)]}'
{"id":"openstack%2Fkolla~840829","triplet_id":"openstack%2Fkolla~master~I09ea136c7354143f45f856d5e8d8ed1192879622","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-05-06 13:24:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I09ea136c7354143f45f856d5e8d8ed1192879622","subject":"prometheus-libvirt-exporter: fix build with newer Go","status":"MERGED","created":"2022-05-06 08:26:02.000000000","updated":"2022-05-06 13:27:14.000000000","submitted":"2022-05-06 13:24:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"840829","meta_rev_id":"d6dda1b6497488856768871fdbbd473174fd2b03","_number":840829,"virtual_id_number":840829,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-06 13:24:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":2,"date":"2022-05-06 09:04:26.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":1,"date":"2022-05-06 12:01:22.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":1,"date":"2022-05-06 12:01:22.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":2,"date":"2022-05-06 12:01:22.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-06 09:04:26.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-05-06 09:57:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ffd1b0e0b21bd030e57bfc4183c9d2df0e61a648","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-05-06 08:26:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"432c5eac7fb0d7b4b16f3b6553102d3c88196aa1","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-06 09:04:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"166dac8719383049336a41db72352c349e9038c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 09:57:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/aaba34e57ef843e59332a3db46125778 : SUCCESS in 37m 03s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/7051857bd86c4f3391c618794b820fc4 : SUCCESS in 1h 22m 40s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/91efe2d7f6664d85aa6593b363a62cb9 : SUCCESS in 1h 25m 27s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c6698059255844f2aa6bf89115b0fe8f : SUCCESS in 21m 13s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/fa72332661884790bcb09e22ccf7552b : SUCCESS in 58m 49s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/a0ba720b2ab64542878da834a5c9cb6c : SUCCESS in 1h 05m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84057e5ced5b45bcba41424b9592c366 : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c2d0cda6bea4427daa7fc6e2b9e8886e : SUCCESS in 3m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/54592e0f45ff44958c57a55c7769b7c2 : SUCCESS in 3m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4ad57c7a824f4c2ea741cf24d0a00310 : SUCCESS in 3m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cc7b88bd1904df1a86733cf7c1592ee : SUCCESS in 6m 49s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d72228188a8541e2995ba2acc608cc61 : SUCCESS in 23m 40s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/dda6db9e33be4858ae648434b6c85667 : SUCCESS in 55m 01s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ec439e8c396f416db9c5021701bbb027 : SUCCESS in 1h 03m 13s","accounts_in_message":[],"_revision_number":1},{"id":"39152035b6f5e35ee198a44ad3f9429dd15e2d57","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 10:35:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-aarch64 https://zuul.opendev.org/t/openstack/build/1f20001c6dc040ad849047475e7eec91 : SUCCESS in 1h 56m 48s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a754db3bac2f4ef6b91a72b8e2c26286 : SUCCESS in 1h 45m 54s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8c8361c2878e44c68cb77b49dcfe4a7b : SUCCESS in 2h 07m 17s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/745e8d69a35346c0a2178930cd07d3ef : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f739844eb2144b399e7158fd93bc275e : SUCCESS in 5m 21s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/6fa3b7f1584f4cd68395d0c393e3b453 : SUCCESS in 1h 27m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"611d6f13b934be2e5dfe1835fcf4a645cadf344d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-06 12:01:22.000000000","message":"Patch Set 1: Backport-Candidate+2 Workflow+1 Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2d5c5939e9b1e54f310b70c487855a05ebe48910","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 12:01:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"43409df9c517a58c58484b0528bd6ba6536d33dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 13:24:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/7285319ea7d04d4daa92e03df9cc3a3b : SUCCESS in 36m 17s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/5167729e14824a57844d0888a872e4d9 : SUCCESS in 1h 16m 07s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/8b6ab07f91aa480eb31167d1270fe66d : SUCCESS in 1h 22m 22s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b36bbc0437a548a79a7fc7bdf56051d4 : SUCCESS in 26m 34s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c92a780d48e844238f9054e85ac59483 : SUCCESS in 1h 00m 46s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/a6d4c4151ece41d98fb2a0ceb531267a : SUCCESS in 1h 15m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a9ff0efe23340bcac95922580bbd6ec : SUCCESS in 3m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/beb7e3ab0e0c49fd88ffbdc528696636 : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/87d5ac867c1248b4a6fe74a9edeb718f : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7793d7f85acf4d9a90a4435980c39b07 : SUCCESS in 6m 17s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/06fcad6c0f0a4782a05db91677637fdc : SUCCESS in 30m 01s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f9c708e39f4b4054904e6ef2361fae71 : SUCCESS in 50m 04s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/314ca345697747e2b73c54c87c5b92c2 : SUCCESS in 1h 00m 42s","accounts_in_message":[],"_revision_number":1},{"id":"8fe60579cb0c7cc3c36a94969a0e461347641ac2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 13:24:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d6dda1b6497488856768871fdbbd473174fd2b03","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 13:27:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71ec15fb50494483b145c72541db2770 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9630032fd61c6f0623bd1ae7577756e7166427f7","revisions":{"9630032fd61c6f0623bd1ae7577756e7166427f7":{"kind":"REWORK","_number":1,"created":"2022-05-06 08:26:02.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/29/840829/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/29/840829/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/840829/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/840829/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/29/840829/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/29/840829/1"}}},"commit":{"parents":[{"commit":"b418bd8592f48980cb6c4875c5827b42c0b136a4","subject":"Merge \"Add multipath to cinder-volume\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b418bd8592f48980cb6c4875c5827b42c0b136a4"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-05-06 08:25:22.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-05-06 08:25:22.000000000","tz":120},"subject":"prometheus-libvirt-exporter: fix build with newer Go","message":"prometheus-libvirt-exporter: fix build with newer Go\n\nI have no idea how Go works. This change make build happen.\n\nChange-Id: I09ea136c7354143f45f856d5e8d8ed1192879622\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9630032fd61c6f0623bd1ae7577756e7166427f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9630032fd61c6f0623bd1ae7577756e7166427f7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
