)]}'
{"id":"openstack%2Fkolla~840954","triplet_id":"openstack%2Fkolla~master~I8245b9e753fa4175dacfa679a078f63876b0d327","project":"openstack/kolla","branch":"master","topic":"redis-sentinel-restarts","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-05-17 07:44:52.000000000","reason":"Marcin Juszkiewicz replied on the change"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2022-05-18 12:54:29.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-05-12 13:49:05.000000000","reason":"Radosław Piliszek replied on the change"}},"hashtags":[],"change_id":"I8245b9e753fa4175dacfa679a078f63876b0d327","subject":"Fix redis-sentinel containers always reporting changed","status":"MERGED","created":"2022-05-06 16:04:03.000000000","updated":"2022-05-18 12:55:46.000000000","submitted":"2022-05-18 12:54:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":32,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"840954-redis-sentinel-restarts","meta_rev_id":"222dec0dc8c95c5b488b99ba7ca19018d0450e9b","_number":840954,"virtual_id_number":840954,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-18 12:54:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2022-05-18 12:55:46.000000000","post_submit":true,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2022-05-18 09:35:55.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2022-05-18 09:35:55.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-06 17:29:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-07 10:03:40.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-05-10 07:47:48.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"}],"messages":[{"id":"5712af8e7b571985fbe480872b1d06472c7f6cbc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-05-06 16:04:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"358f8bbba9879b9797b1ae22f8a8b9203bac0f9d","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-05-06 16:12:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ccdcdcd9730c1c6b32a950d75be05def2aa154e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 17:29:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/21b9658d1e204a85bfd31799863996fe : SUCCESS in 32m 22s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/71836d74a8f243eb88ffce5594fd06a8 : SUCCESS in 1h 15m 41s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/861c9d27a353457bb6134e613e9b302b : SUCCESS in 1h 23m 12s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f19852babdea4a75b321f864869d6100 : SUCCESS in 26m 58s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/63aa7ba000574bf2bb984a5d391eb6d6 : SUCCESS in 57m 54s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/8f7dda31f24a41f4abf04e3fd6165f1e : SUCCESS in 56m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49bddced54f646e5b8e796160c18bfcc : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c5759046cd6543aa97e79543f9aa085a : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c225bad143cc447dbbf8b0ea115cbdbe : SUCCESS in 4m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bc7cb358fc7b40a99e867cbbe44fb911 : SUCCESS in 3m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e2792267c214501bc1b8c6839462131 : SUCCESS in 6m 25s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/94debff08c8c4f23a621598666e16d06 : SUCCESS in 22m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/57d56daa92d5424ba71d385e8562f434 : SUCCESS in 53m 32s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e316e2ba25b44a598fdad9065622aaf3 : SUCCESS in 1h 05m 24s","accounts_in_message":[],"_revision_number":1},{"id":"8894137f848cd89e8512120eb42d890e78f1ed12","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 21:10:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-aarch64 https://zuul.opendev.org/t/openstack/build/9d32a543e0df4ff88cfcab0b932e3c20 : SUCCESS in 1h 59m 15s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3ae040e2381142dea8d286360205617d : SUCCESS in 1h 32m 39s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9ac97b1bb545460fa285158c28dd7330 : SUCCESS in 2h 27m 47s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f9a61678a02248bca50dbc7aff3adc00 : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ea37eeb9b2d8483d8d4e13b42119adc4 : SUCCESS in 5m 35s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/31efcc3eeebf4e98a72a85f16c4b536c : SUCCESS in 1h 37m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c66c346e84f4c19558ae63bc8fcb901da6085926","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-07 10:03:40.000000000","message":"Patch Set 1: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":1},{"id":"7541eff5174aef18c8a25bcc443e8ade5267414e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-05-10 07:40:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"959544882b957bab11929d3f01098a873b3dbf32","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-05-10 07:44:01.000000000","message":"Patch Set 2:\n\n(7 comments)","accounts_in_message":[],"_revision_number":2},{"id":"7230982c065354629ea3a611ed096f229c03dd75","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-05-10 07:47:13.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a66869d623bd242b899512e98e0d8a7945a34000","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 09:06:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/060e7716b16344fdb81ee6510257082d : SUCCESS in 31m 37s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/c2f138dbc0414253ae908bc612600dbf : SUCCESS in 1h 16m 10s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/eceaeca355af4f9f8f8cda09ae00fe65 : SUCCESS in 1h 25m 22s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e234c0f3a10f448c884db5a3e1c76724 : SUCCESS in 27m 21s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8375a683ec044195b592368de841963b : SUCCESS in 1h 01m 32s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/f40452609d0549d491b200b6d5352c0a : SUCCESS in 1h 09m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ff2c1df0e6441059bab0ca294f6086e : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/46beb3bae4f346a5b9557c0d67853934 : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fb7832de055742be89c3b2059a76fe04 : SUCCESS in 3m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/615338bebc1e4b59aaa4e4d4575b1977 : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/458c164cd64a428f8effc09b60f3e0d2 : SUCCESS in 5m 46s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8a074324a4a74eb786551b6563f3ed3b : SUCCESS in 22m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/16d7a8f15fd14fd9b7e6261960ce05db : SUCCESS in 53m 53s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/bb13c914b50f4cfbb53f2a21181a248e : SUCCESS in 1h 03m 10s","accounts_in_message":[],"_revision_number":2},{"id":"528c08db268e088421c88458a7e68bf7c5f8c1e7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 09:40:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-aarch64 https://zuul.opendev.org/t/openstack/build/73a27009770643919a04b0195947fa42 : SUCCESS in 1h 26m 56s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e0114fc9f19f44ea8862a503f33bde21 : SUCCESS in 1h 03m 11s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0a847c6e6aa546a58febf539dee7a122 : FAILURE in 1h 58m 27s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/20221ad39ec7462982d44c58bb0896a7 : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/29b8d9c9fac94881925fa150973556d2 : SUCCESS in 6m 17s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/236f457889384e4bae64644ad9e89641 : SUCCESS in 1h 03m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e0526f6defcc6d232ffd0a7d6f922374612428ff","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-11 13:38:52.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"4a79de5b7536b0886cf007b8fab968a39d666994","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-05-11 15:11:38.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"ea31ce14439e90d60db20726bed1adb883d71461","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-05-11 15:11:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ccaeb2db5bac0bf64ddc92e638ef224113aea4d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 16:44:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/b6931cbc18164da0bbffb84b5960cabb : SUCCESS in 27m 59s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/1062d430bfb34e91b91d4cf65ebb3bba : SUCCESS in 1h 17m 53s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/57dccddf809f4256a4f599025cc5b680 : SUCCESS in 1h 32m 16s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ee5716fdc8604321b49d7b5686c9a361 : SUCCESS in 22m 36s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e7cc66bcf3874ef3a11f682deb2e5e50 : SUCCESS in 1h 09m 26s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/9457e07ea4bb4ad3b3fa25c6d0459c42 : SUCCESS in 1h 20m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fde164cdcc424c6784378950ab7f6037 : SUCCESS in 3m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/766d79b65d4244feab08d9ef30708860 : SUCCESS in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e000319fc72a456fba0b825484fb1e8b : SUCCESS in 3m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/984a907645c5469aa59687091e935a93 : SUCCESS in 3m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a35d5a7d25c14df88ced9f2e839eb855 : SUCCESS in 6m 07s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/9df99b72780d40fcba0ce7b9af33a2a7 : SUCCESS in 21m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9a47d21f72a14c83b328103db165ffbf : SUCCESS in 52m 53s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/fe06dd05c05742d89c8d4e39f80d2256 : SUCCESS in 1h 08m 15s","accounts_in_message":[],"_revision_number":3},{"id":"cba539571362931eb3a9f1089d32315a7cad2fd5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 17:12:29.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-aarch64 https://zuul.opendev.org/t/openstack/build/00397be8c3f6414b8280d2e2723a8a67 : SUCCESS in 1h 57m 59s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e010586735b14ab0b9ac716c05549051 : SUCCESS in 1h 41m 26s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5e0ec26a8585423da93bc13add0bb434 : FAILURE in 1h 49m 54s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/746cfe63ce994898a59c5f11c3740ec3 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/54ab83e51d09435f8e96f19609dc6ae1 : SUCCESS in 6m 26s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/ffb39821783c40ec8678c2f0e4faa186 : SUCCESS in 1h 20m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c087e5fc6d555f03199e10f0055a4b6272253760","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-12 07:07:37.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"4a4e9c689bbfc51fba21f391fa895fa13be6e1db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-05-12 07:10:58.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b76a089a811d34facd8477695865fe2da9b9a26b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-12 07:26:57.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5f390fef0498a2e7b66a6959c981cc041ddebe8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 08:27:02.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/511ffff1b590413880b3e0b5e466f53a : SUCCESS in 27m 35s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/1e4348a3781342648b9b205cfef17dba : SUCCESS in 1h 14m 00s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/19edaf511cff4586984d8235667013a4 : SUCCESS in 1h 11m 14s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/557de54a4e3744ffa11a1ad3b4eea5db : SUCCESS in 19m 44s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ede5f57b720743d5a24a68f42f51162e : SUCCESS in 1h 01m 48s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/41a2f5f4f0854732a97179e0268c6bc4 : SUCCESS in 1h 09m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52493ecb44f24f3a8e22e9afb17cfd82 : SUCCESS in 3m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/33c71ab48fcc4eb2a73afb434123c601 : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/95d79b9f37b34605bff8d2968e200f28 : SUCCESS in 3m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b1b9194818bd409485bec21f0ef40668 : SUCCESS in 4m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69cb4aa650404500b1d9fc19cc7a4c4a : SUCCESS in 5m 55s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/44f4155c745f4507b68422e521ff0460 : SUCCESS in 17m 42s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3a8a1c9835d0496d8a9d55791c0f25f4 : SUCCESS in 53m 42s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f8cd3fc6d0a748479962e5fbba198012 : SUCCESS in 1h 03m 11s","accounts_in_message":[],"_revision_number":4},{"id":"5c1baef94099a695d597c7748b31b373b92f745e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 08:59:42.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-aarch64 https://zuul.opendev.org/t/openstack/build/e11304deda5540db878e9b3c99048c95 : SUCCESS in 1h 40m 45s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a694404ecb71471cbeef32ca09f05936 : SUCCESS in 1h 25m 37s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/38f8da0eb4a248ef8447561cafcf6738 : FAILURE in 1h 46m 30s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/26aa0e5d2a7e4184a421f84dce3c6883 : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/010a5346baa448208c0e1e2f9aede615 : SUCCESS in 5m 31s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/bb179a9a4b514c5c94d9f77be1f91793 : SUCCESS in 1h 05m 00s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4738130f58f824800a5a8b151d43d8c32d5b1f1e","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-05-12 12:07:39.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"709c1c34a8520c30efb3b3a5584fca35f1a25fc0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-05-12 12:09:23.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ca03be88cf9e8d94aa7038afe0d814a61c7bc06d","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-05-12 12:10:50.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2493dd74e09da63640226b489f823f8c707b0b99","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-05-12 12:11:35.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"fa622d4bb93d0e4a4d13f5e43e2733231a6760b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 13:43:03.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/3f50a8259e1c4201af03cca244e65d5e : SUCCESS in 36m 18s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/0c8e52009bd644368f514e2c71ec7b33 : SUCCESS in 1h 14m 00s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/09b3c64027e041ea8cb1be50ec46077c : SUCCESS in 1h 27m 41s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8537d97161db4edba308cbc4600736b8 : SUCCESS in 23m 37s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5ca51d17eda843e59573c036586e3da5 : SUCCESS in 1h 03m 31s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/954925107f64423b926032b63e5ed3f5 : SUCCESS in 1h 09m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8f9359e5340485298cad6a0cef88d1e : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4e35dc3a6d724615ab311d5f188bd8d9 : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7a16467662784f64a5a70693740fc921 : SUCCESS in 3m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9b8ec453503a48cb98c3d26b1e55bb56 : SUCCESS in 3m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc15183319684a158215067ca651360f : SUCCESS in 6m 53s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/026cca113fe9496db8d4bf2656a22475 : SUCCESS in 23m 06s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ce871d5379184a83aa779f2c9b785945 : SUCCESS in 50m 08s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/59c045ee027849ef9327e65e9a0943b9 : SUCCESS in 1h 01m 45s","accounts_in_message":[],"_revision_number":5},{"id":"128253be691786705de79c8a643e5ffd5140afc9","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-12 13:49:05.000000000","message":"Patch Set 5: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"459baaf1645e903f5cbcf86b458546957c42f45d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 14:08:50.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-aarch64 https://zuul.opendev.org/t/openstack/build/6525ea433f4b4eafb4b5b1653ff85086 : SUCCESS in 1h 56m 49s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e46f534cc5dc43918a3c4d39f855cba6 : SUCCESS in 1h 34m 34s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5a30660e54c84512a767b0707cd9df16 : FAILURE in 1h 53m 28s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8e3445b22a59462c8fa35c37b0dad18b : SUCCESS in 7m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fd0de21d2b2347d496ec1db68f02b4cc : SUCCESS in 7m 55s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/6c5a1a9a5d1d4737bc478d8c89223e31 : SUCCESS in 1h 45m 34s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1b0598f2cd89f1c75807d6932a225b1c19fcc93f","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-05-17 07:44:52.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"abe8d6436b44990b96ac1ba4a8030c33c75027c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-17 07:45:16.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"4a6edc90072a9e55021dd172a3f9a6dee9f7efe5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-17 09:16:02.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/b49288e6090249d3a3e1a28b809c2968 : FAILURE in 57m 16s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/0a6200d13a4c4a96ba6c6aa37471e28d : SUCCESS in 1h 10m 08s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/68eb7d34effe4ee594ae52f67eb705da : SUCCESS in 1h 26m 50s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/23603b51980f4e5c856165b4e6336f56 : SUCCESS in 20m 02s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1ce6b3db35f74fb08ff57bf8147ffd8c : SUCCESS in 54m 49s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/91978bdcd9b1479aa481fab8194048da : SUCCESS in 1h 03m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7b32c0be78b4734b27b77b6d60ef7c2 : SUCCESS in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0dd696bd34744bf882ded2fe82ab71cc : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c3f96be6fb0348f582785cc147ec8189 : SUCCESS in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81b88729493a47faa3dd86813a69ab7f : SUCCESS in 6m 05s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/6a40c57af42a478598829205819e6153 : SUCCESS in 39m 41s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ff4c395f5f034c18ade19bd7b6025d4b : SUCCESS in 52m 05s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8f8674f9207f4570a63a4f39eda886db : SUCCESS in 1h 00m 26s","accounts_in_message":[],"_revision_number":5},{"id":"a7dc131889c8e92a6fe9758814793292384f79a0","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-05-17 09:27:26.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"543efa1e263b386b3341e4672c7be95d83dc66fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-17 11:05:39.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/7ae7ae276f9d4b5bb208f77af1a02d8b : FAILURE in 52m 47s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/f49a609392d54279816243cad55fdce5 : SUCCESS in 1h 19m 07s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/b1443a181e234b1898ff0d58f78ce544 : SUCCESS in 1h 23m 22s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9f53fdcffb5145f982d6e487b308ef9d : SUCCESS in 29m 53s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/fff3a84152c94141b818cdc081e69c7d : SUCCESS in 1h 02m 18s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/c13d2573f3e54a90a870af49eb4a10ad : SUCCESS in 1h 05m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e0ce067543745cca7d43751b9498005 : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/82f5ec367b124b08b6ce7f2f9d597621 : SUCCESS in 3m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e040980ce5304067831637b2d7c9f39e : SUCCESS in 3m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a85673c6bfd1452d9f70aa96a2549ad3 : SUCCESS in 3m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f70e20a078842309cae0cd10217a4cd : SUCCESS in 5m 41s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/18a107c9581545bcb5d58618b278c607 : SUCCESS in 24m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/025ca66e38d742e7a8f781ad9cfc891b : SUCCESS in 54m 08s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f676fcc48d96454eba74f07aa224ebda : SUCCESS in 1h 02m 29s","accounts_in_message":[],"_revision_number":5},{"id":"621bd29eb01fe92742a06e2ba87c7ef118130d37","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-17 11:22:08.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-aarch64 https://zuul.opendev.org/t/openstack/build/fe9f459dd2244c8baaba12c383dc79b8 : SUCCESS in 1h 40m 50s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/fbd699dbe0e04348b0b77a8390624a83 : SUCCESS in 28m 45s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/dbae5a34850741e79a1f0bd1ed0c2fd5 : FAILURE in 1h 01m 45s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cdf0c65605574183a7497dff371e9d31 : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5ab4d65442464411b97a02762491d01d : SUCCESS in 5m 47s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/d7f3b414fcbe41d899d5204d391f3456 : SUCCESS in 1h 07m 44s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"22801d9f4518be21f1c34f60eae71baecf8bf51d","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-05-17 11:28:58.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6c794fe0a0b53fd5e10b2285c83b77d98dd4d902","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-17 12:55:43.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/6b6b9e5641794afabd6cb2bccad2fb4e : FAILURE in 55m 05s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/70b20299e2774cd58700ce5b2a16b213 : SUCCESS in 1h 08m 01s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/9df4552f4bda49b885bea0748412eb04 : SUCCESS in 1h 24m 14s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c076b6986f7f47e9864919732509a778 : SUCCESS in 23m 51s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/978bac44a2db4fba8787df856e8983e7 : SUCCESS in 57m 03s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/d1b3eacfc46e4890a1a3f2198d1b22a1 : SUCCESS in 1h 11m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81ee53f1dd724abf8232325d802f6b8c : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba37968e5dae4bf6bc44f802569b5a88 : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5f8dc09d85c84a2ab3dbb9a05a1a076c : SUCCESS in 4m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3c7e6a5ec5cb41f5978e46b3a432530a : SUCCESS in 3m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8f6e8b9e2b148beb1cf11fb0f82f149 : SUCCESS in 6m 13s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/0fffbeee240047bd97f57699ac0997c2 : SUCCESS in 23m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9103a0e173224bb9a63d9cba0e41241e : SUCCESS in 54m 16s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/165e2be0fee54e88bb6275a10d073bc0 : SUCCESS in 1h 04m 20s","accounts_in_message":[],"_revision_number":5},{"id":"0215db5838941026e166b286aab0966c197f525c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-17 13:13:36.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-aarch64 https://zuul.opendev.org/t/openstack/build/6c45e1803878476ea46e566e2781bfe1 : SUCCESS in 1h 31m 41s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/fe3bf6df8f7446b5ade12207daca5658 : SUCCESS in 22m 30s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/80fea34900284a7ba1e1b3c15f74ae1b : FAILURE in 1h 42m 23s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ea25de6f131241019618a3885f242293 : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3167bc1d2b7f4494ad39e709f1e74c3d : SUCCESS in 5m 18s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/fc054a0cb5894661b966bb01cb76fb88 : SUCCESS in 1h 13m 19s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"cd8a3d54acf1b45df7271f4f8ed73ba3b05fb263","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-18 09:35:50.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"1d760c590d4c5631229708043c9dc60eadef397d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-18 09:35:55.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"be8e2af413a2b7b25833f31e3f2482e058ac4aa5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 11:13:53.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/416d222fdda144f6b0156106cfa5843d : SUCCESS in 49m 46s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/2d3011fe92f44ab6a7b5eb242f86d03b : SUCCESS in 1h 07m 18s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/58239e53ff674730a2c140c186fad7f7 : SUCCESS in 1h 28m 36s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/32972c79901440c5ab303ed5524efc6c : SUCCESS in 19m 40s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d6fc87806c3645ea89e4a630297621e3 : SUCCESS in 1h 04m 58s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/7e5364aa0aca44bf9de21c5bb050c542 : SUCCESS in 1h 19m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/378a90d43c4347009da5bd628a11f928 : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6cfc583dad5249eea56a25d97b743231 : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab57dc1be8384845b0d251a3f7cc9a09 : SUCCESS in 3m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/67e6b7e4c1694c09a65a4e7853be59f8 : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bad402b859e241de9f590d1825bc4872 : SUCCESS in 6m 25s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/67ac39262a8248ca80053b6bafb9c2d8 : SUCCESS in 23m 06s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/19ffd00c191f44e19bb3b9aa182f20cc : SUCCESS in 54m 48s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/010a23f0818545b8ab18a72c98023bf1 : SUCCESS in 1h 09m 43s","accounts_in_message":[],"_revision_number":6},{"id":"140c00c32c4d39bf35ccc90db1daa5508a123977","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 11:14:39.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-aarch64 https://zuul.opendev.org/t/openstack/build/e8ffca2119494cf0ab0fc11338e68276 : SUCCESS in 1h 34m 06s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/42612576728a4db08ebed7476146562a : RETRY_LIMIT in 2m 34s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/618feced86644ef1a5f6e70488c1d8b2 : RETRY_LIMIT in 3m 00s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d3a9dda2737b45faa930c57813fac504 : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/33e9fd65e2514c1fbc26d96cf510840a : SUCCESS in 6m 03s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/e8b745c66c3445498a9110f07790e9eb : SUCCESS in 1h 29m 32s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"1aef9fb8709b00f3c28b2f7f522de060083ef969","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 11:17:43.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"1e44c1b289f1593d44f2cc9dc670d741143bc6a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 12:54:28.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/3b637b6b39b64aabb55af3d9ae79ea79 : SUCCESS in 31m 04s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/3be7d6af5e4e468d89b5770f43c35a95 : SUCCESS in 1h 09m 45s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/e2a71fdcfbb0486f9df1b48b3b976643 : SUCCESS in 1h 32m 26s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c5b18db8519144068d2c24e6e4eb82ed : SUCCESS in 19m 41s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ad3e7b09b9bb458eb3dfb79dbbf53e09 : SUCCESS in 1h 03m 00s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/c90b48e3feaa48b2ad503f94a1487cb1 : SUCCESS in 1h 22m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9252140aa2614e279b993b73cc561236 : SUCCESS in 3m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6be470ecceaa4860a8e20bd9f0438ae4 : SUCCESS in 3m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d450ee61f004c1e94e0b226e6bb3406 : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16b5e01306704dccb7f0989ce01fc9e5 : SUCCESS in 5m 21s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b5cce1a4e5ca4a5b9a675f2c3f298781 : SUCCESS in 20m 56s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ad2b1711be8a456a8b05f3c4f06d56d2 : SUCCESS in 54m 20s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3a3512d292484222aa6e137f0a60d8bd : SUCCESS in 1h 11m 02s","accounts_in_message":[],"_revision_number":6},{"id":"e3137a0e74b7c02ed839b4995bf2d81bad4b119b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 12:54:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"28096270c97ad0a857d09f48319d549f1a5d0715","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 12:55:46.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf5b84c8a56949608e064812b8dd8d1d : SUCCESS in 51s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"9040d4f5271912f94635e495d65463956ac9ab75","revisions":{"96d873711cf1f44491f1d7bd59f900c7397830e4":{"kind":"REWORK","_number":1,"created":"2022-05-06 16:04:03.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/54/840954/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/840954/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/840954/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/840954/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/840954/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/840954/1"}}},"commit":{"parents":[{"commit":"684eeab65e88e45adcf6d96eb4b0234fe5ed0e51","subject":"Merge \"prometheus-libvirt-exporter: fix build with newer Go\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/684eeab65e88e45adcf6d96eb4b0234fe5ed0e51"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-05-06 16:00:08.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-05-06 16:04:00.000000000","tz":120},"subject":"Fix redis containers always reporting changed","message":"Fix redis containers always reporting changed\n\nThe CONFIG REWRITE command rewrites the redis.conf\nfile the server was started with, applying the minimal\nchanges needed to make it reflect the configuration\ncurrently used by the server, which may be different\ncompared to the original one because of the use of\nthe CONFIG SET command.\n\nhttps://redis.io/commands/config-rewrite/\n\nBecause of above behaviour it\u0027s needed to\nhack kolla\u0027s CMD.\n\nWithout this hack container\u0027s\n/usr/local/bin/kolla_set_configs --check\nis always reporting changed state even if\nconfig is not changed.\n\nTherefore redis containers are always restarted\nby kolla-ansible. This patch is fixing this.\n\nChange-Id: I8245b9e753fa4175dacfa679a078f63876b0d327\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/96d873711cf1f44491f1d7bd59f900c7397830e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/96d873711cf1f44491f1d7bd59f900c7397830e4"}]},"branch":"refs/heads/master"},"68c8254fcf6a0df390c5a540ea925a2c75ca041c":{"kind":"REWORK","_number":2,"created":"2022-05-10 07:40:04.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/54/840954/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/840954/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/840954/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/840954/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/840954/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/840954/2"}}},"commit":{"parents":[{"commit":"684eeab65e88e45adcf6d96eb4b0234fe5ed0e51","subject":"Merge \"prometheus-libvirt-exporter: fix build with newer Go\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/684eeab65e88e45adcf6d96eb4b0234fe5ed0e51"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-05-06 16:00:08.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-05-10 07:40:02.000000000","tz":120},"subject":"Fix redis containers always reporting changed","message":"Fix redis containers always reporting changed\n\nThe CONFIG REWRITE command rewrites the redis.conf\nfile the server was started with, applying the minimal\nchanges needed to make it reflect the configuration\ncurrently used by the server, which may be different\ncompared to the original one because of the use of\nthe CONFIG SET command.\n\nhttps://redis.io/commands/config-rewrite/\n\nBecause of above behaviour it\u0027s needed to\nhack kolla\u0027s CMD.\n\nWithout this hack container\u0027s\n/usr/local/bin/kolla_set_configs --check\nis always reporting changed state even if\nconfig is not changed.\n\nTherefore redis containers are always restarted\nby kolla-ansible. This patch is fixing this.\n\nChange-Id: I8245b9e753fa4175dacfa679a078f63876b0d327\nPartial-Bug: #1967664\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/68c8254fcf6a0df390c5a540ea925a2c75ca041c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/68c8254fcf6a0df390c5a540ea925a2c75ca041c"}]},"branch":"refs/heads/master"},"c327177aace57f7b1200dbf7a1a75b9b3cf821c5":{"kind":"REWORK","_number":3,"created":"2022-05-11 15:11:40.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/54/840954/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/840954/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/840954/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/840954/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/840954/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/840954/3"}}},"commit":{"parents":[{"commit":"684eeab65e88e45adcf6d96eb4b0234fe5ed0e51","subject":"Merge \"prometheus-libvirt-exporter: fix build with newer Go\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/684eeab65e88e45adcf6d96eb4b0234fe5ed0e51"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-05-06 16:00:08.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-05-11 15:11:38.000000000","tz":120},"subject":"Fix redis-sentinel containers always reporting changed","message":"Fix redis-sentinel containers always reporting changed\n\nThe CONFIG REWRITE command rewrites the redis.conf\nfile the server was started with, applying the minimal\nchanges needed to make it reflect the configuration\ncurrently used by the server, which may be different\ncompared to the original one because of the use of\nthe CONFIG SET command.\n\nhttps://redis.io/commands/config-rewrite/\n\nBecause of above behaviour it\u0027s needed to\nhack kolla\u0027s CMD.\n\nWithout this hack container\u0027s\n/usr/local/bin/kolla_set_configs --check\nis always reporting changed state even if\nconfig is not changed.\n\nTherefore redis containers are always restarted\nby kolla-ansible. This patch is fixing this.\n\nChange-Id: I8245b9e753fa4175dacfa679a078f63876b0d327\nPartial-Bug: #1967664\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c327177aace57f7b1200dbf7a1a75b9b3cf821c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c327177aace57f7b1200dbf7a1a75b9b3cf821c5"}]},"branch":"refs/heads/master"},"a0fb6558df16981097da30505456ff0ab41a5fce":{"kind":"REWORK","_number":4,"created":"2022-05-12 07:10:58.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/54/840954/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/840954/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/840954/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/840954/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/840954/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/840954/4"}}},"commit":{"parents":[{"commit":"684eeab65e88e45adcf6d96eb4b0234fe5ed0e51","subject":"Merge \"prometheus-libvirt-exporter: fix build with newer Go\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/684eeab65e88e45adcf6d96eb4b0234fe5ed0e51"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-05-06 16:00:08.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-05-12 07:10:52.000000000","tz":120},"subject":"Fix redis-sentinel containers always reporting changed","message":"Fix redis-sentinel containers always reporting changed\n\nThe CONFIG REWRITE command rewrites the redis.conf\nfile the server was started with, applying the minimal\nchanges needed to make it reflect the configuration\ncurrently used by the server, which may be different\ncompared to the original one because of the use of\nthe CONFIG SET command.\n\nhttps://redis.io/commands/config-rewrite/\n\nBecause of above behaviour it\u0027s needed to\nhack kolla\u0027s CMD.\n\nWithout this hack container\u0027s\n/usr/local/bin/kolla_set_configs --check\nis always reporting changed state even if\nconfig is not changed.\n\nTherefore redis containers are always restarted\nby kolla-ansible. This patch is fixing this.\n\nChange-Id: I8245b9e753fa4175dacfa679a078f63876b0d327\nPartial-Bug: #1967664\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a0fb6558df16981097da30505456ff0ab41a5fce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a0fb6558df16981097da30505456ff0ab41a5fce"}]},"branch":"refs/heads/master"},"e4199e7fb176d1f8d25070060d48ab8720086cfe":{"kind":"REWORK","_number":5,"created":"2022-05-12 12:09:23.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/54/840954/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/840954/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/840954/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/840954/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/840954/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/840954/5"}}},"commit":{"parents":[{"commit":"684eeab65e88e45adcf6d96eb4b0234fe5ed0e51","subject":"Merge \"prometheus-libvirt-exporter: fix build with newer Go\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/684eeab65e88e45adcf6d96eb4b0234fe5ed0e51"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-05-06 16:00:08.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-05-12 12:09:21.000000000","tz":120},"subject":"Fix redis-sentinel containers always reporting changed","message":"Fix redis-sentinel containers always reporting changed\n\nThe CONFIG REWRITE command rewrites the redis.conf\nfile the server was started with, applying the minimal\nchanges needed to make it reflect the configuration\ncurrently used by the server, which may be different\ncompared to the original one because of the use of\nthe CONFIG SET command.\n\nhttps://redis.io/commands/config-rewrite/\n\nBecause of above behaviour it\u0027s needed to\nhack kolla\u0027s CMD.\n\nWithout this hack container\u0027s\n/usr/local/bin/kolla_set_configs --check\nis always reporting changed state even if\nconfig is not changed.\n\nTherefore redis containers are always restarted\nby kolla-ansible. This patch is fixing this.\n\nChange-Id: I8245b9e753fa4175dacfa679a078f63876b0d327\nPartial-Bug: #1967664\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e4199e7fb176d1f8d25070060d48ab8720086cfe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e4199e7fb176d1f8d25070060d48ab8720086cfe"}]},"branch":"refs/heads/master"},"9040d4f5271912f94635e495d65463956ac9ab75":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2022-05-18 09:35:50.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/54/840954/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/54/840954/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/840954/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/840954/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/54/840954/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/54/840954/6"}}},"commit":{"parents":[{"commit":"0e3b29c5fdfccd80a8951f38f015103359d24783","subject":"[bifrost] Force Bifrost to use the correct git branch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0e3b29c5fdfccd80a8951f38f015103359d24783"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-05-06 16:00:08.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-05-18 09:35:50.000000000","tz":0},"subject":"Fix redis-sentinel containers always reporting changed","message":"Fix redis-sentinel containers always reporting changed\n\nThe CONFIG REWRITE command rewrites the redis.conf\nfile the server was started with, applying the minimal\nchanges needed to make it reflect the configuration\ncurrently used by the server, which may be different\ncompared to the original one because of the use of\nthe CONFIG SET command.\n\nhttps://redis.io/commands/config-rewrite/\n\nBecause of above behaviour it\u0027s needed to\nhack kolla\u0027s CMD.\n\nWithout this hack container\u0027s\n/usr/local/bin/kolla_set_configs --check\nis always reporting changed state even if\nconfig is not changed.\n\nTherefore redis containers are always restarted\nby kolla-ansible. This patch is fixing this.\n\nChange-Id: I8245b9e753fa4175dacfa679a078f63876b0d327\nPartial-Bug: #1967664\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9040d4f5271912f94635e495d65463956ac9ab75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9040d4f5271912f94635e495d65463956ac9ab75"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
