)]}'
{"id":"openstack%2Fkolla~841147","triplet_id":"openstack%2Fkolla~stable%2Fyoga~I1f17717af38f1ad66a6a568701b6fcf5a15077fb","project":"openstack/kolla","branch":"stable/yoga","topic":"venus-yoga-not-released","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-05-09 19:39:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1f17717af38f1ad66a6a568701b6fcf5a15077fb","subject":"venus: mark as unbuildable as it was not released","status":"MERGED","created":"2022-05-09 16:52:54.000000000","updated":"2022-05-09 19:41:04.000000000","submitted":"2022-05-09 19:39:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"841147-venus-yoga-not-released","meta_rev_id":"b0eb1f56e6187d62a98950ed50a31e9d44742124","_number":841147,"virtual_id_number":841147,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-09 19:39:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-05-09 16:57:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-05-09 16:54:39.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-05-09 16:57:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-09 16:54:39.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-05-09 16:57:46.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-05-09 18:17:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3847a51b404565885f01763a6a2c2dd15308a5df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-05-09 16:52:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8f2f4bf07a2fd37df22c5759e058a5ccf8c1710d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-09 16:54:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d7b0e502773ff2cb46a39bb775e53553714f54ca","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-05-09 16:57:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"23418a39b3338a2131389792e68dc0afb6391a0f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-09 17:11:55.000000000","message":"Topic set to venus-yoga-not-released","accounts_in_message":[],"_revision_number":1},{"id":"97ec184449adbf13325d74ae98219ecd83852168","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-09 18:17:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/7b1e41a63b8f4ec78c5d54cc8b347479 : SUCCESS in 31m 46s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/a07338accc094d928f1cfe3a07176275 : SUCCESS in 33m 05s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/92eaeb836ea24275a5e7273bb65bbbef : SUCCESS in 1h 15m 08s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/9b8316d2ea384460b72f2259137fa00e : SUCCESS in 1h 06m 28s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/8419acefd3c74396953fb7b442a87d17 : SUCCESS in 1h 19m 44s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/951aeb64c6c045a3bf82c288ecdc020d : SUCCESS in 22m 00s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f427206dd4c74c45bd985044233e2f25 : SUCCESS in 17m 50s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4e7ad43a8eeb4b24890225380484e0f2 : SUCCESS in 59m 33s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/c7d34c3a939b4113a66ed8c049291f51 : SUCCESS in 1h 12m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08913c9e5f4c4f8f98cbba5be27e2126 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd8b122b177a4f15b5f96eb3cc97db7e : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5ee94eadcd9247208fbc6996e936c237 : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db7390fafcff4c79b9e584a8271fb821 : SUCCESS in 6m 25s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2fd46d899b674ed39c8f79ef56e80ad5 : SUCCESS in 18m 36s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a087f89c76304777aaf3ec07409ce0fe : SUCCESS in 22m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6be0522a8b5944ef9ea7c50190b9e67f : SUCCESS in 50m 01s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/90b446af7b3e43f397dc72d0dbdbe2e7 : SUCCESS in 53m 01s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/85a7519cae9946dd9bb37359ee2b2d6a : SUCCESS in 59m 11s","accounts_in_message":[],"_revision_number":1},{"id":"68d80073c1e67123f2f8561e0a8af821b60a531c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-09 18:17:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a2887cd9075150042d7e389776a996f62d1f9463","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-09 19:02:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/7c0dc520cac64f499e8c99a51ea1f992 : SUCCESS in 2h 06m 41s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e35370b77fdd4e4385647a52abf5f809 : SUCCESS in 1h 34m 12s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/86491467b2804fd3a738029b9ad1b4ca : FAILURE in 1h 52m 59s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/16fc33f1292b49e9a6e2c0732fb6b240 : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/04cde3bc76cc4e089c535cd49a2cdfc8 : SUCCESS in 5m 35s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/3a620ef0946e40ae850fca0f239ed95e : SUCCESS in 1h 30m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f636760a7d5471dd4a429fb7909c3f5b978fcd5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-09 19:39:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/4d47d750adbb4e39983d5f615d2462b8 : SUCCESS in 32m 52s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/cac3d5b6f67f43ffaf6b89cfef22e514 : SUCCESS in 1h 11m 35s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/529441768f7243ef8af4f603515449b6 : SUCCESS in 1h 17m 58s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/aaac88e1c35048f49bd5af9a3a8fc9d9 : SUCCESS in 22m 02s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8fa6e3ec030c410c855a50f01c13e30c : SUCCESS in 1h 13m 56s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/da0850abbb6b4dba83f3c14909b562b3 : SUCCESS in 1h 09m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/868db6e315f14f2481faef1eee5bb485 : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/197cf7d5f9e2448f9df796284d600930 : SUCCESS in 3m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b9d5fdd99e15423383fd8c2c2d5ddfd2 : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc24efe184834e6481ca3ba2c6b559b1 : SUCCESS in 5m 46s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4a7a0ce5a4384351ae819da2867fba36 : SUCCESS in 22m 42s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/af3f1099e79842dc8a0cbea1ac628d75 : SUCCESS in 51m 40s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5796f1f459664f618364da9a47b71e02 : SUCCESS in 1h 02m 42s","accounts_in_message":[],"_revision_number":1},{"id":"40cd68f12da197703651cb547b35744c6872401c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-09 19:39:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b0eb1f56e6187d62a98950ed50a31e9d44742124","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-09 19:41:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eafc0d02ff91410f8477af2f7ab9ff75 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2db1201827654344e9b9b381192f4b7147ab8648","revisions":{"2db1201827654344e9b9b381192f4b7147ab8648":{"kind":"REWORK","_number":1,"created":"2022-05-09 16:52:54.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/47/841147/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/841147/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/841147/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/841147/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/841147/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/841147/1"}}},"commit":{"parents":[{"commit":"3d837170ea7196a08692a5f9f715e8b41da189a8","subject":"cloudkitty: disable building for ubuntu/binary","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3d837170ea7196a08692a5f9f715e8b41da189a8"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-05-09 16:52:08.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-05-09 16:52:08.000000000","tz":120},"subject":"venus: mark as unbuildable as it was not released","message":"venus: mark as unbuildable as it was not released\n\nhttps://review.opendev.org/c/openstack/releases/+/824394 is still open\nwhich means Venus is not released for stable/yoga still.\n\nChange-Id: I1f17717af38f1ad66a6a568701b6fcf5a15077fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2db1201827654344e9b9b381192f4b7147ab8648"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2db1201827654344e9b9b381192f4b7147ab8648"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
