)]}'
{"id":"openstack%2Fkolla~841227","triplet_id":"openstack%2Fkolla~stable%2Fyoga~I83ab903fa74bbb6e963c0f05e4be8792f0253f3b","project":"openstack/kolla","branch":"stable/yoga","topic":"revert-venus","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-05-10 10:53:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I83ab903fa74bbb6e963c0f05e4be8792f0253f3b","subject":"Revert \"venus: mark as unbuildable as it was not released\"","status":"MERGED","created":"2022-05-10 08:10:46.000000000","updated":"2022-05-10 10:55:14.000000000","submitted":"2022-05-10 10:53:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"841227-revert-venus","meta_rev_id":"62f0fbf01cf922e97f09206b07394efbdcf6710c","_number":841227,"virtual_id_number":841227,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-10 10:53:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2022-05-10 08:14:07.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2022-05-10 08:57:02.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2022-05-10 08:57:02.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-10 08:14:07.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2022-05-10 08:57:02.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-05-10 09:28:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a0a2eb7c97106616cc0f861140e770f0e83cc3f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-05-10 08:10:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"97588ab12037fe65018ff96e5166fdfbb9bcdb8a","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-05-10 08:14:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d50c3e88c13661a0fca5d7c3ff05fe56add6523f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-10 08:57:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9856e5f28caafdf4f4ef8bb279a8dba43fbd2ad6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 09:28:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/1f56d9d8614140ce9314650d529c0bff : SUCCESS in 26m 33s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/6d9a482d92c2467c851d2fccf1efb4c6 : SUCCESS in 23m 57s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/a0ff71c55fe74d92a2b2cb465683bacf : SUCCESS in 1h 01m 46s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/93d40a2ee456457081db52d646758662 : SUCCESS in 1h 02m 35s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/c7430d4a94d1400db48d03788d2923a9 : SUCCESS in 1h 07m 33s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b3331d33cb6840339ff5cfc0c9d2dfc0 : SUCCESS in 22m 05s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f60350e0502b4a948742bf9ca0c7f73d : SUCCESS in 15m 44s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b133209b957d42d99dbe09269bc2aeaf : SUCCESS in 55m 52s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/c198a789885447e48aec0c9feb56b992 : SUCCESS in 1h 14m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d88aaa8ba15446aad6bc243d3bd3c36 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/401ce4c49f8f421b86c42d407d4343b7 : SUCCESS in 3m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e90bab36cb51481d9d3080a66e554601 : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee8eda2314184b90aaf978115da8062c : SUCCESS in 5m 22s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3af6b36d867f4304883efeb2898ed0bc : SUCCESS in 17m 06s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/011f37bab2424e54b92a105acb589f27 : SUCCESS in 27m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4c328adca9fe498abe43e5b83d8803e0 : SUCCESS in 51m 21s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b596199bf346407890ab068a96501d02 : SUCCESS in 49m 34s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/63cab9bafc754b2ba72526064b5f3bd0 : SUCCESS in 57m 59s","accounts_in_message":[],"_revision_number":1},{"id":"9ed2c0b61eceebd6f8edab3b6d7eb0bb9b706b50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 09:29:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"276d157d600c80d30cfd614896af3d5690c6828a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 10:44:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/b916a105586c4603bd2c78eddc4cff0a : SUCCESS in 1h 56m 29s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d3d0fd6072b04961a51126db28224701 : SUCCESS in 1h 31m 22s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c6402a2793ca4bbfafad42eafa079636 : FAILURE in 1h 39m 44s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7c4ff7246e804dc6abfd8e4f40350e1b : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fd712722e997449394adf1e414079185 : SUCCESS in 6m 09s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/b79f772d162147859925b6f10a19e529 : SUCCESS in 1h 32m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dbbdd7ce263ed9c63b37372028e917f50f3e046f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 10:53:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/aeee63cf95db4e57b7cba0864654c2aa : SUCCESS in 36m 23s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/596d7c8ca1134489b21592ff357b68ea : SUCCESS in 1h 14m 27s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/80caa3fc1b25485ca7e2e0c026fa1717 : SUCCESS in 1h 23m 21s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b0e2b0aeb9274398bc14de7926d2f26b : SUCCESS in 19m 50s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/30fda47e29954797b945724499e9d067 : SUCCESS in 57m 57s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/a772cd45caca413da3c1291afc376c1e : SUCCESS in 1h 02m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca35aeadf8354197ad3ddc7bbc952ff5 : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7f4052e9947748439cab99087d345abf : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/195a2bc8d7f14315861077f3af3e1e83 : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84433955cc7b4907a1f957a8ed9b5549 : SUCCESS in 5m 20s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8ca34f78fd6446c48edec6187c99d469 : SUCCESS in 31m 47s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/df01fb0dcd184d92aa54c45ddb878fa5 : SUCCESS in 55m 05s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/595cc627fb924476b78f5e15566d251e : SUCCESS in 1h 00m 41s","accounts_in_message":[],"_revision_number":1},{"id":"052ff6d34c46ca5c957824a9fee485e6d5964960","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 10:53:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"62f0fbf01cf922e97f09206b07394efbdcf6710c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 10:55:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/344d39d885c448639cfdc3d2571ee8ba : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f77793b9eb36c088cfddc1f73eca5b4cb112bc04","revisions":{"f77793b9eb36c088cfddc1f73eca5b4cb112bc04":{"kind":"REWORK","_number":1,"created":"2022-05-10 08:10:46.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/27/841227/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/27/841227/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/841227/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/841227/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/841227/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/27/841227/1"}}},"commit":{"parents":[{"commit":"2db1201827654344e9b9b381192f4b7147ab8648","subject":"venus: mark as unbuildable as it was not released","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2db1201827654344e9b9b381192f4b7147ab8648"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-05-10 07:56:36.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-05-10 08:10:23.000000000","tz":120},"subject":"Revert \"venus: mark as unbuildable as it was not released\"","message":"Revert \"venus: mark as unbuildable as it was not released\"\n\nReason for revert: venus was not released for yoga and thus isn\u0027t\navailable for this branch\n\nThis reverts commit 2db1201827654344e9b9b381192f4b7147ab8648.\n\nChange-Id: I83ab903fa74bbb6e963c0f05e4be8792f0253f3b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f77793b9eb36c088cfddc1f73eca5b4cb112bc04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f77793b9eb36c088cfddc1f73eca5b4cb112bc04"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
