)]}'
{"id":"openstack%2Fkolla~841228","triplet_id":"openstack%2Fkolla~stable%2Fyoga~Ic298608c5711916942c1001a4c1530c4d15912ef","project":"openstack/kolla","branch":"stable/yoga","topic":"revert-venus","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-05-10 11:16:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic298608c5711916942c1001a4c1530c4d15912ef","subject":"Revert \"venus: add log management system\"","status":"MERGED","created":"2022-05-10 08:10:46.000000000","updated":"2022-05-10 11:23:14.000000000","submitted":"2022-05-10 11:16:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"841228-revert-venus","meta_rev_id":"548f805e4c8e7092ecd6232b00831abd5fcb6337","_number":841228,"virtual_id_number":841228,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-10 11:16:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2022-05-10 08:14:00.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2022-05-10 08:57:08.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2022-05-10 08:57:08.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-10 08:14:00.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2022-05-10 08:57:08.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-05-10 09:45:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cc6208ff1264aea4732a12147e4c41c29d0f0526","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-05-10 08:10:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"03bef29c2ac0387979badd6e69a40095ab4ed105","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-05-10 08:14:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4aa94a0be3b113093595079f7cf4ce8a7fc52991","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-10 08:57:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"49297c881ff8a1310e496c2be3e1a2c5c77f85cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 09:45:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/40f01881b1e242239984df1fcbcc8400 : SUCCESS in 28m 34s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/ca988e3f81c24cb498973c6268e86105 : SUCCESS in 27m 48s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/82b6c39338cd497599c25ca753475f5a : SUCCESS in 1h 05m 36s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/381a28abccc94122b7aaf813accb7f91 : SUCCESS in 1h 00m 13s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/7fb3654e44b14a80b7e2ae4cf3f13f62 : SUCCESS in 1h 33m 09s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1f3d3388bc1d492d854fd1cf230445d7 : SUCCESS in 23m 25s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/73e4f6f1679642e481d58cebc90dcfaa : SUCCESS in 18m 04s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/fd0151ba59ca445ea85b49f2d8be6a3f : SUCCESS in 56m 16s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/f5d767ad09524661a392846d91752464 : SUCCESS in 1h 14m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3eb90e5b3ca445fbd899f978def30b0 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5410c98a5555468a8bd9641d98df027f : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f557be320c545ff8c6c163d2fc79b5f : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ea51b0d4ac74c7eb5f89806cc6fef04 : SUCCESS in 6m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6a12ddbf6a1646c386197af2f91a9799 : SUCCESS in 4m 18s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f97526785caa4873afc5509204e16e3a : SUCCESS in 34m 18s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5786619a7fe94879a417578bc0ed26c9 : SUCCESS in 26m 42s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9eba268f514c4631973de4dfd1c16ae4 : SUCCESS in 57m 49s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8c452c9cefb347ef9da96b407d26803b : SUCCESS in 50m 37s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/82d89d680aa3424ca0ff2216cf09c651 : SUCCESS in 1h 01m 19s","accounts_in_message":[],"_revision_number":1},{"id":"b7850d10cb3562a40433642ef2fce7ef5185cecf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 09:45:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7f17ff1cbf739ab3b228f5ba08d51584fd9efbf4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 11:16:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/b5b5786c229642bab4f51c611cf120e0 : SUCCESS in 26m 08s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/14003d66cf7b40a58294f807cd1f6914 : SUCCESS in 1h 10m 45s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/0427ff6833a541cd90f8d78b2e7fbd9c : SUCCESS in 1h 22m 38s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c71b6e121c1f42abbf56a27fbeec9c83 : SUCCESS in 24m 23s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0870c95422e94a3b81f9ff937d181c2f : SUCCESS in 1h 00m 13s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/cda32c0f21534054bef1a3894ca69220 : SUCCESS in 1h 10m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd779300645847808ceb9644f41c6af3 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/99f1a97e96214f129f8fa33efa907e31 : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6f09fbd6e93647419b6d2bb632f81c97 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed04d1b32d0b41998477bfd0ec490d35 : SUCCESS in 5m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa88d9a8af314e37ad3bda7a535c5ba3 : SUCCESS in 3m 53s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6425f53491384d119b3ddcf4717aee23 : SUCCESS in 22m 46s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9cb47575007042eba0532323207d7851 : SUCCESS in 50m 02s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/56f7b51300044d59b56928f68d166724 : SUCCESS in 1h 00m 40s","accounts_in_message":[],"_revision_number":1},{"id":"1b1d6fc9d66f2163bb93325bc7814e3bda69d8f3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 11:16:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"eb7aad7bba46d2a64bc50233804300bf844eae4f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 11:17:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/440135dde1ab4015b4ddf7760cfa6cad : SUCCESS in 1m 10s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e15fd0d93168486d823e42e360f7a3bd : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1},{"id":"548f805e4c8e7092ecd6232b00831abd5fcb6337","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 11:23:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/3d6acc1fd81c420f9bbabc543e7a1c0e : SUCCESS in 1h 48m 49s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/83681cb7964c4df3aa501f044c6ebd5f : SUCCESS in 1h 26m 11s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5657d2970b984816bde2fd2bee091812 : FAILURE in 1h 51m 07s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/437c73f046954b068b5ac117791b68e5 : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d44d9fb334934d028c32eec312db8e00 : SUCCESS in 7m 48s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/002356ae423b4f83984a7f7218d85aea : SUCCESS in 1h 08m 31s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c8b8c91ecfc1fb63aa492e5bb45ec23c55af01ea","revisions":{"c8b8c91ecfc1fb63aa492e5bb45ec23c55af01ea":{"kind":"REWORK","_number":1,"created":"2022-05-10 08:10:46.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/28/841228/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/841228/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/841228/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/841228/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/841228/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/841228/1"}}},"commit":{"parents":[{"commit":"f77793b9eb36c088cfddc1f73eca5b4cb112bc04","subject":"Revert \"venus: mark as unbuildable as it was not released\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f77793b9eb36c088cfddc1f73eca5b4cb112bc04"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-05-10 08:00:03.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-05-10 08:10:29.000000000","tz":120},"subject":"Revert \"venus: add log management system\"","message":"Revert \"venus: add log management system\"\n\nReason for revert: venus was not released for yoga and thus isn\u0027t\navailable for this branch\n\nFixed merge conflicts introduced by [0] and [1].\n\nThis reverts commit 3d51f064646940ce5034489238120ab2a6752879.\n\n[0] I8bdce3b76bc02fdcc029f90cf1984e6b3d4eb961\n[1] I4fff070ffef0a10887e1f26809bb43d20f48ea35\n\nChange-Id: Ic298608c5711916942c1001a4c1530c4d15912ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c8b8c91ecfc1fb63aa492e5bb45ec23c55af01ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c8b8c91ecfc1fb63aa492e5bb45ec23c55af01ea"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
