)]}'
{"id":"openstack%2Fkolla~841271","triplet_id":"openstack%2Fkolla~stable%2Fvictoria~I97102e18e4ec47bca94d17d964936f039580b06b","project":"openstack/kolla","branch":"stable/victoria","topic":"pip-setuptools-layout-error","attention_set":{},"removed_from_attention_set":{"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2022-05-10 14:04:08.000000000","reason":"Michal Arbet replied on the change"}},"hashtags":[],"change_id":"I97102e18e4ec47bca94d17d964936f039580b06b","subject":"macros/pip: revert to old setuptools way","status":"MERGED","created":"2022-05-10 13:46:28.000000000","updated":"2022-05-10 16:19:09.000000000","submitted":"2022-05-10 16:17:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"841271-pip-setuptools-layout-error","cherry_pick_of_change":835787,"cherry_pick_of_patch_set":1,"meta_rev_id":"20bf04fd63ccc191bad508b6ee8620d056f15bf6","_number":841271,"virtual_id_number":841271,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-10 16:17:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2022-05-10 14:02:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2022-05-10 13:53:21.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2022-05-10 14:02:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-10 13:46:28.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"CC"},{"updated":"2022-05-10 13:53:21.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2022-05-10 14:02:49.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2022-05-10 15:09:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e3518ea5951fcb6d7b44419c1e1b938551494419","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-05-10 13:46:28.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* docker/macros.j2","accounts_in_message":[],"_revision_number":1},{"id":"fa288969d6318eb4cb3b4516cf7e56a163dcd276","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-05-10 13:49:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e21700eb740cf36d95c4ec8321942ef1585d713b","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-05-10 13:49:58.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"55f54ed3166a10aa6954c4d698313140abcbd85e","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-05-10 13:53:21.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"df7794fc248ad638974535bf28ae0310f49c946b","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-05-10 14:02:49.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"454bc34d2393c0e24e734327f644bb2e23630653","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-05-10 14:04:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"013468937c1bfb8f7d19560d943425cff3bc146f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 15:09:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/6c4261295b204febb91711237b5bf41e : SUCCESS in 31m 12s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/4de8a53493994eb7a3bbc17dea0fe39f : SUCCESS in 32m 35s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/004d49d353c74dd8ada922a65a4c19d2 : SUCCESS in 1h 05m 11s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/e4f71e73dde04c0a852b2d30827de37c : SUCCESS in 1h 02m 15s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/540406b21aeb409eb402a06604a2e4f2 : SUCCESS in 1h 15m 40s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3fa0d73e111c4cd8b0262d6073978300 : SUCCESS in 18m 38s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ef9458ebdaee4beb8605302bcd708bf3 : SUCCESS in 17m 01s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9bbbccf5180547c7aac8f4d2d7e06590 : SUCCESS in 54m 39s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5dead573dea1477a89008c870433f4ab : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/75a2e8f44bd741f0be2f2435f8421f1e : SUCCESS in 3m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2728d4ee5ad94ce4b066f76924e8c882 : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a40ae6868534d2284f58f770039a4af : SUCCESS in 5m 15s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a9cd208ac97445bdba8f2f4f9b24b30c : SUCCESS in 19m 22s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1fabe7bdac794b50894de2740575b166 : SUCCESS in 20m 31s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d3e9f9d1ab9e4e22a7a40d7e82e924f4 : SUCCESS in 47m 18s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cdcc0b2725a648af98fb943b29a5317d : SUCCESS in 46m 53s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c0693b15d75645c39ebcaccf8d1e08a5 : SUCCESS in 58m 52s","accounts_in_message":[],"_revision_number":2},{"id":"6e1990c8934d1764a60e637b49849ca925eb558f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 15:10:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ef788a64a0787a6143fb680e7b77a7671580e18f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 15:54:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/8950481e4b1b4fd78958057cb2016443 : SUCCESS in 1h 46m 23s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4f69a581a4d749f39fc1d426972b8efd : SUCCESS in 1h 37m 11s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/567d28c199cd49e2bae57c28230b1de9 : SUCCESS in 2h 05m 05s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/381b7875044c44f6af506ca767f49c4a : SUCCESS in 1h 34m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a859cb6b9b92e7f6e7132a29caeeb0fcd94bdb42","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 16:17:33.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/72ef16b1bead4a7481fae9b5e4abbc55 : SUCCESS in 35m 57s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/6873be3418e648f5a97e642d904d3f62 : SUCCESS in 1h 03m 40s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/18a144f410814398b0941e39d931d502 : SUCCESS in 1h 04m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be66dc56a2044a1eae57d6de31bca239 : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3cfb4909978c4398885532c8621c8134 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/10af58ef4cf64f85946d21a9f148c1e0 : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/782c0a25d6c4456eb5ab03c6f2652d94 : SUCCESS in 5m 56s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/60d48342b87942f3b8c9dd10db2ff7c0 : SUCCESS in 31m 45s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/13bef93b8cc2486cbe8be4050c76047d : SUCCESS in 51m 32s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a3172cfbf7e641d188dd25ff2d69d9a3 : SUCCESS in 1h 06m 36s","accounts_in_message":[],"_revision_number":2},{"id":"2e651108d76ac1a24574126b2bbb6ca872fb4d52","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 16:17:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"20bf04fd63ccc191bad508b6ee8620d056f15bf6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 16:19:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6670cc296c584594853db2a75c55b4ce : SUCCESS in 53s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3548fa66c8c6b781e734a28dab1823f02b0a080c","revisions":{"b39f8c8cf64bcbdae050ac8221562c5b5a42a94c":{"kind":"REWORK","_number":1,"created":"2022-05-10 13:46:28.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/71/841271/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/71/841271/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/841271/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/841271/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/841271/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/71/841271/1"}}},"commit":{"parents":[{"commit":"fe872a875facdff71a8ffc65929e89b8c67be513","subject":"Merge \"enable logging to file for quiet mode\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fe872a875facdff71a8ffc65929e89b8c67be513"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2022-03-30 07:52:58.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-05-10 13:46:28.000000000","tz":0},"subject":"macros/pip: revert to old setuptools way","message":"macros/pip: revert to old setuptools way\n\nAArch64 builds fail with \"AttributeError: install_layout\". This is due\nto setuptools issue in Debian:\n\nhttps://bugs.debian.org/cgi-bin/bugreport.cgi?bug\u003d1003252\n\nSwitching setuptools to use distutils from stdlib makes error go away.\n\nBackports: Xena, Wallaby (this is where it was first reported).\nChange-Id: I97102e18e4ec47bca94d17d964936f039580b06b\n(cherry picked from commit 497b40fe5a4984b959c63f0c5320311f5810192e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b39f8c8cf64bcbdae050ac8221562c5b5a42a94c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b39f8c8cf64bcbdae050ac8221562c5b5a42a94c"}]},"branch":"refs/heads/stable/victoria"},"3548fa66c8c6b781e734a28dab1823f02b0a080c":{"kind":"REWORK","_number":2,"created":"2022-05-10 13:49:06.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/71/841271/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/71/841271/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/841271/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/841271/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/841271/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/71/841271/2"}}},"commit":{"parents":[{"commit":"fe872a875facdff71a8ffc65929e89b8c67be513","subject":"Merge \"enable logging to file for quiet mode\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fe872a875facdff71a8ffc65929e89b8c67be513"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2022-03-30 07:52:58.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-05-10 13:49:04.000000000","tz":120},"subject":"macros/pip: revert to old setuptools way","message":"macros/pip: revert to old setuptools way\n\nAArch64 builds fail with \"AttributeError: install_layout\". This is due\nto setuptools issue in Debian:\n\nhttps://bugs.debian.org/cgi-bin/bugreport.cgi?bug\u003d1003252\n\nSwitching setuptools to use distutils from stdlib makes error go away.\n\nBackports: Xena, Wallaby (this is where it was first reported).\nChange-Id: I97102e18e4ec47bca94d17d964936f039580b06b\n(cherry picked from commit 497b40fe5a4984b959c63f0c5320311f5810192e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3548fa66c8c6b781e734a28dab1823f02b0a080c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3548fa66c8c6b781e734a28dab1823f02b0a080c"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
