)]}'
{"id":"openstack%2Fkolla~841519","triplet_id":"openstack%2Fkolla~master~Id75b35a09b14280ed0e3b5aec8c9822357e40a93","project":"openstack/kolla","branch":"master","topic":"drop-py36","attention_set":{},"removed_from_attention_set":{"32029":{"account":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"last_update":"2022-05-12 03:53:14.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Id75b35a09b14280ed0e3b5aec8c9822357e40a93","subject":"The Python 3.6 and Python 3.7 Support has been dropped since zed","status":"ABANDONED","created":"2022-05-12 01:21:55.000000000","updated":"2022-05-12 03:53:14.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"66d9b1412fd5cd1a402c24dd17199cd2b3c3a2cd","_number":841519,"virtual_id_number":841519,"owner":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-05-12 02:36:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-12 02:36:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2e615ae3ed1fc1e55f3d4586ed6c8916d9da2292","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2022-05-12 01:21:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9a3b37649f835cc5d6ea66bf5ac556aa88728dba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 02:36:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/cea44a865721464388ddeac89fd0249e : SUCCESS in 2m 56s\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/a4430e33a038483eb51c00585611dc13 : RETRY_LIMIT in 3m 07s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/78d798e536ff4df5ad76d5f25ba06ebe : FAILURE in 15m 22s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/e08640c14295495296a25b0e3d95e7be : FAILURE in 14m 01s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/cb481e4ec07348fcb2a9db09e5ba1f82 : SUCCESS in 20m 29s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7d3b75dfda9d404794068c50e11ff8d2 : SUCCESS in 1h 00m 21s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/33c543311bf64c159e6ae2330564452b : SUCCESS in 1h 05m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16983a0e2d0a4cb2b4bde6e6fe11b4f1 : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7057012b776c42f9911f844c38123595 : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7dd10006f4e6469ea65906741a84341f : SUCCESS in 4m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bbe4bb44316f46fc9e872c41a3ab5595 : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d811e9b1ae8d46698d3fc49a9725c660 : SUCCESS in 7m 30s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/82e8d59234d24aeeb7ab938df6c78313 : SUCCESS in 22m 54s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e5d58dc2761e4a09b13136d28de50545 : SUCCESS in 48m 13s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3bb08a6d46cf4f7d8f1c40a08c119c99 : SUCCESS in 1h 03m 49s","accounts_in_message":[],"_revision_number":1},{"id":"609dad00f6faaa842b1099d73b9b5d81652a9c40","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 03:31:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-aarch64 https://zuul.opendev.org/t/openstack/build/ea298c4330374c659dcc1729191f796a : RETRY_LIMIT in 4m 46s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/fa3787131f594afd97c23ed91abb9307 : SUCCESS in 1h 13m 14s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/194c08b81bf547e6b87da6ac9f51c40c : FAILURE in 2h 06m 44s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a831eb58951e429c9766ceb791f2534a : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/da9cb653db584715acc84010f6f939d1 : SUCCESS in 6m 01s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/361df0c2d08442af908453e3465e854e : SUCCESS in 1h 05m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"66d9b1412fd5cd1a402c24dd17199cd2b3c3a2cd","tag":"autogenerated:gerrit:abandon","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2022-05-12 03:53:14.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"89f176e446a25578679138fe8e3c1209cdf690f6","revisions":{"89f176e446a25578679138fe8e3c1209cdf690f6":{"kind":"REWORK","_number":1,"created":"2022-05-12 01:21:55.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/19/841519/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/19/841519/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/841519/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/841519/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/841519/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/19/841519/1"}}},"commit":{"parents":[{"commit":"684eeab65e88e45adcf6d96eb4b0234fe5ed0e51","subject":"Merge \"prometheus-libvirt-exporter: fix build with newer Go\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/684eeab65e88e45adcf6d96eb4b0234fe5ed0e51"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2022-05-12 01:21:36.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2022-05-12 01:21:36.000000000","tz":480},"subject":"The Python 3.6 and Python 3.7 Support has been dropped since zed","message":"The Python 3.6 and Python 3.7 Support has been dropped since zed\n\n[1] https://governance.openstack.org/tc/reference/runtimes/zed.html\n\nChange-Id: Id75b35a09b14280ed0e3b5aec8c9822357e40a93\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/89f176e446a25578679138fe8e3c1209cdf690f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/89f176e446a25578679138fe8e3c1209cdf690f6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
