)]}'
{"id":"openstack%2Fkolla~841656","triplet_id":"openstack%2Fkolla~stable%2Fxena~Ic4b6171789c3d360317599310492771bd78828ec","project":"openstack/kolla","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2022-05-13 08:30:04.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-05-13 06:59:47.000000000","reason":"Radosław Piliszek replied on the change"}},"hashtags":[],"change_id":"Ic4b6171789c3d360317599310492771bd78828ec","subject":"masakari: add Cyrus SASL packages to monitors image","status":"MERGED","created":"2022-05-13 00:35:38.000000000","updated":"2022-05-13 08:31:15.000000000","submitted":"2022-05-13 08:30:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"841656","meta_rev_id":"d47b11e4c1d3495c86705717b762ab764d4d677e","_number":841656,"virtual_id_number":841656,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-13 08:30:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":2,"date":"2022-05-13 06:59:47.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2022-05-13 07:03:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":1,"date":"2022-05-13 07:03:04.000000000","_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}],"CC":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-13 00:35:38.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2022-05-13 00:46:18.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-05-13 02:05:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-13 07:03:04.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"}],"messages":[{"id":"a19cea64dd850ca77e81d5c6997c31ad0f105c72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-05-13 00:35:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"28ea91fff4fe116a4ad93899e71122d69ba5490d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-13 02:05:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/805b229283f04dd0aa7f4d1c41311680 : SUCCESS in 47m 16s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/bd6ba6ba4405483f9c27f6d4e4628ded : SUCCESS in 51m 32s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/8eab1b4a868c448b9b9774823a16162e : SUCCESS in 1h 21m 16s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/fd26a19652a243e299bce12e3c3dd0fa : SUCCESS in 1h 19m 31s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/cf8d2311a4f54286bf5b12cb317fcd5f : SUCCESS in 1h 27m 59s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/03b0c2fe49c3420ab87a49dc1095111f : SUCCESS in 16m 21s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/161bcd9d34b340ffaf17c293b0762e40 : SUCCESS in 15m 48s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8ffc5414d4624452805e59347c53b376 : SUCCESS in 51m 01s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c6d7174875d46ddbe2cf3da20d427d5 : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8fba12e13e94845981f7380b955e03c : SUCCESS in 3m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c625dfed7ff84b929124ac74ce2e4774 : SUCCESS in 3m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fdd7872315ba416b92dd1c8fa402990a : SUCCESS in 3m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad1baec76dfa4b338159eb88ba0d9a45 : SUCCESS in 5m 45s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6ce6f6c444f44944b8bbae9a49ff8d9e : SUCCESS in 46m 37s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ac12af56e73949ec9476738ec0623088 : SUCCESS in 46m 14s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9ab01f9316a54d8cbd3dd2821608867c : SUCCESS in 1h 08m 27s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a198b68573ba43fe969fd1b66eb1a8e6 : SUCCESS in 1h 06m 19s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d97190cbb117419fb9bbe1418c0d6fe9 : SUCCESS in 1h 21m 25s","accounts_in_message":[],"_revision_number":1},{"id":"db72367396851ba4d4c37d2e0c111fe7d73101ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-13 02:31:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/b683eff7f44f45a49e78696c0f17542f : SUCCESS in 1h 53m 59s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/43eecaf7328c45bab98ae67eefec8442 : SUCCESS in 1h 08m 40s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/71c4159eb0a34f9ba849127e7c8c6b78 : SUCCESS in 1h 49m 56s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4f64b931d9f64f3480eb681793b49cfe : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f8f02d22e3f64c2a96642bdd96f1e796 : SUCCESS in 6m 14s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/19e0538085194de58e6e2b8931d9db2c : SUCCESS in 1h 41m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b662f4ee6d1e422d7fb4eae0a8bdbbbd537332a4","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-13 06:59:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c8c3a9b1b3801d6bfd52ef67ab98b9fda685bc1b","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-05-13 07:03:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"eeacb57ba06b10412ca953057d637ab4316f7330","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-13 07:03:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2b772064985cd0e70a79f1c29fcdc5b44b70ea67","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-13 08:30:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/6a1da983cdaa4e10b963200130cad824 : SUCCESS in 35m 55s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/af80c50053ee4e6fbbcb942db0011413 : SUCCESS in 1h 22m 14s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/996459b845bc4762b4d39668a1dd3cd2 : SUCCESS in 1h 26m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/544094ac6a0a4eb889ca6d7e15feed09 : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/884c2d4d95794f98b5c72aea34f18146 : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/787c772735d24f25bcd0ded19da1183f : SUCCESS in 3m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b42cc7a1d81e4a0181ea36c2774ad95b : SUCCESS in 5m 50s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/de892dcdd8134058b15e11270ee08210 : SUCCESS in 22m 13s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d7b44b0ec3bf4fdfa231a6c48d2e6921 : SUCCESS in 49m 55s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e111e2848459497f99fcbcf5d055ad6b : SUCCESS in 1h 00m 45s","accounts_in_message":[],"_revision_number":1},{"id":"72bb8efb74913c6265a5e62d813a107ce6db3e1b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-13 08:30:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d47b11e4c1d3495c86705717b762ab764d4d677e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-13 08:31:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68594dbacf5d4477ad2c63ce8c13ad68 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0009ffb7e8b6dd08510b67b9f94b2f96595837ff","revisions":{"0009ffb7e8b6dd08510b67b9f94b2f96595837ff":{"kind":"REWORK","_number":1,"created":"2022-05-13 00:35:38.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/56/841656/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/841656/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/841656/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/841656/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/841656/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/841656/1"}}},"commit":{"parents":[{"commit":"cfd0fb0f2804090d9a220e31f6293b84febbcf21","subject":"cloudkitty: disable building for ubuntu/binary","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cfd0fb0f2804090d9a220e31f6293b84febbcf21"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-21 11:44:38.000000000","tz":0},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-05-13 00:35:15.000000000","tz":180},"subject":"masakari: add Cyrus SASL packages to monitors image","message":"masakari: add Cyrus SASL packages to monitors image\n\nAdds Cyrus SASL packages necessary for the DIGEST-MD5 and SCRAM-SHA-256\nmechanisms. These can be used for libvirt SASL authentication.\nFollow up to I13e19ca29eeab40cd08fa3afe2cdf7531867f81b.\n\nPartial-Bug: #1964013\nChange-Id: Ic4b6171789c3d360317599310492771bd78828ec\n(cherry picked from commit 5cc480a617c68ed20c80bffec16321a937f5a32b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0009ffb7e8b6dd08510b67b9f94b2f96595837ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0009ffb7e8b6dd08510b67b9f94b2f96595837ff"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
