)]}'
{"id":"openstack%2Fkolla~841657","triplet_id":"openstack%2Fkolla~stable%2Fwallaby~Ic4b6171789c3d360317599310492771bd78828ec","project":"openstack/kolla","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2022-05-13 08:30:06.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-05-13 06:59:51.000000000","reason":"Radosław Piliszek replied on the change"}},"hashtags":[],"change_id":"Ic4b6171789c3d360317599310492771bd78828ec","subject":"masakari: add Cyrus SASL packages to monitors image","status":"MERGED","created":"2022-05-13 00:36:17.000000000","updated":"2022-05-13 08:31:16.000000000","submitted":"2022-05-13 08:30:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"841657","meta_rev_id":"94a02804b57b8a5d0f9606ca6e8fa778ae1d8a9f","_number":841657,"virtual_id_number":841657,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-13 08:30:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":2,"date":"2022-05-13 06:59:51.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2022-05-13 07:03:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":1,"date":"2022-05-13 07:03:34.000000000","_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-13 00:36:17.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2022-05-13 00:46:28.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-05-13 02:01:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-13 07:03:34.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"}],"messages":[{"id":"b12234986a9bc7eff95818e5322984c0e7a92b91","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-05-13 00:36:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"35fbe0a551633c3f3f5ff89ec200deed6661da6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-13 02:01:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/09182e430b194bbb97106aebdcad7c7d : SUCCESS in 48m 07s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/19be6105e4b4462191f07f428c3df942 : SUCCESS in 51m 09s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/209f39e2175647f888cb857306bd089c : SUCCESS in 1h 19m 47s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/08290c3aeaaa44c29c4a521328e42e77 : SUCCESS in 1h 12m 21s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/1d1196acb9874e619b06e1359ed5ed0d : SUCCESS in 1h 24m 06s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/bbf4ab50df0644ba869fdb745532c317 : SUCCESS in 20m 35s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a394ae12ccc546ea97e714b178a47a14 : SUCCESS in 15m 17s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7190e103394246209fe1c0cc32eb293f : SUCCESS in 57m 03s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2913320e6c44ccaa09d12f0c4b8ce28 : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f418ff2b1b774f9183727bf01b443851 : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fac7ca07e21b414ab532d01293ce73ab : SUCCESS in 3m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fb79efeb04df402e8e0ab068062162a6 : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7951aa2cf1804568a2d532eb4b321fba : SUCCESS in 5m 28s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8bd2677b70704b94a4bd5c26f42d2562 : SUCCESS in 20m 10s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/aa04e0e76f4742bba255d74385dd4650 : SUCCESS in 13m 53s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e3026d6c8bd0480e85984fc8bcf6bb29 : SUCCESS in 51m 35s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/83dee052fbe9495b847c8bde7ad24d5b : SUCCESS in 56m 12s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f1ebbd8baa2a46a1a9b77af62176b7cb : SUCCESS in 1h 03m 02s","accounts_in_message":[],"_revision_number":1},{"id":"6a0dbd860878b32679851b4790b661b9787f6f47","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-13 02:47:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/fb93dd70b7404535888a35776b949494 : SUCCESS in 2h 05m 02s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4b35211f92cc430c8cddc20e97932eac : SUCCESS in 1h 23m 15s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/fd55f9504f26488fbf438381f1a93e89 : SUCCESS in 2h 09m 05s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/7f41420ae93842f3bea9b44f3602bc11 : SUCCESS in 1h 05m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6627f77d5259cbfca9f34646951db22bc90821db","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-13 06:59:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"10a686acebbfac1966ae12cf205462f643facd96","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-05-13 07:03:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dac4ad56a1336deba6eabd7ef6a5884d85849030","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-13 07:04:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cff089d225ee636efcf8471bc65262f9c7d5e1c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-13 08:30:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/3fa1cc6771c74cd892b0bc85973bcd9b : SUCCESS in 25m 04s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/872eeb68399548a29aabcb6836dd5e9c : SUCCESS in 1h 02m 50s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/9dadf252f4f84a4aa7683d3348b69728 : SUCCESS in 1h 18m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01bb251e653c4954a877c0813db6cd93 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b85a427fd2b445839883e2b26dbc71b7 : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6594a56068f34e13aca888b0bf8c78f7 : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a77d902217e448db526f29ecaa84173 : SUCCESS in 5m 30s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/57967dc029664401ba0ecfe903e8c4c7 : SUCCESS in 20m 02s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0576695931f6405ca1a112c43eee68f6 : SUCCESS in 52m 46s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/164955b883fe40f6a009c32a11aa7f35 : SUCCESS in 1h 03m 01s","accounts_in_message":[],"_revision_number":1},{"id":"3a631537fdfc11db5b9e27f43406b10b748ac16b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-13 08:30:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"94a02804b57b8a5d0f9606ca6e8fa778ae1d8a9f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-13 08:31:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/feacc63441d34a6d8d3f3af347852787 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"444bdffcc7d4612fa6dd25fdead8f56428151735","revisions":{"444bdffcc7d4612fa6dd25fdead8f56428151735":{"kind":"REWORK","_number":1,"created":"2022-05-13 00:36:17.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/57/841657/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/841657/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/841657/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/841657/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/841657/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/841657/1"}}},"commit":{"parents":[{"commit":"cfc36552087fb28a2ab0aacfe3f27497a0945a26","subject":"cloudkitty: disable building for ubuntu/binary","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cfc36552087fb28a2ab0aacfe3f27497a0945a26"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-21 11:44:38.000000000","tz":0},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-05-13 00:36:04.000000000","tz":180},"subject":"masakari: add Cyrus SASL packages to monitors image","message":"masakari: add Cyrus SASL packages to monitors image\n\nAdds Cyrus SASL packages necessary for the DIGEST-MD5 and SCRAM-SHA-256\nmechanisms. These can be used for libvirt SASL authentication.\nFollow up to I13e19ca29eeab40cd08fa3afe2cdf7531867f81b.\n\nPartial-Bug: #1964013\nChange-Id: Ic4b6171789c3d360317599310492771bd78828ec\n(cherry picked from commit 5cc480a617c68ed20c80bffec16321a937f5a32b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/444bdffcc7d4612fa6dd25fdead8f56428151735"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/444bdffcc7d4612fa6dd25fdead8f56428151735"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
