)]}'
{"id":"openstack%2Fkolla~842244","triplet_id":"openstack%2Fkolla~master~I23ca2e402c491f20985b2416c173edd1564cc170","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"3031":{"account":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"last_update":"2024-03-28 14:49:33.000000000","reason":"Change was abandoned"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-03-28 14:49:33.000000000","reason":"Change was abandoned"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-06-20 09:04:07.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_30491\u003e by clicking the attention icon","reason_account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}}},"hashtags":[],"change_id":"I23ca2e402c491f20985b2416c173edd1564cc170","subject":"Add script to start apache2 easily","status":"ABANDONED","created":"2022-05-18 00:22:02.000000000","updated":"2024-03-28 14:49:33.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5f4db45739798e18573072261aa38eed6f9f9242","_number":842244,"virtual_id_number":842244,"owner":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-05-18 01:36:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":-1,"date":"2022-08-09 12:57:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-18 01:36:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-18 08:29:09.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-08-09 12:57:24.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2023-03-25 20:27:11.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REMOVED"},{"updated":"2024-03-26 12:55:14.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"be84195086d8444c10625516205d5717d7cfbbd0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2022-05-18 00:22:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3a7670344c8c9cf234c5adbab6b44f368b0a1704","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 01:36:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/2a6d2a82fa1843f89c631213cc008fd9 : FAILURE in 57m 47s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/dd2ee1217a8447f7a087706ff685c0a8 : SUCCESS in 1h 05m 08s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/bea0d65c1d3d41b083a9906acd1a0015 : SUCCESS in 1h 13m 41s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/92783f2236ab4094bab41d81a6d11139 : SUCCESS in 17m 56s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7cc2ad8518734e33828d5a3e55f7551e : SUCCESS in 53m 46s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/cfd8fa987fac4360ae28d24ec808c173 : SUCCESS in 49m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dda6d51d938b46b7915cdeb6fb15a525 : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5434885970c340569061d519245ce43a : SUCCESS in 3m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/14042811ea1d4e2595979e24643f6040 : SUCCESS in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4c7da6dec4a944c0a3ab98e0beb3ea80 : SUCCESS in 4m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5991d4474ef348849a065166270c4310 : SUCCESS in 5m 53s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/3de2f2c1a0b149c183b9b72f4fd0804b : SUCCESS in 27m 02s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b0658d8baa844b6190a6827ce04b29fe : SUCCESS in 1h 00m 19s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/982b1a81dd9b4429b3c2980d4edf21e3 : SUCCESS in 1h 06m 00s","accounts_in_message":[],"_revision_number":1},{"id":"d5e6db57c692162165bf975990ee71e1b4b91d9e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 06:25:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-aarch64 https://zuul.opendev.org/t/openstack/build/cbfc03fd4ceb40ebb2d1f589362829ea : SUCCESS in 1h 15m 43s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/83ca0f0d26624354a88cf5c19b49c349 : SUCCESS in 1h 06m 35s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ab78463f975a467ba507db487879ee2c : FAILURE in 59m 58s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/87b664d07a3240da8298654ec251c11c : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cf3208a369334cad97acf76538996b4c : SUCCESS in 5m 51s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/8aab16c5f75944b6905be9febae8944f : TIMED_OUT in 6h 01m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"185abc1cd56424e82ce399549146f8c8d42c13db","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-18 08:29:09.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e7f55744df98f34c5c11dbe6ff0946309a88618a","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2022-06-10 13:55:14.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"4edca34ddfb2dc7d56795b2a8be0cf661bfe3359","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-10 15:42:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/6aeaa1bfa15f46599b4e8dd620390f81 : SUCCESS in 22m 08s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/699dbd5392c24f269a4c66e7397f5601 : SUCCESS in 1h 02m 30s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/22039deaafc245609f0713364bcd6700 : SUCCESS in 1h 22m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/baa28ec973174e6d88f9b60e500b9640 : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1fb4dfaefea84cc9a51ac0081f258787 : FAILURE in 3m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/87735490118f49eab378ed1f4bb522fb : SUCCESS in 3m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7e0e2b945592479382d72e336cffd2c9 : SUCCESS in 3m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7e9dfbeed49454f87d28ddffef954b3 : SUCCESS in 5m 25s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f6278e4c7ad04d878b152d7a52abb437 : SUCCESS in 30m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/899380c9c89a429ab33d7e2567c4b837 : FAILURE in 5m 35s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a7460f852c9d4dbba950e595100e0959 : SUCCESS in 57m 21s","accounts_in_message":[],"_revision_number":1},{"id":"00856b619fa30263920c091bd79d202bffcdf4a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-10 15:48:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6d0df89f873645c1ac8b38151290b827 : SUCCESS in 1h 46m 03s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4a878c2090fd4c6cbaf590ff93d85954 : SUCCESS in 1h 50m 56s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a5bda29a4d07462eb1bf80ee38943185 : SUCCESS in 5m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1e512d27e14a46a09e07631e4f808e51 : SUCCESS in 5m 43s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/6a6ce93f121640c092a2bb561d94a4e1 : FAILURE in 52m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6d5ea2b966b01c584cf21a2af5c5001a568c1540","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-08-09 12:57:24.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"de5ebe3e4a7f02f4e380953e8186ed9c7f1b7677","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2023-03-25 20:27:11.000000000","message":"Removed reviewer \u003cGERRIT_ACCOUNT_30491\u003e with the following votes:\n\n* Code-Review-1 by \u003cGERRIT_ACCOUNT_30491\u003e\n","accounts_in_message":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}],"_revision_number":1},{"id":"5f4db45739798e18573072261aa38eed6f9f9242","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-03-28 14:49:33.000000000","message":"Abandoned\n\nno updates since 2022","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0105c90625f193bc61ebf5b37cf0653a58d818c6","revisions":{"0105c90625f193bc61ebf5b37cf0653a58d818c6":{"kind":"REWORK","_number":1,"created":"2022-05-18 00:22:02.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/44/842244/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/842244/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/842244/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/842244/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/842244/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/842244/1"}}},"commit":{"parents":[{"commit":"9edcdb0a7ab57764b6e8622e5a75d4a1ecce2515","subject":"Merge \"[publishing] use a new naming scheme\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9edcdb0a7ab57764b6e8622e5a75d4a1ecce2515"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2022-05-18 00:20:31.000000000","tz":600},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2022-05-18 00:21:58.000000000","tz":600},"subject":"Add script to start apache2 easily","message":"Add script to start apache2 easily\n\nSometimes it is handy to be able to just run apache\nbut this isn\u0027t possible without first sourcing the\nhttpd_setup.sh. Add a little helper script so\nit can be used as the docker command.\n\nChange-Id: I23ca2e402c491f20985b2416c173edd1564cc170\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0105c90625f193bc61ebf5b37cf0653a58d818c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0105c90625f193bc61ebf5b37cf0653a58d818c6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
