)]}'
{"id":"openstack%2Fkolla~842392","triplet_id":"openstack%2Fkolla~stable%2Fxena~I77b0dcd2e9dfd8ea8390a471b80c8954b67ef91b","project":"openstack/kolla","branch":"stable/xena","topic":"local-git-stable/yoga-stable/xena","attention_set":{},"removed_from_attention_set":{"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-05-24 10:42:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I77b0dcd2e9dfd8ea8390a471b80c8954b67ef91b","subject":"Fix local sources of git repositories","status":"MERGED","created":"2022-05-18 17:28:20.000000000","updated":"2022-05-24 10:45:38.000000000","submitted":"2022-05-24 10:42:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"842392-local-git-stable/yoga-stable/xena","cherry_pick_of_change":842391,"cherry_pick_of_patch_set":1,"meta_rev_id":"67c7fb34d525a11529ce5eb5f17cca913f0ca76e","_number":842392,"virtual_id_number":842392,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-24 10:42:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-05-24 08:50:56.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2022-05-18 17:28:25.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2022-05-24 08:50:56.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-18 18:56:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-18 19:00:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-24 08:50:56.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"ccf3ab5a8b756fa68d88ca0b8c466e0d1d65fb25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-18 17:28:20.000000000","message":"Patch Set 1: Cherry Picked from branch stable/yoga.","accounts_in_message":[],"_revision_number":1},{"id":"0d2b412489dd3e64e9b1b77c51449a3bdbf7af06","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-18 17:28:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"126807f5d2917607e0e1af9b9f2f441754df4fe0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 18:56:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/dddf366bcb5f4921be1757df7e5d1c49 : SUCCESS in 42m 09s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7290e631d1a348cdb6c343adee3caefc : RETRY_LIMIT in 2m 44s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7c745b1f1f7f47cab39e791351e465e3 : RETRY_LIMIT in 3m 13s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c46375fb33cc4f5794fd0827dc2e49b7 : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e25945ab106444ea807ed520961676f2 : SUCCESS in 6m 43s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/0558c5067e1e43469ae60ecd8e4e9d7a : SUCCESS in 1h 25m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a5aee8542e4e531a81a92714ddce3bc2a8ef0aaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 19:00:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/7a8361baa93d493e8b06b0fa79481e8c : SUCCESS in 31m 39s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/b5b65d1e389548c9871df5e4fc955183 : SUCCESS in 31m 42s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/4ebabb3b2b6843939d622a3d79f7dd33 : SUCCESS in 1h 10m 14s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/fd1686add1f6444e9f019ca0653376c6 : SUCCESS in 1h 09m 06s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/1fbedc540b384d0285f82b6ac5781aad : SUCCESS in 1h 25m 41s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/aa54486e7413403c9fa61e072df67a0b : SUCCESS in 17m 26s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/5c0eebf541ff447399cb2ca8ccce498f : SUCCESS in 18m 33s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/71a2a81fc8b9465ab5e483bf7046597b : SUCCESS in 1h 07m 40s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d675d8238f4457c865c2a605c97dc50 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2cfe408ea1884a7fad37af41e33f6cd7 : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7da7cd299f7e43888c6ea0317c5f6b89 : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/211c522edfd9416a95bcace32e7f366c : SUCCESS in 3m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c413819dc9e42598eac502b94ecceb5 : SUCCESS in 5m 55s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f4ade7564d304df6be907a630ff0d816 : SUCCESS in 21m 44s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1fa337c517ac4e7f9ac36b371993ccad : SUCCESS in 20m 26s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7eed4416ce0a4463979680f4b3de7f8b : SUCCESS in 47m 42s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/084e2f5e2a6546b18cbcf3c21d9deda9 : SUCCESS in 50m 40s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2d1a6453b3824a778612b7cc308d40f0 : SUCCESS in 53m 04s","accounts_in_message":[],"_revision_number":1},{"id":"6fde811b70686c493c621eeceff94b3584fe36a1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-05-24 08:50:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"54b0249212a3a993df4d8c328e18d92260ec5b03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-24 08:51:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e6c5137e855712cdb0970371dbd32435f68f0548","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-24 10:42:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/4bc67043c13442c39e1be5e6861f8913 : SUCCESS in 23m 44s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/5cefc5ede9134d36857445f72125c5c4 : SUCCESS in 1h 08m 58s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/cc81b0ac341b446985a3a4127d383bd3 : SUCCESS in 1h 10m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e331437f649c4c35a5bc150b2fca10d2 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0bd005e7e6404c3bbf6fb338496072eb : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f166106097f84577b47a22fb03577147 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6102af0e8d5c4f69ad0434002e3a175d : SUCCESS in 6m 05s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e7896d4a0d04416a9bdbba91edda098f : SUCCESS in 22m 45s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ebcb763fade44eab920ea778a8d0ece6 : SUCCESS in 51m 21s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0bde6a1aa4824ca1808ba31a8b82cda0 : SUCCESS in 1h 08m 02s","accounts_in_message":[],"_revision_number":1},{"id":"78a7597595107068520caf14793bb1903a216a39","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-24 10:42:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"67c7fb34d525a11529ce5eb5f17cca913f0ca76e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-24 10:45:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be22de6fba22491595a237cf385e925b : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0e54080e61ad194906c8d49ba1e33b2ff0388504","revisions":{"0e54080e61ad194906c8d49ba1e33b2ff0388504":{"kind":"REWORK","_number":1,"created":"2022-05-18 17:28:20.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/92/842392/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/92/842392/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/842392/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/842392/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/842392/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/92/842392/1"}}},"commit":{"parents":[{"commit":"0009ffb7e8b6dd08510b67b9f94b2f96595837ff","subject":"masakari: add Cyrus SASL packages to monitors image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0009ffb7e8b6dd08510b67b9f94b2f96595837ff"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-05-18 14:20:12.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-05-18 17:28:20.000000000","tz":0},"subject":"Fix local sources of git repositories","message":"Fix local sources of git repositories\n\nThis is I2cbf1f539880d512aa223c3ef3a4b19ee18854ac extended to fix\nthe case when a git repository is used with a git repo.\nThis is probably a rarer use case but, still, we use it in CI\nfor in-review changes testing.\n\nChange-Id: I77b0dcd2e9dfd8ea8390a471b80c8954b67ef91b\n(cherry picked from commit b888f68daf6cb3ed9782fe85045858b6a04d4911)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0e54080e61ad194906c8d49ba1e33b2ff0388504"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0e54080e61ad194906c8d49ba1e33b2ff0388504"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
