)]}'
{"id":"openstack%2Fkolla~842394","triplet_id":"openstack%2Fkolla~stable%2Fvictoria~I77b0dcd2e9dfd8ea8390a471b80c8954b67ef91b","project":"openstack/kolla","branch":"stable/victoria","topic":"local-git-stable/yoga-stable/xena-stable/wallaby-stable/victoria","attention_set":{},"removed_from_attention_set":{"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-05-24 10:42:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I77b0dcd2e9dfd8ea8390a471b80c8954b67ef91b","subject":"Fix local sources of git repositories","status":"MERGED","created":"2022-05-18 17:28:42.000000000","updated":"2022-05-24 10:45:39.000000000","submitted":"2022-05-24 10:42:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"842394-local-git-stable/yoga-stable/xena-stable/wallaby-stable/victoria","cherry_pick_of_change":842393,"cherry_pick_of_patch_set":1,"meta_rev_id":"474b24c205e520cab687edf8d234c4e270c8bdb7","_number":842394,"virtual_id_number":842394,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-24 10:42:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-05-24 08:51:22.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2022-05-18 17:28:48.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2022-05-24 08:51:22.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-18 18:37:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-24 08:51:22.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"502d4b6d546e7ed94c7e61af1abb1494378ac13f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-18 17:28:42.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.","accounts_in_message":[],"_revision_number":1},{"id":"3455f66b41d90c2512aaa4d007dfed58c3ae32fc","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-18 17:28:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"947f2742b96453bf15c42bfcae4991103d18dcce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 18:37:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/dd817eb64cb841a0a978921323a9d8e6 : SUCCESS in 34m 38s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/8d5f19e0601246838b9f68d1724451ae : SUCCESS in 32m 53s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/c494194ae7c04bc4b0b1c99f2c699fa9 : SUCCESS in 1h 00m 03s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/0369b972188d4dd795e517b52aa1715b : SUCCESS in 1h 02m 50s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/2bcf599dff7443ef85498719e7734aa6 : SUCCESS in 1h 06m 20s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/491da64c0de4473992510c0eaa43c384 : FAILURE in 11m 37s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ad5f76e86998417e9c321ee1fcf660bf : FAILURE in 19m 04s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f615f281ca3f4440a305ac192afdd8ae : FAILURE in 12m 14s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67132ff215734f688d0d58b8e5022a1c : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d783f3c37f67444190ef05c59810393a : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21ef18bcf2044c4db98a7e1c4f90e7e0 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8793db5a3e994d83a05f3b9f1bf6fa01 : SUCCESS in 6m 39s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7dd637e5f6654b27939395cac57bb297 : SUCCESS in 28m 42s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/edde90031e5a4413b1abde30aabd71cf : SUCCESS in 27m 31s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/57ece0ef68bb4f51ab3c41713dce8e60 : SUCCESS in 39m 42s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a9b898d9cf304e3c8592b5396c15ef32 : SUCCESS in 44m 44s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/15c5f501849e42a5ba86b3d4808419d9 : SUCCESS in 48m 49s","accounts_in_message":[],"_revision_number":1},{"id":"8954e12406e9615c4968d1c0b16205c954ddf6a4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 19:51:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/2a846c92b0204576835a3c42fc2b025d : SUCCESS in 2h 20m 13s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ba56e8edaa554a178acf684822198830 : RETRY_LIMIT in 3m 32s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ca2788612e734ab68b0ef94b6997e4a6 : RETRY_LIMIT in 2m 50s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/29ebde98896344bd8503a72b49180920 : SUCCESS in 1h 51m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2ed412e8affe0a1b821a3cb5006c666b71270341","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-05-24 08:51:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a16f491ef0247f1200c4dc4529d339ea66004d01","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-24 08:52:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9452ffa252b304dfd4458d65586a5dabec8e231c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-24 10:42:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/f00fc10fb4a7461cb47533bea4114102 : SUCCESS in 35m 34s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/66bca36a78db45519da92c4321575d96 : SUCCESS in 1h 08m 18s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/9ef5238071d74205a36eb8cc34fb4304 : SUCCESS in 1h 11m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94f6c84cbf824c859fea53aa2cc00fb2 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eae433d7e36e44a5abade6f213f71aa8 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b27786db2c2c407992b98fdfda04d9d8 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87291d57c41f4318922bb3d25d507054 : SUCCESS in 7m 23s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/fe53b91b53ee4f4a966b3fe501ea9eb1 : SUCCESS in 33m 17s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c4dbdb15231049bd9aa9799a4a3c0bb1 : SUCCESS in 55m 32s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/fe2cdb72a7ff4713b2a83220b861aad8 : SUCCESS in 1h 02m 57s","accounts_in_message":[],"_revision_number":1},{"id":"8ea2c27bc0aca6139d2af86ad0dd18583155db20","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-24 10:42:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"474b24c205e520cab687edf8d234c4e270c8bdb7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-24 10:45:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49135baa1dae46448b313aa1fedcaea3 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b716b5a5a761fa4a5cb826be2c38389727660715","revisions":{"b716b5a5a761fa4a5cb826be2c38389727660715":{"kind":"REWORK","_number":1,"created":"2022-05-18 17:28:42.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/94/842394/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/94/842394/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/842394/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/842394/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/842394/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/94/842394/1"}}},"commit":{"parents":[{"commit":"3548fa66c8c6b781e734a28dab1823f02b0a080c","subject":"macros/pip: revert to old setuptools way","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3548fa66c8c6b781e734a28dab1823f02b0a080c"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-05-18 14:20:12.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-05-18 17:28:42.000000000","tz":0},"subject":"Fix local sources of git repositories","message":"Fix local sources of git repositories\n\nThis is I2cbf1f539880d512aa223c3ef3a4b19ee18854ac extended to fix\nthe case when a git repository is used with a git repo.\nThis is probably a rarer use case but, still, we use it in CI\nfor in-review changes testing.\n\nChange-Id: I77b0dcd2e9dfd8ea8390a471b80c8954b67ef91b\n(cherry picked from commit b888f68daf6cb3ed9782fe85045858b6a04d4911)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b716b5a5a761fa4a5cb826be2c38389727660715"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b716b5a5a761fa4a5cb826be2c38389727660715"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
