)]}'
{"id":"openstack%2Fkolla~842395","triplet_id":"openstack%2Fkolla~stable%2Fussuri~I77b0dcd2e9dfd8ea8390a471b80c8954b67ef91b","project":"openstack/kolla","branch":"stable/ussuri","topic":"local-git-stable/yoga-stable/xena-stable/wallaby-stable/victoria-stable/ussuri","attention_set":{},"removed_from_attention_set":{"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-05-24 10:42:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I77b0dcd2e9dfd8ea8390a471b80c8954b67ef91b","subject":"Fix local sources of git repositories","status":"MERGED","created":"2022-05-18 17:28:56.000000000","updated":"2022-05-24 10:45:40.000000000","submitted":"2022-05-24 10:42:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"842395-local-git-stable/yoga-stable/xena-stable/wallaby-stable/victoria-stable/ussuri","cherry_pick_of_change":842394,"cherry_pick_of_patch_set":1,"meta_rev_id":"cdd70ff24b0cf8db9801367ed7c7e6b366900df4","_number":842395,"virtual_id_number":842395,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-24 10:42:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-05-24 08:51:35.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2022-05-18 17:29:02.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2022-05-24 08:51:35.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-18 18:37:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-24 08:51:35.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"32edf3a35484fece1be797ce6472130cf973333b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-18 17:28:56.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.","accounts_in_message":[],"_revision_number":1},{"id":"cafb76977449834226d03796daf0cb1968442f79","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-18 17:29:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c84c55994a0ac6343f06781dc3aad5d8a352408c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 18:37:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/bcd70b41da3f45e8afa4916c1f26c69d : SUCCESS in 1h 05m 38s\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/04368fe1f53c4c329c1247ef4cc88c19 : SUCCESS in 36m 43s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e7d1d9f317874d8abb29299e1049dc41 : FAILURE in 9m 37s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ac34f33cee1b4410bfbb2335e315cde3 : FAILURE in 25m 07s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/034bf313b7184a81bf0970d3d7dd2eab : FAILURE in 11m 25s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa9a97a167cc4b54bac51dbee1023305 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/12f08e9b2ed84b0fa1ebe7577f97aa10 : SUCCESS in 3m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/84792982321a430486f30f31f4ac174e : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4eabb806d4c64d469460ebdf7bdb83f4 : SUCCESS in 5m 48s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cd9268322c7f4f3b889fef8d70fa3161 : SUCCESS in 36m 08s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/23eab56da8054bf5b0175b729017545b : SUCCESS in 34m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5dccabb154994a8d9e7a0f9e22c84a8e : SUCCESS in 56m 34s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/dca78413841d44b28b3a230d142ac046 : SUCCESS in 51m 32s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5b75e43955cd47b1ad7d52105f721c53 : SUCCESS in 1h 02m 58s","accounts_in_message":[],"_revision_number":1},{"id":"fb4b250bd4b8d78442bb79803103ca1c473d63f8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 19:53:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/596286d19ea54963b35fc8e7c98cd77c : SUCCESS in 2h 14m 02s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/90e10e2a08c748e3a22f65a087493570 : RETRY_LIMIT in 2m 42s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/27cbb4a1e78d40668021f89fbba9da4f : RETRY_LIMIT in 4m 25s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/651b48a046994b2ab284c465fdc60034 : SUCCESS in 1h 35m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cafe9d497fb59f438739e8c7f8a699f7e157dbed","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-05-24 08:51:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ed5bae0f7c2b1eb0713cbc648c602b4c4f6c9a72","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-24 08:52:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d1c1370dd3cd6da5cffb65a653454b7549b858d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-24 10:42:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/f6b3dec608884c5a82e870a8d3db207e : SUCCESS in 1h 04m 30s\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/065b57c6e558488c92d679380ac8d608 : SUCCESS in 38m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f305a6f597084c9b8ea70e3b40bf30ec : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d59dd3aa5a643fa89421ce8c3e1b579 : SUCCESS in 4m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0e86e807864d464892aa04172baeb3b7 : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03210823c8b74d3d83c637120705dd37 : SUCCESS in 7m 10s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9067a669cffd403c977aeb03098fa7a7 : SUCCESS in 35m 01s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c62272bdbe834e93b4e73152527c4dad : SUCCESS in 57m 26s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b2f0f497f0e54412be27bf35aa879eef : SUCCESS in 1h 07m 47s","accounts_in_message":[],"_revision_number":1},{"id":"99140dd6b28759fb09951360441f556cdf82ea7b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-24 10:42:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"cdd70ff24b0cf8db9801367ed7c7e6b366900df4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-24 10:45:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/256b54f19ab6457e9d4023db30a56158 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"917a7c04380fc3e83b3ea9d4b2b3d7fa8b82dc0f","revisions":{"917a7c04380fc3e83b3ea9d4b2b3d7fa8b82dc0f":{"kind":"REWORK","_number":1,"created":"2022-05-18 17:28:56.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/95/842395/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/95/842395/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/842395/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/842395/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/842395/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/95/842395/1"}}},"commit":{"parents":[{"commit":"13b4506b8e0873fd9f5836f832fd4fe7341236d9","subject":"fluentd: pin yajl-ruby 1.4.1 as we have old Ruby (!x86-64 only)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/13b4506b8e0873fd9f5836f832fd4fe7341236d9"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-05-18 14:20:12.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-05-18 17:28:56.000000000","tz":0},"subject":"Fix local sources of git repositories","message":"Fix local sources of git repositories\n\nThis is I2cbf1f539880d512aa223c3ef3a4b19ee18854ac extended to fix\nthe case when a git repository is used with a git repo.\nThis is probably a rarer use case but, still, we use it in CI\nfor in-review changes testing.\n\nChange-Id: I77b0dcd2e9dfd8ea8390a471b80c8954b67ef91b\n(cherry picked from commit b888f68daf6cb3ed9782fe85045858b6a04d4911)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/917a7c04380fc3e83b3ea9d4b2b3d7fa8b82dc0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/917a7c04380fc3e83b3ea9d4b2b3d7fa8b82dc0f"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
