)]}'
{"id":"openstack%2Fkolla~842472","triplet_id":"openstack%2Fkolla~master~Idc404d70df06543342dbb0a2c370231ae9c41762","project":"openstack/kolla","branch":"master","topic":"override-extra-repos","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-06-10 13:13:13.000000000","reason":"Change was abandoned"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2022-06-01 14:56:00.000000000","reason":"Maksim Malchuk replied on the change"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-06-10 13:13:13.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Idc404d70df06543342dbb0a2c370231ae9c41762","subject":"Add ability to override repos","status":"ABANDONED","created":"2022-05-19 09:19:02.000000000","updated":"2022-06-10 13:13:13.000000000","total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fbb080106e7c29957f9954d5ec939926b440d51c","_number":842472,"virtual_id_number":842472,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-05-30 09:47:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":-1,"date":"2022-06-01 14:50:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-19 10:44:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-19 10:55:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-25 15:04:55.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"CC"},{"updated":"2022-06-01 14:50:56.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"834d560d79b861a7bb4557e0320d57dd0a517b6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-05-19 09:19:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d7d830aca1008366319e89ac4554834b5a7b3871","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-05-19 09:29:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"891ff92a07db8e977574c9e43f13349b57cbaa77","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 10:44:10.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-aarch64 https://zuul.opendev.org/t/openstack/build/1fceeeb5234945f49cf6b90b668843ca : SUCCESS in 33m 46s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f18745f563264aadb74d593c9dd269a6 : RETRY_LIMIT in 2m 14s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/137196a721b148a186a1a2be709618af : RETRY_LIMIT in 4m 27s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a92b7bfa91c548bb8096d5b02c8969f7 : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7fe0ba71dead4860b80bccf98d65ad51 : SUCCESS in 5m 02s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/2bf8e171ef0545c5b650d32f23be5407 : SUCCESS in 1h 12m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"814d1d2759281b0cce73701ffbdd220e90b27546","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 10:55:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/4a5e53d55bde438db420e93b410ac066 : SUCCESS in 32m 34s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/078f36e1531447139b40076b21640761 : SUCCESS in 1h 07m 32s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/180bb2341eaa468bb1f4f73989f0bc1d : SUCCESS in 1h 20m 51s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9530262ce5e44506ac5768c2c4aca24f : SUCCESS in 21m 58s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6e94e432f89643318d5744dfb9327a1e : SUCCESS in 58m 43s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/e447d6f4a7914fe8a6f7d36a156ab990 : SUCCESS in 1h 04m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8fd54c839874216a0f785b06e496317 : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8ab87e1d77074378a0368df0e3ffec9b : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/77fb5e2e4d2346dca93576ab9867788d : SUCCESS in 4m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e1b4de56491c4e0f80ec001fa88c05c7 : SUCCESS in 4m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4d6987c68a2483697949410235b4087 : SUCCESS in 6m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4aa312e11a464265801bbd153f790b38 : SUCCESS in 4m 28s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/92901c082be04b6dacd411a544ac1dba : SUCCESS in 19m 54s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9a305e08a9614a7ab8c52ca165e947a9 : SUCCESS in 51m 23s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8fd9f7cf146b475fa20adbc3c94f7f5a : SUCCESS in 1h 00m 42s","accounts_in_message":[],"_revision_number":2},{"id":"e4e36782f29aa6ce2737b0c163be00d84d74ffe9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-05-25 14:37:06.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"00e72fcbd30a013a42a34df37c7e1629259e1f01","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-05-25 15:04:55.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d9d010c67cc27c9545baaeab415e20036802b946","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-05-25 15:31:26.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"46b636447fa7e16259f9ec9cd5f3171fb01679a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-25 15:51:38.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/33d6dfb8370041fd858911954c92c3be : SUCCESS in 34m 07s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/009bc8fa760a4dfd8841f76cd370a3fd : FAILURE in 27m 44s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/cac8b79ee588495ebbf4f1450dc2f3d0 : FAILURE in 1h 07m 34s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/704a6075c09147eaa176c1e71d728f98 : SUCCESS in 19m 37s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/10ffb7333aa6456a9beb2a36b030e7dc : FAILURE in 32m 15s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/790097c30b2c4b2db85fbe1c1dde26fa : FAILURE in 45m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/529f617975704a0ab1d1b9ea9fce03fd : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4c6ec2d927ba4c8c89b4dcb20c3f851b : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5b95594b4b9241299957d8fc0789374d : SUCCESS in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4c2997bd8a8c4cc79deb4a82fea61d3d : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/156717c187754998bb914044c60b1c54 : SUCCESS in 6m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/294e978fb0214efa834b5ac2d81392a2 : SUCCESS in 4m 06s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/07742f932f594bdb9437a44ed8cacc62 : SUCCESS in 29m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4f5aee2500f24e54b7733619a1b2db2d : FAILURE in 25m 07s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/bb97e4de165e4237a921af0648ff7d4a : FAILURE in 53m 15s","accounts_in_message":[],"_revision_number":3},{"id":"5975dca2364cf5a120ccb63d41153ed808b2a800","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-05-25 15:58:26.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1832e24f92d4f8fc004ad056de2d566c344db699","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-25 16:59:51.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-aarch64 https://zuul.opendev.org/t/openstack/build/94eaeaa93b064614ab5765222438fe04 : SUCCESS in 2h 07m 56s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b501809758dc4ca2b8259fe882d448d6 : SUCCESS in 2h 19m 48s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/261d0ad011ce4045a8968e30a16f4078 : FAILURE in 1h 24m 47s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8cbdec3f58c54390ba04bb9378cf69dd : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/87503175547d4bc9a658f78a7158b61e : SUCCESS in 5m 36s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/d3983e0cf60a405a8175d97d383add70 : SUCCESS in 1h 48m 05s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3833dd4bf499969f65044b036e4c101a3b74c04a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-25 17:13:50.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/566bebd59a5a4817a98b9eadf2e75f9d : SUCCESS in 27m 53s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/a1c6314313cd4b71b132251876c922bd : FAILURE in 31m 05s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/de8b404af26b4c99b07268e9d978f49d : FAILURE in 1h 05m 40s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d28888c137294e9ca4dc88197e1afd44 : SUCCESS in 21m 24s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d4075eaf43b94bcf96d207b0e73e1510 : FAILURE in 28m 20s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/5bda4e44d63d4c5c995c8c8cfe5076c7 : FAILURE in 55m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56d17618331d4f0dafd3baf8b91478fc : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d1f46b4315eb499a91eaa4b11ff1f5f5 : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df18312ab221453a80e36cfd26011c9c : SUCCESS in 4m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/65e06f784b6f46fcbcccfa8deb7a9ab2 : SUCCESS in 3m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f2ea7dd9a624ac28a4568bf92c24805 : SUCCESS in 5m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/742019769ee84be29436f4f4c7319a11 : SUCCESS in 4m 21s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b6494aec3dee4c698db0ffc71da7151c : SUCCESS in 19m 03s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/871b2d8c2ea643b4bee19956a1292608 : FAILURE in 21m 29s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e1bf650ba97d4935b5a86b1bfb2dfba5 : FAILURE in 45m 30s","accounts_in_message":[],"_revision_number":3},{"id":"1d0e8d093b94ad41e23b29489dc8c5b781da1e37","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-26 08:33:02.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"2c208e85f0a82bd564a9f0c416010a0c64ec74d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-26 10:09:48.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/909c65add1c442558ed304f67bb065dd : SUCCESS in 32m 41s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/9a9210305d0e467aadea1e323772418e : SUCCESS in 1h 15m 54s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/13d34c652d8e4630b4a8cd575c7aaa92 : SUCCESS in 1h 31m 12s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/7fd40cdce8f3476a82cb75e1a54bb603 : SUCCESS in 20m 08s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6131a88bae1b40f0932c20c40e37fae4 : SUCCESS in 1h 03m 45s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/79163254e2604a96bf1e3d61db99c844 : SUCCESS in 1h 13m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0663d53fdf214442b07cf8673dc472a2 : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/41e3e1844b684ec7825304d9ed7242a3 : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/070b3cf863004145a10ed041709d7ab9 : SUCCESS in 3m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bb4c9a1144f2478490d0d57ef4edb6e4 : SUCCESS in 3m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4506b37a8d4449bb27ec389af920d85 : SUCCESS in 5m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/251ee983ed93446d870cdbc222e60d86 : SUCCESS in 3m 49s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/6e76d6691f95436b9387885fd9f1a880 : FAILURE in 24m 01s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/988d234ddf334945807043473a4f46a6 : FAILURE in 18m 50s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/af46a11db8a744f6a677c70a62574a09 : FAILURE in 20m 03s","accounts_in_message":[],"_revision_number":4},{"id":"60835d5b5b29a268acf139251c72c1851bfd3bb2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-26 10:58:28.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-aarch64 https://zuul.opendev.org/t/openstack/build/399efa6928fa4d5aae31abb35e151cf3 : SUCCESS in 1h 46m 53s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5fd4ebf9783948ec8d62f9c31e336569 : SUCCESS in 1h 20m 25s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/92a839b8b5db48319d95f20aade6f830 : FAILURE in 1h 48m 34s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0e3088666d234aea8874c14db00a9d37 : SUCCESS in 7m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8e53a01bee6045cfa1f56e85a1107dec : SUCCESS in 8m 11s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/54ed298973bd43feb482620be640d30a : FAILURE in 1h 23m 02s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"456e7118a861a46b89116f90ffbf42e241d74ef0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-05-27 14:07:46.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"473bb64c0d935b8ee7534fb990891ed400712d82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-05-27 14:10:20.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"a07eaf1425a6c97881149ee36032d70fe9d016be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 15:37:36.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/12fcd02c62ba4ad690c89ce45dc844a0 : SUCCESS in 29m 04s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/86d13d59a3f04271b51e210baaef2986 : SUCCESS in 1h 12m 10s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/e5e93e33016e425db71a455937d82f8a : SUCCESS in 1h 19m 10s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e7795e3f07214839962e460f666f4d91 : SUCCESS in 19m 33s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d41373e92518453094e804e0f29a0990 : SUCCESS in 57m 28s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/ca5eb2107844422e9739b35c3eb1c925 : SUCCESS in 1h 02m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8177513befb4a3ab128e349fbb303c0 : SUCCESS in 3m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7f8d5c13e8c940838fdc3c587469727b : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6a43bd94842c42e9adc8db976e9bb6c4 : SUCCESS in 3m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4a3a1984915b44ccb9c0ba8b6721ffea : SUCCESS in 4m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/199561e5f0c14f13a61db1efa1fef861 : SUCCESS in 5m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a09208e37e3c482dbe1cf95aaa52ea21 : SUCCESS in 3m 56s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/82d60d5c7a69458dab7f923b76cba2c6 : FAILURE in 23m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/49d7bf41dcd24e8d806bbca1424c2a2d : FAILURE in 19m 09s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/96e9717de80a45fd9c6898e9a6d7bb31 : FAILURE in 14m 51s","accounts_in_message":[],"_revision_number":6},{"id":"8cc400a16295dc091942a4223637536bc865460f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 16:20:19.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-aarch64 https://zuul.opendev.org/t/openstack/build/7b90fc1ba5844e8eb1860e976dac176f : SUCCESS in 2h 06m 53s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/cbac949506d040738632fe6320591fc3 : SUCCESS in 1h 46m 54s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/84c94e16ecb14fde9d61d31cf3b8707c : FAILURE in 2h 03m 55s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f9e739443f4940048a6d998e80a0c497 : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5317f4609d7141b08fc7ba9b4c94db96 : SUCCESS in 6m 26s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/3a69749ddb7c434abd7fd9c7200d2a7b : FAILURE in 1h 48m 42s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"4655c1173d5f1bdcbd15ff065dd060d1cc1083dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-27 19:34:54.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"b14b2c343d26f9336cf3bbfddb3de5c92a5bd2f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 21:24:34.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/b5efd7b8a364417e9d68827fd83727cb : SUCCESS in 33m 53s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/004feeae968742c8889f6605a72388c0 : SUCCESS in 1h 15m 10s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/558276937d99496e89b423bcbbe32166 : SUCCESS in 1h 44m 52s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4aefe73d0d11407581f62b77acdc3a86 : SUCCESS in 20m 30s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/cfa4c9f004ae4bc58ead9004a3ca2231 : SUCCESS in 1h 00m 49s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/78cc1fbb82b84ddc9f95a73c0e138f39 : SUCCESS in 1h 12m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d83742befdc9454e9e5f0d427d137717 : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b0dc17e0d8cb4924b3e12f6a6cf29b8b : SUCCESS in 3m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7c54a3b26cfb4fe2ae2f824d875e472a : SUCCESS in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ddbe6f2f7a484901a90ef7349b17a4b8 : SUCCESS in 3m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d85984fdcf9441c7a66595a368d79ba3 : SUCCESS in 5m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f4ed8ab2685a4574983ca2b978051450 : SUCCESS in 3m 24s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/dd84f0124bed4c52830d88cccf93beec : SUCCESS in 26m 28s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/001a3ccbad454c329ece0c99b0c2c02e : SUCCESS in 38m 04s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9db87520c8734112bd4493f2c5e01045 : SUCCESS in 1h 06m 39s","accounts_in_message":[],"_revision_number":7},{"id":"eb03a6689081b0ccd32f76263df449a77c4092c9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 22:53:11.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-aarch64 https://zuul.opendev.org/t/openstack/build/00d54b12b228449490f18c74cb85ede8 : SUCCESS in 2h 13m 44s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/eca228260c404b17a2146d76e9633eaa : SUCCESS in 1h 44m 23s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c10e29472e3f483eac697a1a86cc538f : FAILURE in 2h 01m 23s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/41c98dfdf9dd4cbda8c92daeac9c2b20 : SUCCESS in 9m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/71096dd107b541759278a514c8ff3a71 : SUCCESS in 8m 26s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/1004764c5fac4bd29860f1bc7f5025df : FAILURE in 1h 41m 44s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"df674d59965d754c14f241ef95fa7923ea0aa29e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-30 08:12:09.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"e01b78557b364d3cd79f4da20a78858ce90dc207","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 09:47:52.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/6868727783ae4d3098ef4ccf3da55042 : SUCCESS in 33m 38s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/ca369c6d09ec463c8e8e16d9cffe9085 : SUCCESS in 1h 17m 32s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/855d0f26ac0f40d099bbcf1f0ba3a959 : SUCCESS in 1h 29m 37s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/638aa16b01e54632b561497b72cd87a7 : SUCCESS in 22m 10s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c769ba2903cc4b6fa2345c5350dbc491 : SUCCESS in 52m 45s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/005c1d478b99426bae238c3cb80835e5 : SUCCESS in 1h 17m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f9fbcc0e41244b0b64fe51fbb1466c2 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4d0c9c5cd9f94835a2ba1c9750681735 : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/28b6d8927212468ebecb933e73f41180 : SUCCESS in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d36fb775587a43bba8263f3c3dc3fb7e : SUCCESS in 3m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1eb7145796b4497bf5dd2ade547290b : SUCCESS in 6m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b0f31c1fc0c04a0fb5a3308397f786fd : SUCCESS in 3m 33s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c1829fbdad3d453d88aba01b6aaee6f4 : SUCCESS in 21m 22s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1f7ffcb85c064825bc3c34bda2e3e8ec : SUCCESS in 43m 12s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/836e025d062f4462b9592dbe3245e538 : SUCCESS in 1h 09m 37s","accounts_in_message":[],"_revision_number":8},{"id":"dad3c6d3776cc294df801936da7198f260365fe7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 10:42:06.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-aarch64 https://zuul.opendev.org/t/openstack/build/665ef5bf55684341a54ccbc81d0dcae5 : SUCCESS in 2h 03m 41s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8e9a123a33d94024ad85e92a74c7da5f : SUCCESS in 1h 53m 15s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8e37774c8cfe494ca76f6420de45b161 : FAILURE in 2h 20m 58s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4e83f5a0104f49b6882924066be90183 : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/81e98db5ebe347b19fc7bdb69c9ffec3 : SUCCESS in 7m 22s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/7f274f5f83bc424a9b39b817ae920343 : FAILURE in 1h 49m 42s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"d628bd92b2a73c669621b18eb96a4ba09c4653be","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-06-01 14:50:56.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"ea737d9aac49665cf914b07e5f1c8a79c7bf5866","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-06-01 14:56:00.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"fbb080106e7c29957f9954d5ec939926b440d51c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-06-10 13:13:13.000000000","message":"Abandoned\n\nhttps://review.opendev.org/c/openstack/kayobe/+/845338","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"aabcc9ffbb9d35ad9b5d5970e6636285140d2f14","revisions":{"299ca00d56fc4fef632b7b8c668b998c0b11f84f":{"kind":"REWORK","_number":1,"created":"2022-05-19 09:19:02.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/72/842472/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/72/842472/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/72/842472/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/72/842472/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/72/842472/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/72/842472/1"}}},"commit":{"parents":[{"commit":"b888f68daf6cb3ed9782fe85045858b6a04d4911","subject":"Fix local sources of git repositories","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b888f68daf6cb3ed9782fe85045858b6a04d4911"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-03-27 21:32:42.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-05-19 09:18:00.000000000","tz":180},"subject":"Add ability to override repos","message":"Add ability to override repos\n\nSince Icf2a8397a8349e0fe849d88d160409fd234480a9 the way of handling\nexternal repositories has changed, all repos are moved to the file\n\u0027kolla/kolla/template/repos.yaml\u0027 which is the part of the Python\ncode instead of configuration files in \u0027docker/\u0027 directory. The macro\n\u0027enable_extra_repos\u0027 was added but without any ability to override\nsomething if needed. This change add an ability to override\n\u0027repos.yaml\u0027 defaults.\n\nChange-Id: Idc404d70df06543342dbb0a2c370231ae9c41762\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/299ca00d56fc4fef632b7b8c668b998c0b11f84f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/299ca00d56fc4fef632b7b8c668b998c0b11f84f"}]},"branch":"refs/heads/master"},"797e8b93f99abcae19acfd6bfcb2494dbff05076":{"kind":"REWORK","_number":2,"created":"2022-05-19 09:29:42.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/72/842472/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/72/842472/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/72/842472/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/72/842472/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/72/842472/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/72/842472/2"}}},"commit":{"parents":[{"commit":"b888f68daf6cb3ed9782fe85045858b6a04d4911","subject":"Fix local sources of git repositories","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b888f68daf6cb3ed9782fe85045858b6a04d4911"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-03-27 21:32:42.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-05-19 09:29:32.000000000","tz":180},"subject":"Add ability to override repos","message":"Add ability to override repos\n\nSince Icf2a8397a8349e0fe849d88d160409fd234480a9 the way of handling\nexternal repositories has changed, all repos are moved to the file\n\u0027kolla/kolla/template/repos.yaml\u0027 which is the part of the Python\ncode instead of configuration files in \u0027docker/\u0027 directory. The macro\n\u0027enable_extra_repos\u0027 was added but without any ability to override\nsomething if needed. This change add an ability to override\n\u0027repos.yaml\u0027 defaults.\n\nChange-Id: Idc404d70df06543342dbb0a2c370231ae9c41762\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/797e8b93f99abcae19acfd6bfcb2494dbff05076"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/797e8b93f99abcae19acfd6bfcb2494dbff05076"}]},"branch":"refs/heads/master"},"22d6bae5077b286b3d1e9de377e5eecfdeb9db8f":{"kind":"REWORK","_number":3,"created":"2022-05-25 14:37:06.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/72/842472/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/72/842472/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/72/842472/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/72/842472/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/72/842472/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/72/842472/3"}}},"commit":{"parents":[{"commit":"576aebec8929514490549f6ef1f4ba16192da11a","subject":"Replace Certbot with Lego for Let\u0027s Encrypt container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/576aebec8929514490549f6ef1f4ba16192da11a"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-03-27 21:32:42.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-05-25 14:36:38.000000000","tz":180},"subject":"Add ability to override repos","message":"Add ability to override repos\n\nSince Icf2a8397a8349e0fe849d88d160409fd234480a9 the way of handling\nexternal repositories has changed, all repos are moved to the file\n\u0027kolla/kolla/template/repos.yaml\u0027 which is the part of the Python\ncode instead of configuration files in \u0027docker/\u0027 directory. The macro\n\u0027enable_extra_repos\u0027 was added but without any ability to override\nsomething if needed. This change add an ability to override\n\u0027repos.yaml\u0027 defaults.\n\nChange-Id: Idc404d70df06543342dbb0a2c370231ae9c41762\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/22d6bae5077b286b3d1e9de377e5eecfdeb9db8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/22d6bae5077b286b3d1e9de377e5eecfdeb9db8f"}]},"branch":"refs/heads/master"},"d5f1ce570c830eef892ed92ec514cbd25412e547":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-05-26 08:33:02.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/72/842472/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/72/842472/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/72/842472/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/72/842472/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/72/842472/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/72/842472/4"}}},"commit":{"parents":[{"commit":"eb4b5420f2ed94c4ef7fe2777dd33d1db7845b30","subject":"Replace Certbot with Lego for Let\u0027s Encrypt container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eb4b5420f2ed94c4ef7fe2777dd33d1db7845b30"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-03-27 21:32:42.000000000","tz":180},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-05-26 08:33:02.000000000","tz":0},"subject":"Add ability to override repos","message":"Add ability to override repos\n\nSince Icf2a8397a8349e0fe849d88d160409fd234480a9 the way of handling\nexternal repositories has changed, all repos are moved to the file\n\u0027kolla/kolla/template/repos.yaml\u0027 which is the part of the Python\ncode instead of configuration files in \u0027docker/\u0027 directory. The macro\n\u0027enable_extra_repos\u0027 was added but without any ability to override\nsomething if needed. This change add an ability to override\n\u0027repos.yaml\u0027 defaults.\n\nChange-Id: Idc404d70df06543342dbb0a2c370231ae9c41762\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d5f1ce570c830eef892ed92ec514cbd25412e547"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d5f1ce570c830eef892ed92ec514cbd25412e547"}]},"branch":"refs/heads/master","description":"Rebase"},"865aee6ce5755f30ec3b076b3e235e1da0446a5e":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-05-27 14:07:46.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/72/842472/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/72/842472/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/72/842472/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/72/842472/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/72/842472/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/72/842472/5"}}},"commit":{"parents":[{"commit":"5b0bb37727201a081baa4e1ae3f96ede35517458","subject":"Replace Certbot with Lego for Let\u0027s Encrypt container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5b0bb37727201a081baa4e1ae3f96ede35517458"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-03-27 21:32:42.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-05-27 14:07:46.000000000","tz":0},"subject":"Add ability to override repos","message":"Add ability to override repos\n\nSince Icf2a8397a8349e0fe849d88d160409fd234480a9 the way of handling\nexternal repositories has changed, all repos are moved to the file\n\u0027kolla/kolla/template/repos.yaml\u0027 which is the part of the Python\ncode instead of configuration files in \u0027docker/\u0027 directory. The macro\n\u0027enable_extra_repos\u0027 was added but without any ability to override\nsomething if needed. This change add an ability to override\n\u0027repos.yaml\u0027 defaults.\n\nChange-Id: Idc404d70df06543342dbb0a2c370231ae9c41762\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/865aee6ce5755f30ec3b076b3e235e1da0446a5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/865aee6ce5755f30ec3b076b3e235e1da0446a5e"}]},"branch":"refs/heads/master","description":"Rebase"},"bd9788bad81dadbb04005c409fed63817116ef0b":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2022-05-27 14:10:20.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/72/842472/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/72/842472/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/72/842472/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/72/842472/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/72/842472/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/72/842472/6"}}},"commit":{"parents":[{"commit":"41b056a107c12e7e02ac2d9a8dd7f620e0dc76a0","subject":"Replace Certbot with Lego for Let\u0027s Encrypt container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/41b056a107c12e7e02ac2d9a8dd7f620e0dc76a0"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-03-27 21:32:42.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-05-27 14:10:20.000000000","tz":0},"subject":"Add ability to override repos","message":"Add ability to override repos\n\nSince Icf2a8397a8349e0fe849d88d160409fd234480a9 the way of handling\nexternal repositories has changed, all repos are moved to the file\n\u0027kolla/kolla/template/repos.yaml\u0027 which is the part of the Python\ncode instead of configuration files in \u0027docker/\u0027 directory. The macro\n\u0027enable_extra_repos\u0027 was added but without any ability to override\nsomething if needed. This change add an ability to override\n\u0027repos.yaml\u0027 defaults.\n\nChange-Id: Idc404d70df06543342dbb0a2c370231ae9c41762\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bd9788bad81dadbb04005c409fed63817116ef0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bd9788bad81dadbb04005c409fed63817116ef0b"}]},"branch":"refs/heads/master","description":"Rebase"},"6baaefdcef9cc080528b218b3f165c84d92fd476":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2022-05-27 19:34:54.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/72/842472/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/72/842472/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/72/842472/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/72/842472/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/72/842472/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/72/842472/7"}}},"commit":{"parents":[{"commit":"bce0f591ddcf25e4f2e5d70152b57cb17a19003f","subject":"Replace Certbot with Lego for Let\u0027s Encrypt container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bce0f591ddcf25e4f2e5d70152b57cb17a19003f"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-03-27 21:32:42.000000000","tz":180},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-05-27 19:34:54.000000000","tz":0},"subject":"Add ability to override repos","message":"Add ability to override repos\n\nSince Icf2a8397a8349e0fe849d88d160409fd234480a9 the way of handling\nexternal repositories has changed, all repos are moved to the file\n\u0027kolla/kolla/template/repos.yaml\u0027 which is the part of the Python\ncode instead of configuration files in \u0027docker/\u0027 directory. The macro\n\u0027enable_extra_repos\u0027 was added but without any ability to override\nsomething if needed. This change add an ability to override\n\u0027repos.yaml\u0027 defaults.\n\nChange-Id: Idc404d70df06543342dbb0a2c370231ae9c41762\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6baaefdcef9cc080528b218b3f165c84d92fd476"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6baaefdcef9cc080528b218b3f165c84d92fd476"}]},"branch":"refs/heads/master","description":"Rebase"},"aabcc9ffbb9d35ad9b5d5970e6636285140d2f14":{"kind":"NO_CHANGE","_number":8,"created":"2022-05-30 08:12:09.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/72/842472/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/72/842472/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/72/842472/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/72/842472/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/72/842472/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/72/842472/8"}}},"commit":{"parents":[{"commit":"d424a63d6049d8d241ab1294aea035d4b187a111","subject":"Replace Certbot with Lego for Let\u0027s Encrypt container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d424a63d6049d8d241ab1294aea035d4b187a111"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-03-27 21:32:42.000000000","tz":180},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-05-30 08:12:09.000000000","tz":0},"subject":"Add ability to override repos","message":"Add ability to override repos\n\nSince Icf2a8397a8349e0fe849d88d160409fd234480a9 the way of handling\nexternal repositories has changed, all repos are moved to the file\n\u0027kolla/kolla/template/repos.yaml\u0027 which is the part of the Python\ncode instead of configuration files in \u0027docker/\u0027 directory. The macro\n\u0027enable_extra_repos\u0027 was added but without any ability to override\nsomething if needed. This change add an ability to override\n\u0027repos.yaml\u0027 defaults.\n\nChange-Id: Idc404d70df06543342dbb0a2c370231ae9c41762\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aabcc9ffbb9d35ad9b5d5970e6636285140d2f14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aabcc9ffbb9d35ad9b5d5970e6636285140d2f14"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
