)]}'
{"id":"openstack%2Fkolla~842847","triplet_id":"openstack%2Fkolla~master~I2cc79c476d089977b57006dd4ce2e9a8c6206aa1","project":"openstack/kolla","branch":"master","topic":"fix-bifrost-u-c","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-05-21 17:41:58.000000000","reason":"Dr. Jens Harbott (frickler) replied on the change"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-05-22 20:39:56.000000000","reason":"Change was submitted"},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-05-22 20:39:56.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-05-22 20:39:56.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-05-22 20:39:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2cc79c476d089977b57006dd4ce2e9a8c6206aa1","subject":"[bifrost] Force Bifrost to use the correct u-c","status":"MERGED","created":"2022-05-21 15:47:46.000000000","updated":"2022-05-22 20:41:01.000000000","submitted":"2022-05-22 20:39:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"842847-fix-bifrost-u-c","meta_rev_id":"284df3c3e0c3a9fa770524c1613fa9842f4bd1af","_number":842847,"virtual_id_number":842847,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-22 20:39:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2022-05-21 17:41:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2022-05-21 17:39:43.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2022-05-21 17:39:43.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2022-05-21 17:39:43.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2022-05-21 17:39:43.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-21 15:50:47.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2022-05-21 15:50:47.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-05-21 15:50:47.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-05-21 15:50:47.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2022-05-21 16:33:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-21 16:54:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"49d4660da6674d1a958c5b131c6530db13031765","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-21 15:47:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"59f6277ec03d4c394f2d0b447a3cba69184170c1","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-21 15:50:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cc461a7650266693c961c3e9101090dcc1cd1a74","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-21 15:54:51.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"a69012895d999c743387665a073275c93de5af9e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-21 15:55:00.000000000","message":"Patch Set 1: Backport-Candidate+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8d82375d8675dd4718354e640f757ef440695c07","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-21 16:33:04.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-aarch64 https://zuul.opendev.org/t/openstack/build/f8a2609735724319bf7f5cc9dcadf971 : SUCCESS in 38m 54s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/38e96c0385c0414cbccaac185e3ff44a : RETRY_LIMIT in 2m 49s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/cd004b771bd7481b8d04d62913c31d46 : RETRY_LIMIT in 2m 23s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ac5778094cdb44158b19db69c702eaa5 : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/214df58ef1a94d49ab42a6ec44bed50e : SUCCESS in 5m 13s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/d6e5d8beaa1b48efba930be91c1fd8cf : SUCCESS in 42m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"43dbfa810fec5435cff5a2799fc15cabdd44011b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-21 16:54:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/1922e610948b47bea90ff255fdee7982 : FAILURE in 34m 30s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/38fcdd19cff8458b87fa0c7d69ed2e57 : FAILURE in 8m 32s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/9444187e28a347d2950bb8abd753aa98 : SUCCESS in 1h 06m 36s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d0f99520b8654fa7898e402a5fff7286 : FAILURE in 21m 46s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4318f8f3caeb48a68d315a435f2b3516 : FAILURE in 30m 47s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/cee60f5d602a488eb418ccb1e25cc21a : FAILURE in 23m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc9c0a4fe33a40b680ed43e8dcd3a527 : SUCCESS in 3m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a8d794d511e4bf58dfc0b96673859cb : SUCCESS in 3m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/102db234045e48ecb40e8e3766a4a8d1 : SUCCESS in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cf78a464fbe045e0ba948ed5d053825a : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a6eb9c2d4bd4cf29cdfb936fdf56eba : SUCCESS in 6m 06s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/acd2479d38ff434aa9bd6c37aa7027ba : SUCCESS in 22m 33s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/eb9871e11f07436f8aa1ee23744ad843 : SUCCESS in 51m 28s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/bb9f01eeb3634ddc833d890799c3060a : SUCCESS in 1h 00m 22s","accounts_in_message":[],"_revision_number":1},{"id":"9cb361beeb284129b7dee0e950d770be5b09a43e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-21 17:39:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0c784fde5f8b9f11cc028ab428206e3b5ad14c0f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-21 17:39:43.000000000","message":"Patch Set 2: Backport-Candidate+2 Code-Review+2 Workflow+1 Review-Priority+2","accounts_in_message":[],"_revision_number":2},{"id":"dbed8f7fdba985eaabd3c905472d049f3c716080","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-05-21 17:41:58.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"eeaf05402b142e6fd9d411cfb1151c5db35a0cfe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-21 19:18:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/2cb98e111b6444ee93a1c4caa62a0ce9 : SUCCESS in 32m 46s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/b66a26ad61ca4e69aa96c6c1905c6a30 : FAILURE in 9m 15s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/f2ede71373e94950bcd97493fb5b4a26 : SUCCESS in 1h 16m 44s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/cf7f5cba05b4450684401a881fc34ade : FAILURE in 21m 17s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2462239050d34c67a0678244980accfd : FAILURE in 24m 57s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/16ac202e862a46469aa4fc94c0e575a9 : FAILURE in 22m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f71e7ca9a5e3475c8b9b0bbd5026ff1f : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8680633e0307420a92b058d2386426fe : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/70e42609e6a84d75b8721dc52111f19c : SUCCESS in 3m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fb6ddf5bd2c743f89645235d8f3bd2c1 : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33c9f60a979d4fb5851cc5142b90b9fb : SUCCESS in 5m 32s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/23b2a27591be4621bbad9b1509e74efd : SUCCESS in 24m 41s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e81d42028f7f4db7ac5f52a134dff2a5 : SUCCESS in 56m 05s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b1df80adb2d04cfc8ac411d2435d6a52 : SUCCESS in 56m 21s","accounts_in_message":[],"_revision_number":2},{"id":"74419af1cbacea0541cede0f39ba5f016bb36b23","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-21 19:21:58.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-aarch64 https://zuul.opendev.org/t/openstack/build/79bdaefaaa7348ae85804eb27b04a405 : SUCCESS in 1h 38m 59s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/25644827223841939b980fbcbcfe129a : RETRY_LIMIT in 2m 50s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0363db61aef9499a8f72b3eed1b878ee : RETRY_LIMIT in 3m 07s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/01ef888c0736469781f4667009a1ed24 : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f1a585e35ae14a8d88549a779a95bd6b : SUCCESS in 6m 03s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/50afdf8ed821400c87c0d407811cbf8b : SUCCESS in 1h 12m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3c6c152d5acd10ebe5d7cc10a41ca1ccbff1346a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-21 21:26:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f0e7adf4fe5c8bac8751d990cef4edac28a2030e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-21 22:31:52.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/51d77ab4afc845a59a7474e142d6682a : SUCCESS in 32m 17s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/663bc139aeb44af2866874b458c3b9e1 : SUCCESS in 59m 18s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/066829d1a86d438c9412a26755f7fd0d : SUCCESS in 1h 04m 13s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/6a6025698cad48e3b9dffafcfccad970 : FAILURE in 17m 48s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3be43f58c05d4101b74f2ec66bb77e17 : FAILURE in 24m 02s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/2ab32e95ba3c4a11beaaf0ddc68ee211 : FAILURE in 20m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/382aae4f13ff417cb5a64be0316fdb58 : SUCCESS in 3m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c52267b23c6c475bb1f275961c620046 : SUCCESS in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/795d6c670ca3495ca4525a92f7a35981 : SUCCESS in 3m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3ad3c0d7b6a24d2dbf49566eb344013a : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13535827f55642e6aaf3db8e95dd1d7d : SUCCESS in 5m 50s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f50b3c0a5c1e49329cdbd54bc37b299e : SUCCESS in 29m 08s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/798b4c8f342440e7a06822040748096c : SUCCESS in 46m 27s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3066f92da8174b0295e8590076c70ca2 : SUCCESS in 49m 48s","accounts_in_message":[],"_revision_number":2},{"id":"59915952915db970ab58b8a33334500540be5725","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-21 23:02:24.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-aarch64 https://zuul.opendev.org/t/openstack/build/babbae57844a4101bf8727726369021e : SUCCESS in 1h 32m 54s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0720e7aeac684dd7b918a68f6d08e182 : RETRY_LIMIT in 2m 30s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/bfad6eb09b5e487eb8219e174e3f66df : RETRY_LIMIT in 2m 31s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/867cc7318b264f8892477b389513db69 : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/05356da521ac436cb111868f5b121e70 : SUCCESS in 7m 02s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/8ccf893caf5f4161b470ddc9a3180b21 : SUCCESS in 1h 21m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"012634775800a34a5853b456244d35283e63eb00","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-22 07:57:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7b4242ebaa8063fb7ef6b49ac6742354a1d9a67a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-22 09:28:50.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/72e4e882c7784551a4e00377689d50f6 : SUCCESS in 31m 54s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/dcf67d4bd3254a3388ad987a1fc05520 : SUCCESS in 1h 15m 23s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/8c1248621d2c4795b98fb6d0b83c8884 : SUCCESS in 1h 30m 01s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b414e88db7084d04bc2ed7e5b7b03329 : FAILURE in 27m 44s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/914055d5dcce433ca6f0bbc417f17e69 : FAILURE in 19m 44s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/27f7698a612347538a2fdd74a48509ed : FAILURE in 25m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33cb9a2bb75c4608b62b5db225ec5d6f : SUCCESS in 3m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f5e9e9c9437045e09c02a9f3087c4a4f : SUCCESS in 3m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/04eeea4a8da845db892111e171d95e5b : SUCCESS in 4m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ccdf36141c0d45c899312969d49ca67c : SUCCESS in 3m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc1b5ec36b1c4eccbf9118f10016c225 : SUCCESS in 5m 56s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d3a551f7931b40b88447a554213724f0 : SUCCESS in 24m 12s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4887dd29ae9d4400974fb6c23e345bed : SUCCESS in 1h 06m 35s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a44cfbf8afa8455fb4ea1f06454556b4 : SUCCESS in 1h 04m 41s","accounts_in_message":[],"_revision_number":2},{"id":"f303f21b72dec5464f55419b0501e1b99c1921e8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-22 09:51:06.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-aarch64 https://zuul.opendev.org/t/openstack/build/68bb1dd5571646f4a2abdd4f84c407cf : SUCCESS in 1h 50m 34s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b1b6ab1e64d84db28f59c05963d4c4ab : RETRY_LIMIT in 1m 56s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/70d60a4c94d94b7d9e6453bd38fad053 : RETRY_LIMIT in 3m 40s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bdf7504a44074b2294eac33bba9388b9 : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bd221f1c9c9d4918a07e3c1e4f11f248 : SUCCESS in 5m 55s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/a2f90206dac94dfe9f2f57762620b613 : SUCCESS in 1h 29m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0dca8f18af9dbd2d00c15a540b3d334633c1dd7c","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-05-22 17:50:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"655fd5e2581b640a684bb601fbc22cba1f239200","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-22 19:12:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/167bb8ab4bfa4d6c9efc36b6d8ed67be : SUCCESS in 29m 58s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/d6ccdeb5a6454e7ba56703b869df0ff8 : SUCCESS in 50m 34s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/0d46ce0e87e64d5397e81bf6d351b4c8 : SUCCESS in 1h 02m 07s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f1188fc3ad684970abef0dfa1103683a : SUCCESS in 20m 04s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/02c81ac46d674e24a1cfb94522968f12 : SUCCESS in 44m 50s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/395a67a311a24c609e5e901094827c2e : SUCCESS in 52m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0dd7f25ae5a04aff829115ab61ed9f78 : SUCCESS in 2m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb7dbc7f1f2e490abd454e648f665c3f : SUCCESS in 3m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e2a4315771a748e5a72e7516b6916b5f : SUCCESS in 4m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ec467516e328480eb34e30c10090b508 : SUCCESS in 3m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26cad6bdddbe45fe9a77a54abeecc716 : SUCCESS in 5m 05s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/4a54a971f8af49528a8fdd2daae0ea2e : SUCCESS in 20m 11s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9708d3be7f8445da8ef4690dfd296454 : SUCCESS in 39m 37s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/190d19c856c7447dbe9eac18e42838fc : SUCCESS in 45m 16s","accounts_in_message":[],"_revision_number":2},{"id":"88904794bfaa8f1a43c1d69b63df25cfe76af130","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-22 19:13:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6fd6725642bf315d0b4004f286082c05e8a86960","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-22 19:21:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-aarch64 https://zuul.opendev.org/t/openstack/build/78c2021f25914cba9a419100bb593acf : SUCCESS in 31m 13s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/79bd4fcaf689421cb3c3c8d1be47b112 : SUCCESS in 1h 24m 29s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8d2ab614021c4e85ab9952e8958e9dcb : FAILURE in 1h 28m 22s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ca0771e24c99444aa08727b2696bb64a : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/682d925b6bd84e9d81aa7bf6f4a9b017 : SUCCESS in 3m 50s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/0568ffe4815d45448a832fee3f1b91d4 : SUCCESS in 1h 06m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8682e2b77f3832e716c1d1e8f51a4419c64504be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-22 20:39:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/244584dfa27548e2a3149458c34a4e06 : SUCCESS in 30m 02s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/9a65c1c2c01a468aaeda3a5b45fe4f7b : SUCCESS in 1h 04m 54s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/031004fece344e5baa73a4b41f86a802 : SUCCESS in 1h 23m 41s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f0fafceddecc47638270f7c1da266aa8 : SUCCESS in 24m 30s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d414e48c059a41b29270e172474921cd : SUCCESS in 1h 01m 52s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/c98499b96f724f57934373c6b67aac70 : SUCCESS in 1h 10m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4544130c6d7644b385ad5f73d397b6ca : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d16a51fe945e465e8e11f3c26803e58e : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c6d07a5f7c764220af5f0da9eb78d6ae : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35d02d1e7b1e454a89a315b27f6b48b6 : SUCCESS in 6m 15s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b62df1b2be9d4770a992b88da93e0eaa : SUCCESS in 22m 23s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2e809055e31d46908262f660752c8422 : SUCCESS in 57m 30s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4432da61c400499eb4a61d859f0bc426 : SUCCESS in 1h 02m 35s","accounts_in_message":[],"_revision_number":2},{"id":"9522b60877af01869942f76f592375accff3301c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-22 20:39:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"284df3c3e0c3a9fa770524c1613fa9842f4bd1af","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-22 20:41:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6090c846a08486ea5d8f77d36519257 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"633c89b423c6d83609120faa87d4b4d63e59ccfa","revisions":{"554bfd3e4e924d42c01deb0ffe0822cef893be8a":{"kind":"REWORK","_number":1,"created":"2022-05-21 15:47:46.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/47/842847/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/842847/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/842847/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/842847/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/842847/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/842847/1"}}},"commit":{"parents":[{"commit":"b888f68daf6cb3ed9782fe85045858b6a04d4911","subject":"Fix local sources of git repositories","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b888f68daf6cb3ed9782fe85045858b6a04d4911"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-05-21 15:45:37.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-05-21 15:47:12.000000000","tz":120},"subject":"[bifrost] Force Bifrost to use the correct u-c","message":"[bifrost] Force Bifrost to use the correct u-c\n\nFollowing up on [1], Bifrost also needs to use correct u-c and,\nsurprise-surprise, this is set via a third variable, this time\nenvironment one for the bash script. Oh well. Let\u0027s do it.\n\n[1] https://review.opendev.org/c/openstack/kolla/+/842276\n\nChange-Id: I2cc79c476d089977b57006dd4ce2e9a8c6206aa1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/554bfd3e4e924d42c01deb0ffe0822cef893be8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/554bfd3e4e924d42c01deb0ffe0822cef893be8a"}]},"branch":"refs/heads/master"},"633c89b423c6d83609120faa87d4b4d63e59ccfa":{"kind":"REWORK","_number":2,"created":"2022-05-21 17:39:35.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/47/842847/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/842847/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/842847/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/842847/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/842847/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/842847/2"}}},"commit":{"parents":[{"commit":"b888f68daf6cb3ed9782fe85045858b6a04d4911","subject":"Fix local sources of git repositories","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b888f68daf6cb3ed9782fe85045858b6a04d4911"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-05-21 15:45:37.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-05-21 17:38:23.000000000","tz":120},"subject":"[bifrost] Force Bifrost to use the correct u-c","message":"[bifrost] Force Bifrost to use the correct u-c\n\nFollowing up on [1], Bifrost also needs to use correct u-c and,\nsurprise-surprise, this is set via a third variable, this time\nenvironment one for the bash script. Oh well. Let\u0027s do it.\nThis has the added benefit of not requiring additional\ndownload of u-c as it reuses the already existing one.\n\n[1] https://review.opendev.org/c/openstack/kolla/+/842276\n\nChange-Id: I2cc79c476d089977b57006dd4ce2e9a8c6206aa1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/633c89b423c6d83609120faa87d4b4d63e59ccfa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/633c89b423c6d83609120faa87d4b4d63e59ccfa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
