)]}'
{"id":"openstack%2Fkolla~843206","triplet_id":"openstack%2Fkolla~stable%2Fvictoria~I793e35ef347c9d44b4e870cd3ca1bd459e557623","project":"openstack/kolla","branch":"stable/victoria","topic":"feature/bump_upstream_alertmanager-stable/victoria","attention_set":{},"removed_from_attention_set":{"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2022-06-23 18:43:58.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I793e35ef347c9d44b4e870cd3ca1bd459e557623","subject":"Bump up Alertmanager version","status":"ABANDONED","created":"2022-05-25 11:37:31.000000000","updated":"2022-06-23 18:43:58.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"cherry_pick_of_change":839367,"cherry_pick_of_patch_set":1,"meta_rev_id":"59836cb98b28137d481bbd8b36165f865fd131ee","_number":843206,"virtual_id_number":843206,"owner":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-05-25 12:53:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2022-05-25 12:53:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0f29fe0aca2972c17d7f149380e511d68f6559e3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2022-05-25 11:37:31.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* docker/prometheus/prometheus-alertmanager/Dockerfile.j2","accounts_in_message":[],"_revision_number":1},{"id":"83a2e3015e396a23e824fe0dd1a0cf446977757a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-25 12:53:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/791a1f2844164ae4a9d1ad7b7b9542f4 : FAILURE in 33m 51s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/d6f6e0c92c9e44dc9af6866e3e1e72a2 : FAILURE in 38m 30s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/1242f3cec77c47a3ab20190bb59f894d : SUCCESS in 57m 41s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/feb81b10e002407788ef014189e4f95d : SUCCESS in 1h 04m 31s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/3bdd0fa495e843378da8816053982722 : SUCCESS in 1h 04m 50s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/7c4d842b10f2460b95a49571e83fa2fc : FAILURE in 21m 12s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/93ee0bd3e6e44978bf035f9b1ad51b92 : FAILURE in 19m 33s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f237230156c64081a647f6fe15fb9bb7 : SUCCESS in 56m 17s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff1556edbbfc415faf7dba58ecb4f713 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3af61a3f21204510b2b14f9bcf3fc6da : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/515149c870ba4a06a4da244ce2cc8f4f : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/385aa59c96b5409da0697fdbd1a6464c : SUCCESS in 4m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f422110a60c4ff381b478d99cca0eac : SUCCESS in 4m 32s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/51dacc565f3f43679999a7c7c8088a65 : FAILURE in 33m 13s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b744ec36351a441fb0dc5aedca3d02d4 : FAILURE in 32m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/eeaebf33f122410facec3178b03c0795 : SUCCESS in 51m 14s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a336f7960f4d4df2bb0ea24d8f6b3020 : SUCCESS in 50m 35s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5c1be74502a64f28a22cc59d44837fb6 : SUCCESS in 44m 56s","accounts_in_message":[],"_revision_number":1},{"id":"d25d2e5ca1ab712ac1425bd0741692327585494e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-25 14:43:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/b44a7dc9beea4e1e99be8f6c2f778e67 : FAILURE in 1h 59m 21s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/76a84f2fd8cd4e2286e45f2a056be1c8 : FAILURE in 1h 23m 06s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/97de77b3873e47eca7f07b583c262096 : SUCCESS in 3h 02m 52s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/ced5a854be7d4c96beb6289ff4a96487 : FAILURE in 1h 30m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"59836cb98b28137d481bbd8b36165f865fd131ee","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-06-23 18:43:58.000000000","message":"Abandoned\n\nabandoning due to no progress","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1e8c0ff4cf5acaa45f23f39fdc5fbc6cf56f00eb","revisions":{"1e8c0ff4cf5acaa45f23f39fdc5fbc6cf56f00eb":{"kind":"REWORK","_number":1,"created":"2022-05-25 11:37:31.000000000","uploader":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"ref":"refs/changes/06/843206/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/843206/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/843206/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/843206/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/843206/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/843206/1"}}},"commit":{"parents":[{"commit":"b716b5a5a761fa4a5cb826be2c38389727660715","subject":"Fix local sources of git repositories","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b716b5a5a761fa4a5cb826be2c38389727660715"}]}],"author":{"name":"Sven Kieske","email":"s.kieske@mittwald.de","date":"2022-04-26 12:01:54.000000000","tz":120},"committer":{"name":"Sven Kieske","email":"s.kieske@mittwald.de","date":"2022-05-25 11:37:31.000000000","tz":0},"subject":"Bump up Alertmanager version","message":"Bump up Alertmanager version\n\nChange-Id: I793e35ef347c9d44b4e870cd3ca1bd459e557623\n(cherry picked from commit d41c19fbcaacc2d8030110fb72ea0612b5f7ab87)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1e8c0ff4cf5acaa45f23f39fdc5fbc6cf56f00eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1e8c0ff4cf5acaa45f23f39fdc5fbc6cf56f00eb"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
