)]}'
{"id":"openstack%2Fkolla~843417","triplet_id":"openstack%2Fkolla~master~I6bc8b489a810849744c8bead6b9e350c5d3e36b7","project":"openstack/kolla","branch":"master","topic":"new-naming-scheme","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-05-30 11:26:57.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-05-30 09:53:54.000000000","reason":"Michal Nasiadka replied on the change"},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-05-30 10:04:50.000000000","reason":"Marcin Juszkiewicz replied on the change"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2022-05-26 08:50:26.000000000","reason":"Maksim Malchuk replied on the change"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-05-30 11:26:57.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-05-30 11:26:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6bc8b489a810849744c8bead6b9e350c5d3e36b7","subject":"Use the new image naming scheme by default","status":"MERGED","created":"2022-05-26 08:31:11.000000000","updated":"2022-06-09 12:16:03.000000000","submitted":"2022-05-30 11:26:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"843417-new-naming-scheme","meta_rev_id":"c99f31eb4db65ecb5eb170ed86d60a155f0c6084","_number":843417,"virtual_id_number":843417,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-30 11:26:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-05-30 09:53:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-05-30 10:04:50.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2022-05-30 10:48:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2022-05-30 10:04:50.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2022-06-09 12:16:03.000000000","post_submit":true,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-26 08:31:49.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2022-05-26 08:31:49.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2022-05-26 08:31:49.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2022-05-26 08:31:49.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-05-26 08:31:49.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-05-26 09:43:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d300093119fae86680a8ae05923967349d067bee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-26 08:31:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5c3bcde69d73b6e9337a68407c2134809494680f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-26 08:31:49.000000000","message":"Patch Set 1: Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ad24c97fe28677268e70d3245bb90c4e638568b9","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-05-26 08:45:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"37b33880c435d10b89eeb35dc6eaf4ccd14534fa","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-05-26 08:50:26.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5752359ea706437baba3c3b43420def8639c69ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-26 09:43:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/39e098ddc4ea43b0a799ec9b151a07ee : SUCCESS in 24m 26s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/523e9ca18f6647f8865a4a836e7e5482 : SUCCESS in 1h 01m 44s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/272796e4e8fc478fbcd384944ce8ebe1 : SUCCESS in 1h 05m 28s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/0b09a3fa4fa84a0392451b2045d656c6 : SUCCESS in 17m 51s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8f215bd47e5643afb47855253408e8ab : SUCCESS in 1h 00m 12s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/3164d99bb9b845fa8de5d2d64eb69d93 : SUCCESS in 58m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/507bbef46e8e4471bb4802919381eac6 : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc599d2be5814669bde84ab76b0120ff : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4441f6c9f36e4fefb99bf278e7a81063 : SUCCESS in 4m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9fcb62ea5e1f4b0b906ceb2adafda84f : SUCCESS in 3m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1dbb7554490428995003a126576fcb1 : SUCCESS in 5m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/632ee1ba66e44cf895a44fbebdece2fc : SUCCESS in 3m 55s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/28f0b228deb74505865756081890ea55 : FAILURE in 18m 28s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4342947051a54b00836ede072d8c234b : FAILURE in 19m 14s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0a253066bfa140fb9c9a0d1d16512fd7 : FAILURE in 17m 39s","accounts_in_message":[],"_revision_number":1},{"id":"e679b9e33f89243f883a037e494bbd4668c986d4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-26 10:25:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-aarch64 https://zuul.opendev.org/t/openstack/build/b47eb9b072ba4dfbaf0697c10d03a273 : SUCCESS in 1h 51m 25s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7e8725407d604adcb9b19f93157e1406 : SUCCESS in 1h 24m 30s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1e05941d7d0a470cb7e9ae5357d71435 : FAILURE in 1h 00m 43s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6479d87db2d24c22923da0dbbcfe8bfd : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e3b0de3dd8734b0ca6dab52471a8a8b0 : SUCCESS in 5m 55s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/8b80b1accc5e415a921de73e74f19398 : FAILURE in 1h 11m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b808571853babc02ab5a0b3ea42b51ec27f94999","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-26 11:20:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"432f32d0620b355ad7edff10ab6f185898460aee","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-26 11:20:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ed868b5ac333e2d4f3d62e715829ff6c66428baf","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-05-27 14:09:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f24a317b94158660e136f569ac76ef6e840a7d88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 15:49:48.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/fb1899c68dab45e4a4a5d36d63f0d5cc : SUCCESS in 25m 01s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/5a0b74def47646a89e45d74290e6275e : SUCCESS in 1h 20m 02s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/4d91cd94aa3c4f84bc4de5299b69715e : SUCCESS in 1h 33m 32s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8da249e723234c25a6341b914a8748ed : SUCCESS in 20m 20s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/58bb437684ac4ea693f3d3b6c1d9bf72 : SUCCESS in 1h 07m 17s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/e7819e305150444db6b873afc49c3c58 : SUCCESS in 1h 12m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35e173dc4d804822902c0a00858d8a45 : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48363e34703a4773af89bbdb5919d12a : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/98e80fcf433349f28a0d6b19153931e1 : SUCCESS in 3m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eab6ec162ce942c280528ba7f5f30798 : SUCCESS in 3m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7765255c686b421d8d9bd1dc8399381b : SUCCESS in 5m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/86e6fd42336841b2ba086205b460a398 : SUCCESS in 3m 47s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/2fe7771595544dd89055e247f0210b58 : FAILURE in 22m 08s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/59b2773c7a47493ca3c620c864a84163 : FAILURE in 17m 57s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/006186596303478b8ce2e5f428afdb6e : FAILURE in 17m 47s","accounts_in_message":[],"_revision_number":1},{"id":"d91470c95cd5dbb68f96400ef31d3986987d6a85","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 16:10:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-aarch64 https://zuul.opendev.org/t/openstack/build/0ba9ec49c85643b4885def44a8131a10 : SUCCESS in 33m 58s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1e46363833f44636b202fffaf80da9b4 : SUCCESS in 1h 35m 11s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6747d40c14524116b96701d54b3e615c : FAILURE in 1h 57m 53s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e1376297a2c94b9ebc4c04bd8965be3f : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9b411d0108fd42fbb1dd0acd4a9b363c : SUCCESS in 6m 04s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/dd1343298b944a6abd5141e37cbb6285 : FAILURE in 1h 28m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ad4ec095d88bd808c86cf1f953dd8bac2fa0e186","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-27 19:33:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"22dfc7934ebb4c1b24c389576016caf79a900e0c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-27 19:33:51.000000000","message":"Patch Set 2: Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f2221fb91cc57c217365427614577d163f1cf75d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 20:49:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/6a786afb05924bf6a30c34c107419489 : SUCCESS in 31m 23s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/f0db7c264267409faf4fbc649a1f6a67 : SUCCESS in 1h 09m 06s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/99db5e6f901348c89bef982d2777dcb5 : SUCCESS in 1h 10m 42s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f3acaa9992cf4d1c81997d40af6b97a6 : SUCCESS in 19m 17s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8fc75c6bf3a24afba1c6c8d15fa174b1 : SUCCESS in 51m 18s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/ac1626ee380646438ede03540c4b7e61 : SUCCESS in 1h 05m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24f12e5204e847229b6354887e811131 : SUCCESS in 3m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/175713be3ead4261b0ce7ddf3f453da8 : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6480134830b54ad2bd9c8bc144fdff62 : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f65207e9e3084ef4985c3f135c887340 : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3b877713d984d678d77b2c489137749 : SUCCESS in 5m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a3c2584e0a4421299d8ed6efcfe6ac2 : SUCCESS in 3m 47s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8d1e772a78594c178980af36f2b03fff : SUCCESS in 18m 32s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/fd74684c58a04ef1b0122c19760c56e3 : SUCCESS in 45m 25s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/86ab8b1d7a0e4a8ab35194a6f4f2514b : SUCCESS in 58m 03s","accounts_in_message":[],"_revision_number":2},{"id":"b809092645763d7ff2b2edeec1662d65533f229f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-05-27 20:58:47.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"cbbbf735fc891a88f73fc2641b54f65591a62c9e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 22:12:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-aarch64 https://zuul.opendev.org/t/openstack/build/401dde95d2344d4da39703faf9e92704 : SUCCESS in 2h 24m 52s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/af02bf89ef404daa99bf3a2c0034453f : SUCCESS in 22m 44s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f1994e78af2342ac8d7c389e329d6950 : FAILURE in 1h 58m 32s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1e8f1e09052a4fa4ae6d41b383aaec06 : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/102eb50d2d584a63900c11a9e98f4084 : SUCCESS in 7m 38s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/5b70a34808af4c46aa5ef1a94268f095 : FAILURE in 1h 34m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"80dd80d4dc398bda1b084d2d7dc8ecc599994476","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-29 18:14:01.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"4da863c58d1055b54daf03f13baf3eb7e8cc9624","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-29 19:34:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/455e278d48314ea2bbff75ce0ec0bd50 : SUCCESS in 29m 33s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/e4cfddc555974e56989a0df883accf78 : SUCCESS in 54m 31s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/d928db827c384115aad176eafd0b0dcc : SUCCESS in 1h 18m 07s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d39d99143813489d9f1fd4d210bcf6e7 : SUCCESS in 16m 30s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/979669d72ce943088c3734e2e949758a : SUCCESS in 48m 41s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/95761440eecc4aacaaa53d2f6ab0ee66 : SUCCESS in 1h 02m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04d301057af34a7cbe333762113ddd9b : SUCCESS in 3m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16762879246445c98d342c350c976755 : SUCCESS in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/713410aa09ae486898a4e26ebc26059f : SUCCESS in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/070fd4ba7a914059807d5e3016a627c2 : SUCCESS in 3m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63437541355a42048781a473bca19b01 : SUCCESS in 5m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b337de1e38614e3e90be6f2c3372d038 : SUCCESS in 3m 30s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/57eb8d89ad454a57bb52beda98e6d34d : SUCCESS in 24m 58s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8338d6790b374b7080837ff7c34fde31 : SUCCESS in 45m 28s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/bf1a855d96b349a886a4afdd4979041e : SUCCESS in 57m 08s","accounts_in_message":[],"_revision_number":3},{"id":"6f5cb217a68f3787c692a8e4a4991ef388d8a083","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-29 20:23:57.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-aarch64 https://zuul.opendev.org/t/openstack/build/708e73a3de09404290110dc51c923a03 : SUCCESS in 2h 07m 33s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0bae9bf935c74b35a3b6f8babd0e19d5 : SUCCESS in 1h 15m 56s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c91e85f900bb4e1f844e6ca6cf15164b : FAILURE in 1h 04m 32s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c5704f7eebdb46f89d5e10247160987b : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1b5f26fb0dd04906aac837988666f3f2 : SUCCESS in 6m 52s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/c4effca04c314c11b5ed8d9313154419 : FAILURE in 1h 32m 14s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fc5982b19b0e9e06ae760575b739c31f81dec755","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-05-30 09:53:54.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"9bd98e0784646626ed498535a8e1c009bfab03ea","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-05-30 10:04:50.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"f43202208f97c6a08d4008583e0f77df7c8558b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 10:05:32.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"779369ef09823fc3a511af6e0496edfdc4991ce0","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-05-30 10:48:18.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"c6ba1cfe4f8f2088a0534d0d6a7e222057504d4e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 11:26:57.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/d51394c91c294ec680ecc6da015c4ee0 : SUCCESS in 28m 54s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/0c5e9712a6e344a58fc938c6842f1d06 : SUCCESS in 1h 15m 50s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/46ec9f4754ad4814820e744eb1154e5a : SUCCESS in 1h 12m 33s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/eeca2eeafb9c4ad1ba531ede44dc9131 : SUCCESS in 19m 39s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/beeb6a1ed7754e7fb21f682fcc5996df : SUCCESS in 58m 42s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/e12810659e02416fab0ce8e21eea7c98 : SUCCESS in 1h 16m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cb92bbbae7e4575ae6ae9a37d33a80e : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a110484b19b04411b60fbbd2985fc327 : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/14ad5fc98c5944d6a865967864d4c3d1 : SUCCESS in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcfec3b226ec4e6db09a268653416409 : SUCCESS in 6m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/90a8bc21bfb640349ff79a0a6f17b4e9 : SUCCESS in 3m 42s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e29ed98afdf848de848d7610500cf460 : SUCCESS in 21m 12s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/623d777e8d464cad90e544f5597a96d1 : SUCCESS in 47m 49s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a5651c5d520f45bdb83ab092e3b04cbe : SUCCESS in 1h 04m 25s","accounts_in_message":[],"_revision_number":3},{"id":"33341d7169b8783995463d6f5ea275353cb60737","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 11:26:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"03b00238e9e0ec98a58bfa517128172617311247","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 11:28:32.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7514f5040ca44d9e947b4db832468c80 : SUCCESS in 1m 02s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/da774ecd65484961b4e7006eae9bff49 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":3},{"id":"28dedc058eadd428fde7440494fd28e148545c0a","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-06-09 12:16:03.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9d5bdbf36d477da38b0dd7d44b3fc6feb9325378","revisions":{"058f2c0296ed6500a3c259fe02232eaa6ae2610d":{"kind":"REWORK","_number":1,"created":"2022-05-26 08:31:11.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/17/843417/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/843417/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/843417/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/843417/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/843417/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/843417/1"}}},"commit":{"parents":[{"commit":"a51ebc9e7fc2163623cb3d9b7540c23ea984c859","subject":"Merge \"[CI] Move queue setting to project level\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a51ebc9e7fc2163623cb3d9b7540c23ea984c859"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-05-26 08:27:53.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-05-26 08:31:03.000000000","tz":120},"subject":"Use the new image naming scheme by default","message":"Use the new image naming scheme by default\n\nAdds a relevant reno to inform users about the change and the\npossibility to customise the prefix.\n\nChange-Id: I6bc8b489a810849744c8bead6b9e350c5d3e36b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/058f2c0296ed6500a3c259fe02232eaa6ae2610d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/058f2c0296ed6500a3c259fe02232eaa6ae2610d"}]},"branch":"refs/heads/master"},"e5fdc8b22804f9cf50c82b8fcc198f327d6bfc66":{"kind":"REWORK","_number":2,"created":"2022-05-27 19:33:24.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/17/843417/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/843417/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/843417/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/843417/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/843417/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/843417/2"}}},"commit":{"parents":[{"commit":"a51ebc9e7fc2163623cb3d9b7540c23ea984c859","subject":"Merge \"[CI] Move queue setting to project level\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a51ebc9e7fc2163623cb3d9b7540c23ea984c859"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-05-26 08:27:53.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-05-27 19:32:22.000000000","tz":120},"subject":"Use the new image naming scheme by default","message":"Use the new image naming scheme by default\n\nAdds a relevant reno to inform users about the change and the\npossibility to customise the prefix.\n\nAdditionally, this patch includes the fix to Ubuntu image builds\nregarding rabbitmq and erlang.\n\nBoth the changes are required to fix master CI.\n\nChange-Id: I6bc8b489a810849744c8bead6b9e350c5d3e36b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5fdc8b22804f9cf50c82b8fcc198f327d6bfc66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5fdc8b22804f9cf50c82b8fcc198f327d6bfc66"}]},"branch":"refs/heads/master"},"9d5bdbf36d477da38b0dd7d44b3fc6feb9325378":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-05-29 18:14:01.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/17/843417/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/843417/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/843417/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/843417/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/843417/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/843417/3"}}},"commit":{"parents":[{"commit":"a51ebc9e7fc2163623cb3d9b7540c23ea984c859","subject":"Merge \"[CI] Move queue setting to project level\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a51ebc9e7fc2163623cb3d9b7540c23ea984c859"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-05-26 08:27:53.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-05-29 18:14:01.000000000","tz":0},"subject":"Use the new image naming scheme by default","message":"Use the new image naming scheme by default\n\nAdds a relevant reno to inform users about the change and the\npossibility to customise the prefix.\n\nAdditionally, this patch includes the fix to Ubuntu image builds\nregarding rabbitmq and erlang.\n\nBoth the changes are required to fix master CI.\n\nChange-Id: I6bc8b489a810849744c8bead6b9e350c5d3e36b7\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/843752\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9d5bdbf36d477da38b0dd7d44b3fc6feb9325378"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9d5bdbf36d477da38b0dd7d44b3fc6feb9325378"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
