)]}'
{"id":"openstack%2Fkolla~843637","triplet_id":"openstack%2Fkolla~stable%2Fvictoria~Ic4b6171789c3d360317599310492771bd78828ec","project":"openstack/kolla","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2022-05-30 14:45:41.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-05-30 10:09:06.000000000","reason":"Radosław Piliszek replied on the change"}},"hashtags":[],"change_id":"Ic4b6171789c3d360317599310492771bd78828ec","subject":"masakari: add Cyrus SASL packages to monitors image","status":"MERGED","created":"2022-05-27 17:52:15.000000000","updated":"2022-05-30 14:47:37.000000000","submitted":"2022-05-30 14:45:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"843637","cherry_pick_of_change":841657,"cherry_pick_of_patch_set":1,"meta_rev_id":"72c69413784ed6e10222d5d880c377c13b8ac86b","_number":843637,"virtual_id_number":843637,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-30 14:45:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2022-05-27 21:00:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2022-05-27 17:53:27.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2022-05-27 17:53:27.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}],"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-27 17:52:15.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2022-05-27 17:53:27.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"CC"},{"updated":"2022-05-27 17:53:43.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"CC"},{"updated":"2022-05-27 20:13:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-27 20:59:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-27 21:00:12.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"c4f81c2cbdca09f03b7f6014d085f18c66070ad4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-27 17:52:15.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.\n\nThe following files contain Git conflicts:\n* docker/masakari/masakari-monitors/Dockerfile.j2","accounts_in_message":[],"_revision_number":1},{"id":"6e1767b32b4487c5861e7ad4ce2ceaf0c38c296e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-27 17:52:50.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"72eab268badd446393077e239fb1c49524024475","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-27 17:53:27.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"59e7fade1c100bb96cc5d2e4090d64b38b4463e6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 20:13:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/9b649cfee60b49cd85754a560eb2975f : SUCCESS in 38m 28s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1d6dd2bb4bc74b019433e7eb3b339fb8 : SUCCESS in 1h 03m 06s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/62147cb373d34af88948229e6bd637e8 : SUCCESS in 2h 18m 11s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/e606535ab0e543c4b644e2fdcb3d79dd : FAILURE in 26m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b98e4042112f36476a08d5f3e1970e7fc6dc95c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 20:59:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/887e7f388d6d4dec950e51846f16ff41 : TIMED_OUT in 3h 01m 34s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/24e705413b47459ba6b3f868e9374999 : SUCCESS in 33m 51s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/d26dfbd6b9e344259db8df3918e2ff8f : SUCCESS in 1h 07m 48s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/28772ad510e94210aedcfc25813977ae : SUCCESS in 1h 07m 37s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/806d2e08b1f44be29380da8da3be1fb9 : SUCCESS in 1h 19m 05s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1202f3d63b334126b9112351e93b0990 : SUCCESS in 17m 54s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/47a4bfd0c18b460ab2f4086e5f370a96 : SUCCESS in 18m 11s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8beddc84f2a246a1991ae1eee14bd6d7 : SUCCESS in 52m 22s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c94b14a02a274c23b22579b2a56ae244 : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9d416b709de485a907fb8f8d6eebb91 : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a6799ec36198450bb267e9333a64c167 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09345c5f81034828b205c1fa6f3cb5c0 : SUCCESS in 5m 59s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b312170aef1a4e99a1bd410b6928609d : FAILURE in 30m 55s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a63799e04f5b4e18adc0434a2093929d : FAILURE in 32m 17s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ff58d72848774321ae58413f08471bee : FAILURE in 15m 58s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/55e31f48ac4e41a4ae2f0cdb0294be66 : FAILURE in 18m 58s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c5af2994413542469d490f9aa3adee05 : FAILURE in 15m 31s","accounts_in_message":[],"_revision_number":2},{"id":"cd0c71bcac136f6ee6a6f28603900cf0b0b9baaa","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-05-27 21:00:12.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"1fe986eb17dab132950d392032d824f2f1295ee5","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-05-27 21:00:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e88a662f880b328761ccbdbdf868c297e1a52487","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 22:15:50.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/2c14432246c845dc901b0dc3bf00373b : SUCCESS in 32m 48s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/4755107e8f6b4ecbba004aad55e3c6ff : SUCCESS in 34m 43s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/c0882d0c9efc45c5a45b164e0f6fa1bd : SUCCESS in 56m 15s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/4561315fe73440a28304ff03a3727c8b : SUCCESS in 55m 48s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/6dc7d43eb3e54610b616e8dac420d219 : SUCCESS in 57m 45s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f63121c5c9164f3d98bc036963d5c984 : SUCCESS in 17m 26s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/25f87b4baf2f41298d21ac5c3ad3fe9d : SUCCESS in 16m 50s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8e2508aba548442dbcb02641e4846973 : SUCCESS in 43m 44s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bcea68dd02a4f1fab4b483b1741087c : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d83f3de738049ea99ffbd1a0df237d1 : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9bd4565a035e437aa3b65798535f8cfb : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e5c83ad9aa94126a0785b6f97a3c719 : SUCCESS in 5m 15s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d4eecbec157e4f27a7e1f5c090f872f1 : FAILURE in 33m 44s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/61e82a09c5fc422ba498528e2cbd518a : FAILURE in 34m 53s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0c5583dfed0d4bc096f2971c63684f4b : FAILURE in 15m 26s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d2b1be2e76d84e66956dc5259f9e3b37 : FAILURE in 16m 28s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/75f43be2977a49ed9b395ce329b83b96 : FAILURE in 16m 29s","accounts_in_message":[],"_revision_number":2},{"id":"2e35713b15adec3148b746e2d250c55d34bf7a4e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-28 04:15:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/c8508b80080549b8bafa8250b182e629 : SUCCESS in 2h 05m 10s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b201d81eaf9e44dfb33275a22f1c3abf : SUCCESS in 2h 10m 47s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/81a6cd5c16b249b191ab08ea30887525 : SUCCESS in 2h 01m 32s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/6789d0cc8c084e05b5cbdca0bb42eebe : TIMED_OUT in 6h 01m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"57e5efbe411d7d585072053dd69b702fafe43da4","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-28 11:58:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ca733110626deeb211e39e04a9d485d74b9b5e77","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-28 12:03:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5ea9b29fb3067d5a711377aa511990da5f389023","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-28 13:14:44.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/5051ff5307fa4c9f84b95a0f8a9c4fdc : FAILURE in 36m 35s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/55f36f2f9baf4baabc033c4f6a6b267f : FAILURE in 32m 10s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/90ce911c734e4227b11ff760576b1e22 : FAILURE in 24m 19s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/41be83908f344cdd8988050bf03f348e : FAILURE in 27m 50s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/fcfbd985fb324176a2f8bb4ec4b6b526 : FAILURE in 26m 38s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/fe3452859a2e481cbf3d76df4cdd525f : SUCCESS in 20m 34s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/01034c826a0543ffbcce4af24803d77c : SUCCESS in 20m 41s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/45360d86f09c41a396e41b8e57fcf73c : SUCCESS in 52m 34s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cee18685c4114559acc58ba950b42ece : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/863779ef620f4ac496dae03f6f77d4e4 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6278e0698b2c447999b6635e77c8fa21 : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d11bb882f46e4695be6b0893c74d3db5 : SUCCESS in 7m 45s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5f7742f756e34adb87588a9534a8c4c7 : FAILURE in 33m 10s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a98f8a1dc5f242d39da75fa6ed49db2b : FAILURE in 35m 00s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1c676aa0c0d145c2bd16dc6526ffeb03 : FAILURE in 15m 20s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/76ed90f98d894c48ab9ce808176e5475 : FAILURE in 17m 13s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/45c3603d18fa4ec58bc4a49f6aaa6f16 : FAILURE in 17m 49s","accounts_in_message":[],"_revision_number":2},{"id":"cfb0e1bfd712b3bf3ab887ea5c6de4af68d1235f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-30 10:09:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f2a50d089f75328d56fec9a9caf43be9b784f086","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 11:24:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/8e2750b150c54c5f853d58f54d374bee : SUCCESS in 34m 24s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/e874db0e7d68474c8da223d8021f1aa8 : SUCCESS in 34m 10s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/d8d3d0cccfa14f409fbf9eed51e318af : SUCCESS in 55m 49s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/b40d58fdf46749928d5235b9c0cf0c17 : SUCCESS in 59m 52s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/93b4effe1a0347f5bcc2615576e623e1 : SUCCESS in 1h 04m 50s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/804901c1018f44afb0873aaed206fa87 : SUCCESS in 18m 00s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/1e551bca250e45b7b3114c5804ea4962 : SUCCESS in 18m 36s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a9435ace8d6c47ab9f15b32a73bd30fd : SUCCESS in 54m 43s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d03af085d8045b7bbf054fd3a12a2d1 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f025dba9ad764da8a148d86ac02dbb61 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/975177aa2554483bb71a9e8568264b21 : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/240747b60eff4a268162c576ff94bf5d : SUCCESS in 6m 13s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6ccb47acb98c497190b4e187514e0038 : SUCCESS in 32m 09s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2486d2ad3b694a1bb55cb8ded3c46f96 : SUCCESS in 36m 00s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5fd46584b8df424bb4369dc2551ab4b9 : SUCCESS in 47m 50s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/861619bddea9448f8f6525224eae51b0 : SUCCESS in 52m 22s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/af99e66da22841ff8050b101a1399407 : SUCCESS in 54m 37s","accounts_in_message":[],"_revision_number":2},{"id":"e2931a960f3d1b9277e1cd83c3d4d2d51b0b670e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 11:25:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"79c2465f8667ce24a259ec4d889bb86df36ed981","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 12:29:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/bda95524e1464e92a5e4c5e26cd24b18 : SUCCESS in 1h 57m 46s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/fa34b9eb4d2e4953a306145e5715186f : FAILURE in 1h 42m 14s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/58a52d4983ce42d0ae1328f6704ea82b : FAILURE in 48m 54s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/7981809d2d5e4169b3b0705f41be50c4 : FAILURE in 1h 44m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b715c5f5a116a70e8139c6e08f00540578176151","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 14:45:41.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/7927bc628341499f87caf613f3740b27 : SUCCESS in 33m 22s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/2e78b4af657c450bbf4d6d27b4ff49ed : SUCCESS in 1h 17m 29s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/40c75144fd7e4465b4c6767dd8b04d62 : SUCCESS in 1h 21m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5c60cd06b0a433e973fa64837748d95 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/26e00bbf6e5140c8abcec1f5013134b6 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3cc71f3011684e86b3bda916376b551f : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de24bdb5a2a84c12b0e96eb6ca0b74be : SUCCESS in 6m 09s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1551df5bc373480392c5df596330bd19 : SUCCESS in 46m 53s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/85d08ad4eaf340ef83497331c377cb6e : SUCCESS in 1h 02m 48s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7f31f6aa9fa9484cb6bcdef25eb13b5b : SUCCESS in 57m 35s","accounts_in_message":[],"_revision_number":2},{"id":"0bc8acd5c4958de0d2cf584921dd1cb85bd0bd19","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 14:45:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"72c69413784ed6e10222d5d880c377c13b8ac86b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 14:47:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d405a44d3a94797b684376be7052b3b : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"02dfa915a51b9abbb271e69f58fb8290742d5163","revisions":{"f10be4efbbf280371ad720ec37055ef45bd51fcb":{"kind":"REWORK","_number":1,"created":"2022-05-27 17:52:15.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/37/843637/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/37/843637/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/843637/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/843637/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/843637/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/37/843637/1"}}},"commit":{"parents":[{"commit":"b716b5a5a761fa4a5cb826be2c38389727660715","subject":"Fix local sources of git repositories","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b716b5a5a761fa4a5cb826be2c38389727660715"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-21 11:44:38.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-05-27 17:52:15.000000000","tz":0},"subject":"masakari: add Cyrus SASL packages to monitors image","message":"masakari: add Cyrus SASL packages to monitors image\n\nAdds Cyrus SASL packages necessary for the DIGEST-MD5 and SCRAM-SHA-256\nmechanisms. These can be used for libvirt SASL authentication.\nFollow up to I13e19ca29eeab40cd08fa3afe2cdf7531867f81b.\n\nPartial-Bug: #1964013\nChange-Id: Ic4b6171789c3d360317599310492771bd78828ec\n(cherry picked from commit 5cc480a617c68ed20c80bffec16321a937f5a32b)\n(cherry picked from commit 444bdffcc7d4612fa6dd25fdead8f56428151735)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f10be4efbbf280371ad720ec37055ef45bd51fcb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f10be4efbbf280371ad720ec37055ef45bd51fcb"}]},"branch":"refs/heads/stable/victoria"},"02dfa915a51b9abbb271e69f58fb8290742d5163":{"kind":"REWORK","_number":2,"created":"2022-05-27 17:52:50.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/37/843637/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/37/843637/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/843637/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/843637/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/843637/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/37/843637/2"}}},"commit":{"parents":[{"commit":"b716b5a5a761fa4a5cb826be2c38389727660715","subject":"Fix local sources of git repositories","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b716b5a5a761fa4a5cb826be2c38389727660715"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-21 11:44:38.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-05-27 17:52:50.000000000","tz":0},"subject":"masakari: add Cyrus SASL packages to monitors image","message":"masakari: add Cyrus SASL packages to monitors image\n\nAdds Cyrus SASL packages necessary for the DIGEST-MD5 and SCRAM-SHA-256\nmechanisms. These can be used for libvirt SASL authentication.\nFollow up to I13e19ca29eeab40cd08fa3afe2cdf7531867f81b.\n\nPartial-Bug: #1964013\nChange-Id: Ic4b6171789c3d360317599310492771bd78828ec\n(cherry picked from commit 5cc480a617c68ed20c80bffec16321a937f5a32b)\n(cherry picked from commit 444bdffcc7d4612fa6dd25fdead8f56428151735)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/02dfa915a51b9abbb271e69f58fb8290742d5163"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/02dfa915a51b9abbb271e69f58fb8290742d5163"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
