)]}'
{"id":"openstack%2Fkolla~843645","triplet_id":"openstack%2Fkolla~stable%2Fxena~I6bc8b489a810849744c8bead6b9e350c5d3e36b7","project":"openstack/kolla","branch":"stable/xena","topic":"new-naming-scheme-stable/yoga-stable/xena","attention_set":{},"removed_from_attention_set":{"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-05-30 07:32:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6bc8b489a810849744c8bead6b9e350c5d3e36b7","subject":"Fix Ubuntu image builds","status":"MERGED","created":"2022-05-29 18:43:50.000000000","updated":"2022-05-30 07:33:36.000000000","submitted":"2022-05-30 07:32:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"843645-new-naming-scheme-stable/yoga-stable/xena","cherry_pick_of_change":843644,"cherry_pick_of_patch_set":3,"meta_rev_id":"858cef2a51319019b645a912e57a236e98fe3529","_number":843645,"virtual_id_number":843645,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-30 07:32:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-05-30 06:01:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-05-29 18:44:36.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-05-30 06:01:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-29 20:09:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-30 06:01:08.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"512f2e1463104977fad8ba28d5208dcd6c2b749f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-29 18:43:50.000000000","message":"Patch Set 1: Cherry Picked from branch stable/yoga.","accounts_in_message":[],"_revision_number":1},{"id":"bd599ea5592bec1f508775d10f1dc5bd461ca6f4","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-29 18:44:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6be176af005b4692efa5c7835a8bcfcbb1bd6e76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-29 20:09:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/8734d4987bcb4e83be2989047fb3f339 : SUCCESS in 37m 21s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/ccae54bb28df458f8794b43bc441754b : SUCCESS in 35m 13s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/0793309409e746a2a2d3ed9531203be1 : SUCCESS in 1h 12m 40s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/18ee7ae7dc59447fa6742bf4d1539a83 : SUCCESS in 1h 09m 57s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/02574bcbcbcb42da8b162d6e7ad59921 : SUCCESS in 1h 23m 20s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/cf54fd197e5e4982bdde4f2dd069a1ed : SUCCESS in 20m 52s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/5499c2724aa54d9aa601e4110328ff3a : SUCCESS in 21m 17s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7f36c191c3634503a0bcb8e329cf9e56 : SUCCESS in 50m 21s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0d368b3457f43519ce5ad253c304e2d : SUCCESS in 3m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4dfb8f0693644f73abdd48d88c9c729c : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/381662f40fe846ae94fd2c74fca42228 : SUCCESS in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8602924071fd4a70b7a8767d883590d1 : SUCCESS in 3m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef75c84a2d6d41688479b4b1f661a7c2 : SUCCESS in 7m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c4dcce6c2124f738630e28c3c74e91f : SUCCESS in 3m 08s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d08af778243f4bffad3519b800761da4 : SUCCESS in 31m 24s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8b6d81bb717e46d0b1e0324ab09105ea : SUCCESS in 31m 28s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1d685d27959f48249279b982411d5e9e : SUCCESS in 54m 32s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e9dab8a612a8408db33d36aaac2d9743 : SUCCESS in 56m 13s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7887bc481220473cac295b6abb0bffe7 : SUCCESS in 1h 07m 20s","accounts_in_message":[],"_revision_number":1},{"id":"2f8c857b393b1c38131e029fe03ea7b4e351f44a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-29 20:53:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/5eb704f0fc084808a2db1d4c990b62cc : SUCCESS in 2h 07m 49s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3370874cbcb3414db01b8579dd9f1914 : SUCCESS in 1h 14m 06s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/67eb9d708da8421b97d7e370bbe32554 : SUCCESS in 1h 14m 29s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f7bddd1c6e074f96b797409726404c4c : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2265db0386144da3a6a4aea726946233 : SUCCESS in 7m 48s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/2ff94fca494d4429b97eb1355c252fb7 : FAILURE in 1h 26m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"802c0048a1e37d9b65ec44ea464b92be167b3bc2","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-05-30 06:01:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ec65de8c25abbbf90d027d991f4f946055f3a499","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 06:01:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"850f9f2461f1c3cea22b56cf1899101d5859c1e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 07:32:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/f89d098a91ac4f42b3983933ee908d70 : SUCCESS in 29m 47s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/bf6c20284f6b45b98eb9ccb32b6b1fba : SUCCESS in 1h 16m 28s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/bb4e7675679f4f2f90b5ac4a8270aa18 : SUCCESS in 1h 26m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57f7942a8bbb490a8a37f71e862c6a9d : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c4b0b7b3f9c04208b2462d2b81943bec : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/285533a4bcb84897b7dc74f33b06616f : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af38db840991450686129c2acaa29900 : SUCCESS in 4m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/96a38caf1a8043de8fefc750f7f88728 : SUCCESS in 3m 47s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5bf6e64e624b4f21bd4b9a782b174453 : SUCCESS in 20m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7b6b0a4fa5e7444493a1231a26fd7ab1 : SUCCESS in 49m 58s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/214d1b48d64e42bf86f94b442cebf3ac : SUCCESS in 1h 05m 31s","accounts_in_message":[],"_revision_number":1},{"id":"98a6c61ad04bb73ed3a203f8b393863f17959c6b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 07:32:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"858cef2a51319019b645a912e57a236e98fe3529","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 07:33:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/582da86546c74386a245290ab31eb1c2 : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21b1e6f5912d4be9a7c4afd6b40e1aa6 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6298c0e07324757e97627a206f5467a317de6d97","revisions":{"6298c0e07324757e97627a206f5467a317de6d97":{"kind":"REWORK","_number":1,"created":"2022-05-29 18:43:50.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/45/843645/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/45/843645/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/843645/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/843645/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/843645/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/45/843645/1"}}},"commit":{"parents":[{"commit":"b2396ec64e605d0c999d3e811ae185a218e4204a","subject":"Merge \"Fix local sources of git repositories\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b2396ec64e605d0c999d3e811ae185a218e4204a"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-05-26 08:27:53.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-05-29 18:43:50.000000000","tz":0},"subject":"Fix Ubuntu image builds","message":"Fix Ubuntu image builds\n\nRegarding rabbitmq and erlang.\n\nChange-Id: I6bc8b489a810849744c8bead6b9e350c5d3e36b7\n(cherry picked from commit 240b3cf448354bf1a7271f8f81d730e51da856cb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6298c0e07324757e97627a206f5467a317de6d97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6298c0e07324757e97627a206f5467a317de6d97"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
