)]}'
{"id":"openstack%2Fkolla~843646","triplet_id":"openstack%2Fkolla~stable%2Fwallaby~I6bc8b489a810849744c8bead6b9e350c5d3e36b7","project":"openstack/kolla","branch":"stable/wallaby","topic":"new-naming-scheme-stable/yoga-stable/xena-stable/wallaby","attention_set":{},"removed_from_attention_set":{"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-05-30 07:32:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6bc8b489a810849744c8bead6b9e350c5d3e36b7","subject":"Fix Ubuntu image builds","status":"MERGED","created":"2022-05-29 19:15:35.000000000","updated":"2022-05-30 07:33:37.000000000","submitted":"2022-05-30 07:32:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"843646-new-naming-scheme-stable/yoga-stable/xena-stable/wallaby","cherry_pick_of_change":843645,"cherry_pick_of_patch_set":1,"meta_rev_id":"2e395b8595aa27ea4e43ebc202b7880474ef6fdb","_number":843646,"virtual_id_number":843646,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-30 07:32:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-05-30 06:01:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-05-29 19:28:15.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-05-30 06:01:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-29 19:23:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-30 06:01:19.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"6772d0597784b8aa14114a1d19cbc82df1e942d1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-29 19:15:35.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.\n\nThe following files contain Git conflicts:\n* docker/base/Dockerfile.j2\n* kolla/template/repos.yaml","accounts_in_message":[],"_revision_number":1},{"id":"f94b4bc9551230b367bbfb4b82ed96bfac5b0c9e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-29 19:16:33.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"407bd8a43bbbbf41c0ea670d563f7f49dbac488d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-29 19:22:53.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0a878822ea8420c9929e00d1bf28862980ab2cdf","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-29 19:22:59.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":3},{"id":"af8fcb25f4630db94f4330fa6faced00fe6375a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-29 19:23:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b080c8d59f95447cb6f19f2cfd4e3f23 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ee114ba445047cdb8bc6459a0eb46f1 : SUCCESS in 4m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fcebf515ca8a48a384c6040487cdba97 : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":2},{"id":"9b042a1407ad2132a94df7d6eb466586b7ffa603","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-29 19:28:15.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"e2e65ffd2f6002e6eb3a3efd3f61a7748daf9cdb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-29 20:35:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/f3be28c63ba04a7e925883e08fbb6132 : SUCCESS in 28m 46s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/4712d2e34f8541c2921018a68cc95324 : SUCCESS in 25m 46s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/fb13148bd3e943b293c03bec69f10540 : SUCCESS in 1h 06m 50s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/229029a74d0e450188942889982fce26 : SUCCESS in 1h 01m 35s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/9355c0835b974b3cbd02e9ad2dfd88a0 : SUCCESS in 1h 09m 10s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b0f5cc55d09343469204fd07dae019d5 : SUCCESS in 20m 44s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ee8b5fc9fb9c4de7a99ca37def24cacf : SUCCESS in 20m 07s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4275c5311beb4b84afd7f7a63dfe03ad : SUCCESS in 49m 01s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66ff738e0a8546798bd2e3b78edbbda7 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d4041a3b36cf40d59f04222b725832e1 : SUCCESS in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f1182c8d2c0a4e1a9ddc8fd4c46c1bdf : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a19b3824a3c9480f8ff923d41e4d5049 : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2602f229368d4646b801784fe567f022 : SUCCESS in 6m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c7f71507d5c4654926ae4b3ef4fddd8 : SUCCESS in 4m 40s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a214a02e0fab4c0e85a084af1483c30a : SUCCESS in 14m 59s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4d267898f434404dacfa1c6017befe7b : SUCCESS in 21m 50s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8030f77077e24f3f83325355d4de4692 : SUCCESS in 49m 37s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b9ee395e9dc2499eabd9778949c06a7a : SUCCESS in 52m 08s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/bc739744340a4dd1abd05a1597c9d7d2 : SUCCESS in 57m 23s","accounts_in_message":[],"_revision_number":3},{"id":"fd4e8f734bc4d5a7f07488659f8b97fc19f572df","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-29 21:28:41.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/86dabdc31a514e2e8b01911dee2b433b : SUCCESS in 2h 03m 00s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d60a6e435ad64cd7aeb4f07557dd3780 : SUCCESS in 1h 37m 06s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/176a0aabb9e14b609ff2a304ff0610a6 : SUCCESS in 2h 04m 48s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/7c30aad3fb7944a39400bee881080133 : FAILURE in 20m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9d76fbfa81d31d1a9caeded9ed7df8b1ed7c96a3","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-05-30 06:01:19.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"16a35716c25a9bbc2c4f19bffff11a6aef03bffb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 06:01:59.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"10c832ac87055b9b26b8e3844f5765b2eed92838","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 07:32:26.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/63b7f3b0d4e14ccf8dfd34b06a9efa8b : SUCCESS in 26m 29s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/0aba1c56e3f149adbef02e0889a48e10 : SUCCESS in 1h 04m 52s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/003cbc2f18c84a5787e5ae5d495c57ee : SUCCESS in 1h 15m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ea957aba9ce449497f48640a03d6d5f : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f8530f50bb004915b4f15342218b23bb : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c8ac87815364a12911e8f222500d7cd : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47c5aa750224493f8db9713074197d11 : SUCCESS in 5m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca2b9aefde9e42eab76216c459d88306 : SUCCESS in 3m 55s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a5dd7a1ff32545db88370f67ba3627a8 : SUCCESS in 19m 32s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/436084a81850488e8b76806d44af4fbf : SUCCESS in 50m 14s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/300cb0e8b3de4619b9232e9790ec8acc : SUCCESS in 1h 09m 28s","accounts_in_message":[],"_revision_number":3},{"id":"56ee3062de2d2ed692a1984b4cbad2acfecaa0f1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 07:32:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"2e395b8595aa27ea4e43ebc202b7880474ef6fdb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 07:33:37.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd6118085a1249579784705dbe4febe0 : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d41a221751fc480d93c541135ee4cfff : SUCCESS in 48s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b0517b35627d02fa969c223070a9cf4d6e6a5916","revisions":{"f4602611b8c169080bf9779a4e301c82c5efa542":{"kind":"REWORK","_number":1,"created":"2022-05-29 19:15:35.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/46/843646/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/843646/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/843646/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/843646/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/843646/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/843646/1"}}},"commit":{"parents":[{"commit":"8ffdee926405bbc291f6393f2bd8cb91ec01be50","subject":"Fix local sources of git repositories","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8ffdee926405bbc291f6393f2bd8cb91ec01be50"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-05-26 08:27:53.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-05-29 19:15:35.000000000","tz":0},"subject":"Fix Ubuntu image builds","message":"Fix Ubuntu image builds\n\nRegarding rabbitmq and erlang.\n\nChange-Id: I6bc8b489a810849744c8bead6b9e350c5d3e36b7\n(cherry picked from commit 240b3cf448354bf1a7271f8f81d730e51da856cb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f4602611b8c169080bf9779a4e301c82c5efa542"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f4602611b8c169080bf9779a4e301c82c5efa542"}]},"branch":"refs/heads/stable/wallaby"},"a618e6533c5caeb700145c5aeda53de1e10dbda2":{"kind":"REWORK","_number":2,"created":"2022-05-29 19:16:33.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/46/843646/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/843646/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/843646/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/843646/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/843646/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/843646/2"}}},"commit":{"parents":[{"commit":"8ffdee926405bbc291f6393f2bd8cb91ec01be50","subject":"Fix local sources of git repositories","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8ffdee926405bbc291f6393f2bd8cb91ec01be50"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-05-26 08:27:53.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-05-29 19:16:25.000000000","tz":0},"subject":"Fix Ubuntu image builds","message":"Fix Ubuntu image builds\n\nRegarding rabbitmq and erlang.\n\nChange-Id: I6bc8b489a810849744c8bead6b9e350c5d3e36b7\n(cherry picked from commit 240b3cf448354bf1a7271f8f81d730e51da856cb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a618e6533c5caeb700145c5aeda53de1e10dbda2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a618e6533c5caeb700145c5aeda53de1e10dbda2"}]},"branch":"refs/heads/stable/wallaby"},"b0517b35627d02fa969c223070a9cf4d6e6a5916":{"kind":"REWORK","_number":3,"created":"2022-05-29 19:22:53.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/46/843646/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/843646/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/843646/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/843646/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/843646/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/843646/3"}}},"commit":{"parents":[{"commit":"8ffdee926405bbc291f6393f2bd8cb91ec01be50","subject":"Fix local sources of git repositories","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8ffdee926405bbc291f6393f2bd8cb91ec01be50"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-05-26 08:27:53.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-05-29 19:22:46.000000000","tz":120},"subject":"Fix Ubuntu image builds","message":"Fix Ubuntu image builds\n\nRegarding rabbitmq and erlang.\n\nChange-Id: I6bc8b489a810849744c8bead6b9e350c5d3e36b7\n(cherry picked from commit 240b3cf448354bf1a7271f8f81d730e51da856cb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b0517b35627d02fa969c223070a9cf4d6e6a5916"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b0517b35627d02fa969c223070a9cf4d6e6a5916"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
