)]}'
{"id":"openstack%2Fkolla~843647","triplet_id":"openstack%2Fkolla~stable%2Fvictoria~I6bc8b489a810849744c8bead6b9e350c5d3e36b7","project":"openstack/kolla","branch":"stable/victoria","topic":"new-naming-scheme-stable/yoga-stable/xena-stable/wallaby-stable/victoria","attention_set":{},"removed_from_attention_set":{"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-05-30 07:36:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6bc8b489a810849744c8bead6b9e350c5d3e36b7","subject":"Fix Ubuntu image builds","status":"MERGED","created":"2022-05-29 19:23:09.000000000","updated":"2022-05-30 07:38:42.000000000","submitted":"2022-05-30 07:36:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"843647-new-naming-scheme-stable/yoga-stable/xena-stable/wallaby-stable/victoria","cherry_pick_of_change":843646,"cherry_pick_of_patch_set":3,"meta_rev_id":"8594c621b4df191d9f9f6912bea5bd5c8bf2ed17","_number":843647,"virtual_id_number":843647,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-30 07:36:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-05-30 06:01:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-05-29 19:28:20.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-05-30 06:01:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-29 21:36:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-29 22:31:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-30 06:01:30.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"c9126fcc9c0bace4633b99c6727d3dbf90cdac3f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-29 19:23:09.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.\n\nThe following files contain Git conflicts:\n* kolla/template/repos.yaml","accounts_in_message":[],"_revision_number":1},{"id":"58f78187a0931c97513784f07d7b5011b3a3fe1e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-29 19:25:15.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"2dbdef96d182ad989d2b5e0775dbea3a7ccc742e","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-29 19:25:44.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"416fe054c439f503b6c42c2c9ee719517ef5773c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-29 19:28:20.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"401dba6f60042a5c3e85e2b2ad001741b36a9788","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-29 21:36:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/7120b82e956549fdb559d2b3922874f3 : SUCCESS in 2h 09m 17s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8d127e7700e8496eb58d3370bdbd72ff : FAILURE in 2h 07m 14s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/94e5dce8c830472c978d917946f397fe : FAILURE in 44m 59s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/105e3c14362e4c4eb1e501a10396ea93 : FAILURE in 1h 44m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1dd048a7f614a738bd7b83e557d7f9e8bf2f58e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-29 22:31:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/e6f9ebdfd286412d8e32e5228842408d : TIMED_OUT in 3h 01m 19s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/419b320b071e4fb48b8108e25ad1404c : SUCCESS in 35m 41s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/1fdea2f815194ad796695a12b357d5b0 : SUCCESS in 1h 08m 45s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/8f82939ab52e4081a75bfcdd7e098569 : SUCCESS in 1h 03m 20s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/83c306028b4847acb8592ac2fc07dad8 : SUCCESS in 1h 13m 39s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d92e37c278b948bfbb2cedef1d67eca8 : SUCCESS in 18m 07s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/8a160f03339d418b98c8c8bd3e6a9e05 : SUCCESS in 19m 36s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5ae84269692548debf7cf5d4c0d711ab : SUCCESS in 42m 09s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e27cf88527b44e69beefe76249a24159 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c544811dfa64b53bb951d501860cd73 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/55d0f2e2f05b4fc2a60f1cb4533100ce : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5b477b43d404219ad7f1a8d3efca325 : SUCCESS in 5m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b442637748464c6085f8cb171aaabf6a : SUCCESS in 3m 20s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/49db709b6c834696bb4318c2d374ee22 : SUCCESS in 33m 46s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/44e8fe84b63a412b8a595a46cec944cb : SUCCESS in 34m 33s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/cd7e717d4425418fa46abeca7e80c265 : SUCCESS in 45m 59s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/140520b41ae24d6eb79278741a0298aa : SUCCESS in 51m 59s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5fbcca7b0152428384fcf4aba5515d05 : SUCCESS in 57m 52s","accounts_in_message":[],"_revision_number":2},{"id":"fbf9007243de9ad2224a6d9419f5d826d877ca46","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-05-30 06:01:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"064737e844ddaee8273a4e28e656117623262ece","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 06:02:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"450c6005d30fbba238090e9fad858419a31ace37","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 07:36:43.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/3a62f220b2db4093846ed964e88ca210 : SUCCESS in 36m 35s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/4904c89075f043da96e4da5eea141a95 : SUCCESS in 1h 10m 53s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/a63666c1552d42e59403c2d3517dbf2b : SUCCESS in 1h 30m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcff2c3cb2be4fe0b0d79fc4d518f61f : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/63f204f9525f484e93c8892ad37db923 : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f35f2bd51c646079563389c38ae6757 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4a724275df942da9d65265e39d66626 : SUCCESS in 5m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c8c17123219c4181bd1ae6a01087cf38 : SUCCESS in 3m 50s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/705f4106d5ab4b9296986052ce7e150c : SUCCESS in 32m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9b4a099b6de840e3b5ff47a584990cd1 : SUCCESS in 57m 45s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/eada9982c6624a1d8b6ab243ea4b7481 : SUCCESS in 1h 02m 28s","accounts_in_message":[],"_revision_number":2},{"id":"4f4ae9d73fa3e29b18b093fb418ec3a2c88cf879","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 07:36:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"8594c621b4df191d9f9f6912bea5bd5c8bf2ed17","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 07:38:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf178e8c3d994c5f98e6d46dc4d78889 : SUCCESS in 1m 06s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c2acd0d6a47438dba6ba3b03f1aa58a : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ad72543a5d28920e5524eb9fa4a50065d8c5a5b4","revisions":{"066a4eb083f5d8653fbed83dc540c4e55bddbe07":{"kind":"REWORK","_number":1,"created":"2022-05-29 19:23:09.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/47/843647/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/843647/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/843647/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/843647/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/843647/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/843647/1"}}},"commit":{"parents":[{"commit":"b716b5a5a761fa4a5cb826be2c38389727660715","subject":"Fix local sources of git repositories","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b716b5a5a761fa4a5cb826be2c38389727660715"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-05-26 08:27:53.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-05-29 19:23:09.000000000","tz":0},"subject":"Fix Ubuntu image builds","message":"Fix Ubuntu image builds\n\nRegarding rabbitmq and erlang.\n\nChange-Id: I6bc8b489a810849744c8bead6b9e350c5d3e36b7\n(cherry picked from commit 240b3cf448354bf1a7271f8f81d730e51da856cb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/066a4eb083f5d8653fbed83dc540c4e55bddbe07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/066a4eb083f5d8653fbed83dc540c4e55bddbe07"}]},"branch":"refs/heads/stable/victoria"},"ad72543a5d28920e5524eb9fa4a50065d8c5a5b4":{"kind":"REWORK","_number":2,"created":"2022-05-29 19:25:15.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/47/843647/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/843647/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/843647/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/843647/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/843647/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/843647/2"}}},"commit":{"parents":[{"commit":"b716b5a5a761fa4a5cb826be2c38389727660715","subject":"Fix local sources of git repositories","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b716b5a5a761fa4a5cb826be2c38389727660715"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-05-26 08:27:53.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-05-29 19:25:15.000000000","tz":0},"subject":"Fix Ubuntu image builds","message":"Fix Ubuntu image builds\n\nRegarding rabbitmq and erlang.\n\nChange-Id: I6bc8b489a810849744c8bead6b9e350c5d3e36b7\n(cherry picked from commit 240b3cf448354bf1a7271f8f81d730e51da856cb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ad72543a5d28920e5524eb9fa4a50065d8c5a5b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ad72543a5d28920e5524eb9fa4a50065d8c5a5b4"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
