)]}'
{"id":"openstack%2Fkolla~843648","triplet_id":"openstack%2Fkolla~stable%2Fussuri~I6bc8b489a810849744c8bead6b9e350c5d3e36b7","project":"openstack/kolla","branch":"stable/ussuri","topic":"new-naming-scheme-stable/yoga-stable/xena-stable/wallaby-stable/victoria-stable/ussuri","attention_set":{},"removed_from_attention_set":{"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-05-30 07:36:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6bc8b489a810849744c8bead6b9e350c5d3e36b7","subject":"Fix Ubuntu image builds","status":"MERGED","created":"2022-05-29 19:25:52.000000000","updated":"2022-05-30 07:38:29.000000000","submitted":"2022-05-30 07:36:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"843648-new-naming-scheme-stable/yoga-stable/xena-stable/wallaby-stable/victoria-stable/ussuri","cherry_pick_of_change":843647,"cherry_pick_of_patch_set":2,"meta_rev_id":"3938612837fcd0fbaf99d8b3936a70cdaadd8b05","_number":843648,"virtual_id_number":843648,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-30 07:36:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-05-30 06:01:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-05-29 19:28:24.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-05-30 06:01:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-29 20:38:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-30 06:01:39.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"f80d5f4f2ebd4dd564543b654e8c215b586e8b81","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-29 19:25:52.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.\n\nThe following files contain Git conflicts:\n* kolla/template/repos.yaml","accounts_in_message":[],"_revision_number":1},{"id":"7ea2d0c3906fcb1bbf5db2c4cc739a496092193e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-29 19:27:43.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"c7834dc4ac7a416d0146233f1c4c2f867e291df7","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-29 19:28:06.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"b5d134cfc036533f5fd87d3f47888e34dff13232","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-29 19:28:24.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"28e0ea109caa1056d511868563e0fc2dcb1f3747","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-29 20:38:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/c080d008b30e47f2a77d5c7043d74cd5 : SUCCESS in 1h 05m 54s\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/62916e9b20424e1398541f3203219a74 : SUCCESS in 36m 18s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2003d1ad104a4decac5deeaf632f5a51 : SUCCESS in 23m 57s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/413943c2522140a39083681588813bc9 : SUCCESS in 19m 28s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d96e76892d404ab4832bf744438a278f : SUCCESS in 55m 02s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb90b779f8dc49cfb6ccbc2fe7295924 : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c223d086b8b64a2ea7bd688448d6ee29 : SUCCESS in 3m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8af7b971c7de448a95134d4c7157ff2a : SUCCESS in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/654edfea031f4e069bb5be9890a8b804 : SUCCESS in 5m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/03b4fb1c58e14242bdc7a57f7267e105 : SUCCESS in 3m 19s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/45fcc545997646ecac12a74c549393c7 : SUCCESS in 33m 29s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5458a5986843401a9b97d0f19ec64ab7 : SUCCESS in 33m 26s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1a7072e3a4e84370852539d03dcc7cd2 : SUCCESS in 50m 57s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b9208dd41e3e4f4f976f08cfbedb5a3b : SUCCESS in 48m 20s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/aae6c9a84ff54eed8f0906afed616a6a : SUCCESS in 1h 04m 29s","accounts_in_message":[],"_revision_number":2},{"id":"b651c1889d1a0fdd74056a572c2f801d8ad7c2ec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-29 22:21:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/0fcbd9f76dd94e0d8717dcfc39ac3d97 : SUCCESS in 2h 29m 38s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/32e050c7c1cb43ac99d8f34dae170171 : FAILURE in 1h 46m 35s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/eb6153eb02a9446da1bc47eb4fc76eb0 : FAILURE in 46m 45s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/3c6dd5c4fa484cb18b6c85d3a6b9178b : FAILURE in 1h 22m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"45c0e08eb4031bff08ee74d89ffeb266677652a1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-05-30 06:01:39.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d97e8c0223804f34a1ebad343815a5a628840101","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 06:02:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b1ae0a5fe6537af167a37abd832e422ce6d39af4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 07:36:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/d9a5a97453f74b4ebd0b7f1866a89f8d : SUCCESS in 1h 04m 11s\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/5c6e2a09963a4534a0de4a7a30c4fa0b : SUCCESS in 32m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc6234e851c34d21952cfff975fbb6ce : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c1b425b139404215a319ba5b4fbdd9fe : SUCCESS in 4m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d85598b347544ca685baeb9c7b22464c : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/375c2f9b41e24200b1b21e9b32a3a9d6 : SUCCESS in 6m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/12ee79b47f934046b836a0604da32728 : SUCCESS in 4m 26s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f2c6159c128d4326b817d262355fc627 : SUCCESS in 32m 52s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c4a27552d1214cd8b73aea5541f320ab : SUCCESS in 56m 47s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e76a764acd554b6ca330def4865ab99e : SUCCESS in 1h 14m 31s","accounts_in_message":[],"_revision_number":2},{"id":"f00c30e23e8c1a61a3c6a2c28df45234460efacd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 07:36:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"3938612837fcd0fbaf99d8b3936a70cdaadd8b05","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 07:38:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0e063034d7045ea8fccc803936233ac : SUCCESS in 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e3745c49c19f4f58bb440307e3fec73d : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d926c35f0f2e33d728d2390b93e6f5cd3d56ecb2","revisions":{"8e9e89c21aed87e1da818a080e2350aced464c8a":{"kind":"REWORK","_number":1,"created":"2022-05-29 19:25:52.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/48/843648/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/843648/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/843648/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/843648/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/843648/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/843648/1"}}},"commit":{"parents":[{"commit":"917a7c04380fc3e83b3ea9d4b2b3d7fa8b82dc0f","subject":"Fix local sources of git repositories","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/917a7c04380fc3e83b3ea9d4b2b3d7fa8b82dc0f"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-05-26 08:27:53.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-05-29 19:25:52.000000000","tz":0},"subject":"Fix Ubuntu image builds","message":"Fix Ubuntu image builds\n\nRegarding rabbitmq and erlang.\n\nChange-Id: I6bc8b489a810849744c8bead6b9e350c5d3e36b7\n(cherry picked from commit 240b3cf448354bf1a7271f8f81d730e51da856cb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8e9e89c21aed87e1da818a080e2350aced464c8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8e9e89c21aed87e1da818a080e2350aced464c8a"}]},"branch":"refs/heads/stable/ussuri"},"d926c35f0f2e33d728d2390b93e6f5cd3d56ecb2":{"kind":"REWORK","_number":2,"created":"2022-05-29 19:27:43.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/48/843648/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/843648/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/843648/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/843648/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/843648/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/843648/2"}}},"commit":{"parents":[{"commit":"917a7c04380fc3e83b3ea9d4b2b3d7fa8b82dc0f","subject":"Fix local sources of git repositories","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/917a7c04380fc3e83b3ea9d4b2b3d7fa8b82dc0f"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-05-26 08:27:53.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-05-29 19:27:43.000000000","tz":0},"subject":"Fix Ubuntu image builds","message":"Fix Ubuntu image builds\n\nRegarding rabbitmq and erlang.\n\nChange-Id: I6bc8b489a810849744c8bead6b9e350c5d3e36b7\n(cherry picked from commit 240b3cf448354bf1a7271f8f81d730e51da856cb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d926c35f0f2e33d728d2390b93e6f5cd3d56ecb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d926c35f0f2e33d728d2390b93e6f5cd3d56ecb2"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
