)]}'
{"id":"openstack%2Fkolla~843719","triplet_id":"openstack%2Fkolla~master~I0cabb9a1c1cb5d55813d7885319094c34085c6e7","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"28706":{"account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"last_update":"2022-05-30 09:36:08.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I0cabb9a1c1cb5d55813d7885319094c34085c6e7","subject":"test","status":"ABANDONED","created":"2022-05-28 12:34:40.000000000","updated":"2022-05-30 09:36:08.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"378c36c6dbbc3a0c3e7779d66eb390653bf227c4","_number":843719,"virtual_id_number":843719,"owner":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-05-28 13:39:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"all":[{"value":-1,"date":"2022-05-28 12:36:14.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-28 13:39:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ea296b472a98ddf18c8df74f1c03273e66ee67ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-05-28 12:34:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a5c58f71436db2b7aa8bbf018c7c879e684f754e","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-05-28 12:34:52.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"8cf9af532448c0ecac470fb1e1476297c199da48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-05-28 12:35:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f6d15aaa80a4cd621bc87afbd21fea1b96e2fbfd","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-05-28 12:36:14.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"839c47289e2ec59baa4731e00ab7fbc59b203049","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-28 13:39:16.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/6343a14fa19d4b018cb189c6b529c23d : FAILURE in 26m 04s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/174b0392033e41ecb093d3dabf5e59ae : FAILURE in 26m 16s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/68114985747749aab2d9c657f249ba95 : FAILURE in 27m 47s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/239068251d774ad88a25c7686fbd421e : SUCCESS in 20m 22s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e150c9b92353486bbef90935d1e1b7da : FAILURE in 26m 50s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/cee7bfcffa4340e9b63169d861c146d8 : FAILURE in 58m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29f3066535fe4edbbf0b6b1254aef9ff : FAILURE in 2m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cc19f9dac0d944af888381980f59a1d0 : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f502ad629b3648d2b30dad122c1fedea : SUCCESS in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7ec001bc9cfe428ea373da33978b6e38 : SUCCESS in 3m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cd06e7ad5b640e4b13126d032e89270 : SUCCESS in 9m 23s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/2d21db99e8984cb8885186f04f438729 : FAILURE in 22m 38s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9744ee49383147b38650f5cecad11258 : FAILURE in 19m 30s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e9bd596af6af430a8f47aba210f9ea23 : FAILURE in 18m 59s","accounts_in_message":[],"_revision_number":2},{"id":"0b559840f1da73fefb3e19085f21b8bfd73bccc0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-28 14:25:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-aarch64 https://zuul.opendev.org/t/openstack/build/d7417ac0008243d98f4c899d3c0da9dd : FAILURE in 1h 47m 15s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d947d705963b4eee994044101cefa6c3 : SUCCESS in 1h 26m 58s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8b229fbec6954bfbba44b5440db0455a : FAILURE in 1h 18m 49s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3a7148dea946440da7b0b63ad6d1598a : SUCCESS in 7m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d629118785934a1d8eb696d3933a4cb9 : SUCCESS in 7m 16s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/20f98a6818374b1eb55fed928da9b1e9 : FAILURE in 1h 26m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"378c36c6dbbc3a0c3e7779d66eb390653bf227c4","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2022-05-30 09:36:08.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7fddb9aafe44437137c3df2b4e01a231d5780572","revisions":{"47ea24e5984f61bf481161083a77d9e09144de6e":{"kind":"REWORK","_number":1,"created":"2022-05-28 12:34:40.000000000","uploader":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"ref":"refs/changes/19/843719/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/19/843719/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/843719/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/843719/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/843719/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/19/843719/1"}}},"commit":{"parents":[{"commit":"a51ebc9e7fc2163623cb3d9b7540c23ea984c859","subject":"Merge \"[CI] Move queue setting to project level\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a51ebc9e7fc2163623cb3d9b7540c23ea984c859"}]}],"author":{"name":"zhu.boxiang","email":"zhu.boxiang@99cloud.net","date":"2022-05-28 12:34:14.000000000","tz":480},"committer":{"name":"zhu.boxiang","email":"zhu.boxiang@99cloud.net","date":"2022-05-28 12:34:14.000000000","tz":480},"subject":"test","message":"test\n\nChange-Id: I0cabb9a1c1cb5d55813d7885319094c34085c6e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/47ea24e5984f61bf481161083a77d9e09144de6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/47ea24e5984f61bf481161083a77d9e09144de6e"}]},"branch":"refs/heads/master"},"7fddb9aafe44437137c3df2b4e01a231d5780572":{"kind":"REWORK","_number":2,"created":"2022-05-28 12:35:48.000000000","uploader":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"ref":"refs/changes/19/843719/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/19/843719/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/843719/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/843719/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/843719/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/19/843719/2"}}},"commit":{"parents":[{"commit":"a51ebc9e7fc2163623cb3d9b7540c23ea984c859","subject":"Merge \"[CI] Move queue setting to project level\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a51ebc9e7fc2163623cb3d9b7540c23ea984c859"}]}],"author":{"name":"zhu.boxiang","email":"zhu.boxiang@99cloud.net","date":"2022-05-28 12:34:14.000000000","tz":480},"committer":{"name":"zhu.boxiang","email":"zhu.boxiang@99cloud.net","date":"2022-05-28 12:35:19.000000000","tz":480},"subject":"test","message":"test\n\nChange-Id: I0cabb9a1c1cb5d55813d7885319094c34085c6e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7fddb9aafe44437137c3df2b4e01a231d5780572"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7fddb9aafe44437137c3df2b4e01a231d5780572"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
