)]}'
{"id":"openstack%2Fkolla~844456","triplet_id":"openstack%2Fkolla~stable%2Fxena~I0b07da22fea27e0ff4e90aaad19e50d84ff9a121","project":"openstack/kolla","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2022-06-06 10:37:55.000000000","reason":"Change was submitted"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2022-06-06 10:37:55.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-06-03 07:23:58.000000000","reason":"Radosław Piliszek replied on the change"}},"hashtags":[],"change_id":"I0b07da22fea27e0ff4e90aaad19e50d84ff9a121","subject":"Allow to provide own repos.yaml file","status":"MERGED","created":"2022-06-02 13:30:04.000000000","updated":"2022-06-06 10:39:04.000000000","submitted":"2022-06-06 10:37:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"844456","meta_rev_id":"62f533fb1781bacb457948535273a950dc993cb7","_number":844456,"virtual_id_number":844456,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-06 10:37:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-06-06 09:04:31.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2022-06-03 07:23:58.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2022-06-06 09:04:31.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}],"reviewers":{"CC":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}],"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-02 13:30:04.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"CC"},{"updated":"2022-06-02 13:31:35.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2022-06-02 13:31:35.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-06-02 13:52:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-06 09:04:31.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"a30338cefecbc8b7353233083c9fe3acb3b6df9c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-06-02 13:30:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2e35861dda83d044524379cbc29a363709bfdf3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 13:52:10.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/e42cbf9bcd7d4f5ba90bea2c8ec2febd : FAILURE in 8m 03s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/daf9788407224647b0634c64ced79fb7 : FAILURE in 6m 20s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/3ed1d55d877d4ab3ae0b126faf84976c : FAILURE in 14m 20s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/2e77a4d4aac24878b62002242cd4d82f : FAILURE in 12m 16s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/0533194ea88448b0be29c7152c7a4f2a : FAILURE in 12m 35s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b2c56aad56b149d29613f9e7a46498b8 : FAILURE in 2m 56s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e4dc2e060fd84926ac73ef07f1968cdf : FAILURE in 3m 33s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c0b4c72ec1ca46ea90a2ffa0b5513f9e : FAILURE in 15m 23s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfb39a995a504915b968df9d6354ba0a : FAILURE in 3m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb3f18a8333f4c91b15c779d3740b0c3 : FAILURE in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8c4d0171c0042f9b05ce69302252992 : FAILURE in 3m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bee9fb2ecb8a4023af3c938d804df991 : FAILURE in 3m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e79ca6a0728743eaa637ee7571d86bc3 : SUCCESS in 7m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6df4b4e7272149e2b94e4280742ea083 : SUCCESS in 4m 09s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/eb53cf6af8e84eeb9528e12abe78abd0 : FAILURE in 4m 07s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7763d414700f4759b081759cba61b9af : FAILURE in 3m 35s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/17b26d1fde1e4d3a91c7a509063192e6 : FAILURE in 12m 06s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6411e18a7292422993a0de355c2cdbbb : FAILURE in 10m 35s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ebe93c5a66e84bcd8a329e4d056357a1 : FAILURE in 9m 21s","accounts_in_message":[],"_revision_number":1},{"id":"e51a80eda3757d6f585954d664f8f8d9a2eea0b1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 14:12:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/dbdebcf412f644fb8f032d4a164ebe98 : FAILURE in 8m 31s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0b299d76596544e08861f388deb983d4 : FAILURE in 6m 46s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0abc68bbead54491bd0f7b90270e9c25 : FAILURE in 22m 42s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ed22ad1ddfe04fc495ebed3c78255620 : FAILURE in 5m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/044f34a8d3d340e08009db7982711c92 : FAILURE in 6m 37s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/b1a703387f614617bd5f4c7e51e2a07f : FAILURE in 4m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"64d2de6a0a91f7eb31a9b1dd4942dd7c66303778","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-06-02 14:18:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"631091911956cf7007558075e36fd82ae05d3910","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 15:48:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/dcae8d4ac328454091b1b70a81272499 : SUCCESS in 31m 10s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/c5d8d253aeb74e67bbf0a514fda61248 : SUCCESS in 32m 12s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/54451ffe0e0e4878895cd33950d23bc4 : SUCCESS in 1h 13m 19s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/e0b77e1b25214e5baf345f24b37eeaa9 : SUCCESS in 1h 14m 34s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/dbe89031b0dc4aff8c55bf4ff8287b58 : SUCCESS in 1h 28m 20s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a5ae49c9cae7499189ef14ee0b53e8aa : SUCCESS in 20m 09s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6641d2870bd64fee97de57d9710ace53 : SUCCESS in 20m 38s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a9603886bf4140008d452291940b6d9e : SUCCESS in 1h 03m 27s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ef3bde920c64226bcbef298a65dc295 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4982170763104ed3b051cd37765c7b27 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/844e34712502458b8a86381fe3376b75 : SUCCESS in 3m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3be1417b065f4a16b8eb07da8d377860 : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91c5a770cb0a45358e8a4d34574eca06 : SUCCESS in 6m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b6eec7ea3eb24974ac32b1660e597306 : SUCCESS in 4m 54s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1ee79e9dbd7b4085bbe1df48d2f5472a : SUCCESS in 23m 27s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/48fccc4470cf4255a63e80337a6eca28 : SUCCESS in 22m 03s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e20e9e04257b467781af4645a0610eb3 : SUCCESS in 55m 32s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/dc8f1588229047ca82e524d42e0aa29d : SUCCESS in 58m 21s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c900ace69d9141bf82fede504d27de16 : SUCCESS in 1h 08m 00s","accounts_in_message":[],"_revision_number":2},{"id":"ef1d5ae57e1a0a1ff5100f2baebc8f550d5b5421","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 18:48:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/c3323e576efb4757b28d9a6308ce8b2d : SUCCESS in 1h 58m 46s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/995bbd81f8294f5ea2e582c9497c7b5d : SUCCESS in 1h 28m 51s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/68a5a32b821346c79d1b9809b6a90447 : SUCCESS in 2h 13m 26s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e68429607a8e4cd4bc7e1ab69c1acf44 : SUCCESS in 6m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c58675c319ec48babb58644367da1373 : SUCCESS in 7m 05s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/e26aaa110a464228a328e9ffceea5006 : FAILURE in 1h 26m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6274d40177af9609bad2da30c75fbd97d590b1f4","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-06-03 07:23:58.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"75390574287cce4dcb2b53b1123f577698c3db16","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-06-06 09:04:31.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0677731e6d20ac396ad6275b0ba3a0ce0f859f3a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-06 09:05:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a2f9352ff9ef60b52e9a0f05ae8341fc12854e7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-06 10:37:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/cc7e3b7b611548c8b0e54c63606ae3f0 : SUCCESS in 26m 17s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/e0c38948528345c38e63ad7534fae071 : SUCCESS in 1h 10m 42s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/e62bee47a2d647dcbad07426ab2f9acc : SUCCESS in 1h 22m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7571e1e698234a478ca077f9daec1156 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a363de1e258b41e59a160adc9124081c : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/817a4c3010ac491ba9f177344954397c : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddc4f5fe821543f7ad5e9b7510adeffe : SUCCESS in 5m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e027152615554cc0b9df55c762b9e472 : SUCCESS in 4m 30s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6c590d37efb04797b4ff17c28c430734 : SUCCESS in 24m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8ed1f3abae1146babdff11346689effe : SUCCESS in 56m 29s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ec7269d33f9643899c675c7aaae82578 : SUCCESS in 1h 08m 39s","accounts_in_message":[],"_revision_number":2},{"id":"70699e75840381d85dce15ce6e9b1ec8e1147219","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-06 10:37:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"62f533fb1781bacb457948535273a950dc993cb7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-06 10:39:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45a13536917f47d79b34f51495411b68 : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b469f086804417c84003c5a9e5a139e : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e90786ab9affe99e3c6986f54a580844f1598e39","revisions":{"b4541a9e02f393c6a22578c594a7593b7bdb5f17":{"kind":"REWORK","_number":1,"created":"2022-06-02 13:30:04.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/56/844456/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/844456/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/844456/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/844456/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/844456/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/844456/1"}}},"commit":{"parents":[{"commit":"6298c0e07324757e97627a206f5467a317de6d97","subject":"Fix Ubuntu image builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6298c0e07324757e97627a206f5467a317de6d97"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-06-01 15:16:22.000000000","tz":120},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-06-02 13:29:45.000000000","tz":180},"subject":"Allow to provide own repos.yaml file","message":"Allow to provide own repos.yaml file\n\nSeveral people use Kolla behind firewall/proxy. Internal mirrors are\noften in use then. We do not provide a way to replace repos.yaml file in\nan easy way which may lead to complicated solutions or template\noverrides.\n\nThis patch adds a way to provide own copy of repos.yaml file.\n\nChange-Id: I0b07da22fea27e0ff4e90aaad19e50d84ff9a121\n(cherry picked from commit 213190ae030e9971225ea3f103b4be0a7f2532d4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b4541a9e02f393c6a22578c594a7593b7bdb5f17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b4541a9e02f393c6a22578c594a7593b7bdb5f17"}]},"branch":"refs/heads/stable/xena"},"e90786ab9affe99e3c6986f54a580844f1598e39":{"kind":"REWORK","_number":2,"created":"2022-06-02 14:18:56.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/56/844456/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/844456/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/844456/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/844456/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/844456/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/844456/2"}}},"commit":{"parents":[{"commit":"6298c0e07324757e97627a206f5467a317de6d97","subject":"Fix Ubuntu image builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6298c0e07324757e97627a206f5467a317de6d97"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-06-01 15:16:22.000000000","tz":120},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-06-02 14:18:47.000000000","tz":180},"subject":"Allow to provide own repos.yaml file","message":"Allow to provide own repos.yaml file\n\nSeveral people use Kolla behind firewall/proxy. Internal mirrors are\noften in use then. We do not provide a way to replace repos.yaml file in\nan easy way which may lead to complicated solutions or template\noverrides.\n\nThis patch adds a way to provide own copy of repos.yaml file.\n\nChange-Id: I0b07da22fea27e0ff4e90aaad19e50d84ff9a121\n(cherry picked from commit 213190ae030e9971225ea3f103b4be0a7f2532d4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e90786ab9affe99e3c6986f54a580844f1598e39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e90786ab9affe99e3c6986f54a580844f1598e39"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
