)]}'
{"id":"openstack%2Fkolla~845151","triplet_id":"openstack%2Fkolla~stable%2Fxena~I5ce2eb66a647e1805f086aa37942bf5d13eb4106","project":"openstack/kolla","branch":"stable/xena","topic":"apt-get-update-fatal","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-06-10 14:07:52.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2022-06-10 14:07:52.000000000","reason":"Change was submitted"},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-06-10 14:07:52.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-06-10 09:05:17.000000000","reason":"Mark Goddard replied on the change"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2022-06-10 11:53:14.000000000","reason":"Michal Arbet replied on the change"}},"hashtags":[],"change_id":"I5ce2eb66a647e1805f086aa37942bf5d13eb4106","subject":"Don\u0027t ignore errors of \u0027apt-get update\u0027 command","status":"MERGED","created":"2022-06-08 15:15:27.000000000","updated":"2022-06-10 14:09:25.000000000","submitted":"2022-06-10 14:07:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"845151-apt-get-update-fatal","meta_rev_id":"700d9ba74fdfb40e35fcc0427cff2ff666163f7c","_number":845151,"virtual_id_number":845151,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-10 14:07:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-06-10 09:05:17.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2022-06-10 11:53:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2022-06-10 11:53:14.000000000","_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-08 16:45:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-09 10:58:43.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2022-06-09 10:59:14.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2022-06-09 12:31:16.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2022-06-09 13:07:09.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"fce3cdc19580949165cb91696a22de37f37685ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-06-08 15:15:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"12a96ecb5fbe6e47f7fbb82692abdfba35967813","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 16:45:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/7a91fcb2d46c4e119bac2e7d9e207f8c : SUCCESS in 27m 02s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/8da5349664a64117af6873990d7af3b5 : FAILURE in 40m 32s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/df029fd3966e4a82897ee976b16e4bdb : SUCCESS in 1h 12m 57s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/9fd4f1cd426346ef99741da0ef3a0296 : SUCCESS in 1h 11m 05s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/18c3b9753bbc45a38fecb4c84681a4bd : SUCCESS in 1h 27m 29s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/0abd7af0d871492085b773b0a54cb1bb : SUCCESS in 20m 58s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ed6d7d5f20834accb90d051391ed06fd : SUCCESS in 16m 28s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e47e8b34ea684810928e8833b062ac2c : SUCCESS in 1h 01m 18s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6bf42cb349241c2b7454ad803b1fca3 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dfdfad42ef014739b7780bbd6f79efaf : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a2d4ea8b5f2543e999655f6fd889aa10 : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/337cad47ded3449dac5cebed54a02801 : SUCCESS in 4m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10e9cc94af424787a8d8f41c45632218 : SUCCESS in 6m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0675650fc9b944c4b3d5ff92330456b6 : SUCCESS in 4m 08s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/49be866b66dd426194e7ecb985ce0f64 : SUCCESS in 18m 45s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6082868cb224483394417d9aca26fefc : SUCCESS in 25m 31s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/27a0c3af339a43a1b0b1fbe757afc99b : SUCCESS in 54m 55s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/07d8f6ea65f040c3acda05160ad45ed5 : SUCCESS in 48m 52s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/025a8241b9884fa6adec737eb624ce26 : SUCCESS in 1h 03m 00s","accounts_in_message":[],"_revision_number":1},{"id":"efbba836f1d0209191e851a4441ca4267fb39c57","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 17:22:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/aaf9c4e8a554478a8de4377e2900a437 : SUCCESS in 2h 05m 14s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9df3845a2cc0423cb3fa64ea93283305 : SUCCESS in 1h 21m 30s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d412788bd49240d2bf5d4058e0cc4e9f : SUCCESS in 1h 53m 53s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a2c295da900e4a0cba4e7a021febe6df : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f9eff4d3a72d4b2bab4e601cdc7173ce : SUCCESS in 6m 19s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/5b6f2fb6f03f4dd0a44aba16538c7cb4 : FAILURE in 1h 18m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a92d96f3108fcf4eed92d5a79dd777bd2ac57e1e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-06-09 09:23:31.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"0beb22b205153bdd2fdb0e675e78c03415a6d939","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-09 11:18:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/26cdbc85a3054da7ab722ed8f94ba9ab : SUCCESS in 1h 05m 26s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a1240c9dd6804c4a9e6020972bffe810 : SUCCESS in 59m 07s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a158d31b91a4441e854d3bc7cbfcd7b7 : SUCCESS in 1h 44m 26s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/145f9a0e6de24ecd91fb924d4d069720 : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b1739c9f324d4c0d9dd0c0c20308652a : SUCCESS in 4m 37s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/e136305a95ed4aaebfa645c17866d034 : FAILURE in 57m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"944798abe2202c740d75d7932af1027ebd7ee40c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-09 11:20:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/ba7b1cc20e5442a99848f41089b507f1 : SUCCESS in 43m 27s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/a744c9268e1341e6aa72a96e2a64f635 : SUCCESS in 40m 17s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/8701917bc4a04deb812b972b3141ffdb : SUCCESS in 1h 20m 13s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/5d2118b3f9c64c3c8c2b75347e8ca2b8 : SUCCESS in 1h 19m 30s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/3236bd628b19459b908db2b221ac825c : SUCCESS in 1h 25m 40s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/bad55f7e540540c5a4239d3d60301b3f : SUCCESS in 15m 23s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/15034dd39e354fb1bce52631a896f47a : SUCCESS in 22m 40s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e13d61fffdb74d7697b2e2e035a0d20e : SUCCESS in 59m 12s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f263124d0bec40689b4d1749e924c021 : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2caf3151956543a293ec55b44bcd9b2d : SUCCESS in 3m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1474de018f534a2b95b096319244c410 : SUCCESS in 3m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c5092d093a8242acb90faea6e15fdc63 : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a271cfe6b0c04b62a3975fd56802010a : SUCCESS in 6m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a18e62a8ddb460daf5d8a7b6f7a60ac : SUCCESS in 3m 34s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6663e84a6bcc44b89f376c101057017c : SUCCESS in 14m 21s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/16d8702de16e44f9a93aba51076f57f2 : SUCCESS in 17m 57s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/918d3ffb30794fc78761402090a69330 : SUCCESS in 53m 25s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f63cce1e7e9b428ba728125203d5f165 : SUCCESS in 48m 41s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4e58a39c6ee2407d95bb6dbdad2bab8b : SUCCESS in 57m 09s","accounts_in_message":[],"_revision_number":2},{"id":"f42b892f0b0dde2bc05b147035bf8774cb23e8ff","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-06-10 09:05:17.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"cf8638b8ad42300f139965c7b4602a8e97035cd9","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-06-10 11:53:14.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"15278c279e349016f10bb914cfcfb4ace638c787","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-10 11:53:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"163d32dc66abc3933c66c75277b4191988fd766c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-10 14:07:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/5c2758246648481d9e6f37cf11c58c18 : SUCCESS in 31m 00s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/1cdc3c2ca6cb463d9b8a085c4930569b : SUCCESS in 1h 19m 54s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/c4dbdc9b1e444d5095434dfc439f189d : SUCCESS in 1h 27m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/452db37d396b417ab629630d393c394d : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b446d3630cb1477cbdd4afe704e523f5 : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b682c6eff8641d3839c7c477aaf9c48 : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8200aad4633e47efb69995050064b91f : SUCCESS in 5m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0344cccdf38d4cfd9e082caab6f2ad9e : SUCCESS in 4m 00s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a000df90d7be4d2281509286182c6196 : SUCCESS in 24m 33s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a31266f3628946ec89728b861fc62905 : SUCCESS in 48m 53s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/65d9891e5d194585a6a7b829ff693e84 : SUCCESS in 1h 01m 03s","accounts_in_message":[],"_revision_number":2},{"id":"511c9a0c9bad587709ea425042c58072679ef6a6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-10 14:07:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"700d9ba74fdfb40e35fcc0427cff2ff666163f7c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-10 14:09:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3552ed0220dd4dc7ac4cd971fa74fa99 : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e7755e6739a4cadb0bdc0311d9f741e : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1ab4e13fe31be4d3559cdf6288215d0a81788543","revisions":{"00f6cadd6834c12df7699d8cadf5f965e3cb24a2":{"kind":"REWORK","_number":1,"created":"2022-06-08 15:15:27.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/51/845151/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/845151/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/845151/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/845151/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/845151/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/845151/1"}}},"commit":{"parents":[{"commit":"e90786ab9affe99e3c6986f54a580844f1598e39","subject":"Allow to provide own repos.yaml file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e90786ab9affe99e3c6986f54a580844f1598e39"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-06-03 14:01:25.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-06-08 15:15:15.000000000","tz":180},"subject":"Don\u0027t ignore errors of \u0027apt-get update\u0027 command","message":"Don\u0027t ignore errors of \u0027apt-get update\u0027 command\n\nThis change fixes an issue with \u0027apt-get update\u0027 command and stop\ncontainer building with erroneous source repositories.\n\nChange-Id: I5ce2eb66a647e1805f086aa37942bf5d13eb4106\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit bce06d2d5a27fd2e430de0078b7d585cfc98d8a4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/00f6cadd6834c12df7699d8cadf5f965e3cb24a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/00f6cadd6834c12df7699d8cadf5f965e3cb24a2"}]},"branch":"refs/heads/stable/xena"},"1ab4e13fe31be4d3559cdf6288215d0a81788543":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-06-09 09:23:31.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/51/845151/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/845151/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/845151/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/845151/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/845151/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/845151/2"}}},"commit":{"parents":[{"commit":"5baeaf498e147e909524ac3ca094e7c987d18af6","subject":"Pin diskimage-builder version due to drop Python3.6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5baeaf498e147e909524ac3ca094e7c987d18af6"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-06-03 14:01:25.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-06-09 09:23:31.000000000","tz":0},"subject":"Don\u0027t ignore errors of \u0027apt-get update\u0027 command","message":"Don\u0027t ignore errors of \u0027apt-get update\u0027 command\n\nThis change fixes an issue with \u0027apt-get update\u0027 command and stop\ncontainer building with erroneous source repositories.\n\nChange-Id: I5ce2eb66a647e1805f086aa37942bf5d13eb4106\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit bce06d2d5a27fd2e430de0078b7d585cfc98d8a4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1ab4e13fe31be4d3559cdf6288215d0a81788543"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1ab4e13fe31be4d3559cdf6288215d0a81788543"}]},"branch":"refs/heads/stable/xena","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
