)]}'
{"id":"openstack%2Fkolla~845245","triplet_id":"openstack%2Fkolla~stable%2Fxena~I9f9e381716611b71584ecf8b371f8b7f45c694ee","project":"openstack/kolla","branch":"stable/xena","topic":"pin-dib-python36","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-06-09 10:57:22.000000000","reason":"Dr. Jens Harbott (frickler) replied on the change"},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-06-09 17:08:41.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2022-06-09 13:36:44.000000000","reason":"Maksim Malchuk replied on the change"}},"hashtags":[],"change_id":"I9f9e381716611b71584ecf8b371f8b7f45c694ee","subject":"Pin diskimage-builder version due to drop Python3.6","status":"MERGED","created":"2022-06-09 09:22:16.000000000","updated":"2022-06-09 17:09:55.000000000","submitted":"2022-06-09 17:08:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"845245-pin-dib-python36","meta_rev_id":"7c0a83a5159ed885d7fd2377105ffea28ab34150","_number":845245,"virtual_id_number":845245,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-09 17:08:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2022-06-09 10:57:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2022-06-09 11:49:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2022-06-09 11:49:11.000000000","_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2022-06-09 10:57:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-09 09:23:04.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2022-06-09 09:23:04.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2022-06-09 11:26:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-09 11:49:11.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"}],"messages":[{"id":"3e797a8813abdc4e738671dc72534798fe07d83b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-06-09 09:22:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4d8f7d051ebb53c586c3cad1a957ff15038c538e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-09 10:57:22.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"85d030e8b629de3d50948582994df5671a1619db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-09 11:26:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/a120942eef0a4d9ba081655ff7a94095 : SUCCESS in 26m 33s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/b1369af5868f4fac92734516d8e64cad : SUCCESS in 23m 59s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/9c6d546f17ce44aaab026971289aaa2b : SUCCESS in 1h 08m 26s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/8203bd36fef64594a09846ac7c9bb050 : SUCCESS in 1h 06m 49s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/25a16de0143d472fbf778b4a0f6887a4 : SUCCESS in 1h 23m 21s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ee0a8e3b7a8c4363a2c6b2259f3fe336 : SUCCESS in 18m 06s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/cfeb3742022946ac9c1500ba47df11fb : SUCCESS in 17m 58s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8fd388d4c9c74053bcdab369646c000b : SUCCESS in 59m 55s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4dfe5b2e03554488a6fc28152ba14f36 : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/922ff7b1a09240e7ba886b63eb4763de : SUCCESS in 3m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/72d7c8fca2f8409f96ead42bd25a7398 : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ca15fe1130a4995aa48a47b34931210 : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88efa2602a65481e878b148ac28be866 : SUCCESS in 5m 31s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ac7d92a32c394e399f5eb4c87c76710f : SUCCESS in 19m 23s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/db2433491fd54ea094bf6180f9672c8e : SUCCESS in 18m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a94d145924c74add858ae02bfa2549fd : SUCCESS in 45m 05s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/05a4646afe99499c994b8f160184d8e3 : SUCCESS in 51m 28s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/bc979c027048468eb129ccf9d693e471 : SUCCESS in 1h 14m 57s","accounts_in_message":[],"_revision_number":1},{"id":"ecf56b28a99d734ac95ff03a721976de7706d79b","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-06-09 11:49:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"83d972a3f35160186dc248ad21274ddcef38b73a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-09 11:49:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"49ba7a7d1d7822fbd1d53e29b7fe79f326181bdd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-09 12:15:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/395ee2bdb12e4876836dc53b3e7a124c : SUCCESS in 1h 05m 26s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7d2cfa4d96fc4926b9202a9f9803b7ed : SUCCESS in 1h 37m 54s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/723885257e654e048aab71533522821f : SUCCESS in 2h 50m 47s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8ff31504115b4d949e53cd9fd8fb7778 : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8e222b86074240579767c1429da9629d : SUCCESS in 6m 09s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/ab3ec5221f9744feae914653277245c3 : FAILURE in 1h 36m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"33f4b2ef335e95e0e74eb20580444567cdbedda5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-09 13:34:06.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/6f912aca0f1945ea80d842a458a964c4 : SUCCESS in 33m 50s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/5a011e0e55e44e3ea622b0d102553b4e : SUCCESS in 1h 14m 41s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/4182f1e7948741bb8dfdfa600de890fc : SUCCESS in 1h 14m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/247f7244ac5645368dbe58ea1c57999a : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ee531fe06784a538dc377b213424f97 : SUCCESS in 3m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e9ab4c707f28471fa918fb526da9937e : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edf881ad5f8e4359a5bdfe86192fc76e : SUCCESS in 5m 29s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/25eca5ab1e8d4495aef681c5f2967cee : SUCCESS in 24m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/039aed66407e44cfbac9588c4350731d : SUCCESS in 50m 05s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/403e965403c448fdae12a7e489b7a6f0 : FAILURE in 1h 38m 50s","accounts_in_message":[],"_revision_number":1},{"id":"ef018d251c716b69ca71669245781d10bf015968","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-06-09 13:36:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0282d8cfac4beba075ae921357a8ddd6a2909a50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-09 15:10:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/a5ca07503b3c45be8789979f75dfd4c1 : SUCCESS in 27m 35s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/4737e340a9f14bd8be6ffb4580f2a4fd : SUCCESS in 24m 47s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/b089f122d13f467bb5cb376425a276c6 : SUCCESS in 1h 16m 20s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/5ff0a3fa6a4e4e7886b6b8f32c83b2b1 : SUCCESS in 1h 09m 24s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/28eb706b4cc841dea1f500e1729ca644 : SUCCESS in 1h 27m 48s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/0e01c0b1e6934d24b8db42bca7e9aa91 : SUCCESS in 19m 16s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ab9eb5a360204f8693e631cc5f3807c8 : SUCCESS in 19m 13s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/28d700a0b8b84b698629092b3f49df45 : SUCCESS in 1h 02m 16s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ab67d18bbbb45bfade32b6a80e23111 : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f0fb50a9d61488b9480824b27997fe8 : SUCCESS in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/289d1b5fae9b4adc9366f99737a4b97f : SUCCESS in 3m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/01b3071c7df540129bce59e90a3aa03f : SUCCESS in 3m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1a196394ea84b95b66d0b9d1307ddad : SUCCESS in 5m 41s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/acd70cb4dac849b9b496d4bd0bb200fe : SUCCESS in 16m 43s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/cdde96cfcde2470eb2a7c0b4ae83f322 : SUCCESS in 19m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/22e9f56cac664e5789bc189e4f8c4c56 : SUCCESS in 47m 03s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/10e819dde7484160a1b99de184e40c50 : SUCCESS in 54m 36s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/059310a6d78f477dba949e1c8ee8d80e : SUCCESS in 57m 01s","accounts_in_message":[],"_revision_number":1},{"id":"33bb60878eee6726c540cbaa50e847002311926f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-09 15:10:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"23e358ee72d4c8de63800070dbd8fe51243cd099","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-09 16:08:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/d082ea7e7ec54d1497bd6e27204eac89 : SUCCESS in 2h 10m 33s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/204c7534dad74c7793ed64c682acfc8f : SUCCESS in 1h 25m 23s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/730f1d2d563248599244787af8242654 : SUCCESS in 2h 06m 36s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1c6d7aad4b0a4cea85e6ca3b05dc3df4 : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ce74f975fddb41bda0bf2be586965f91 : SUCCESS in 5m 56s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/1f594d559de34a3d8f7645bfe734faf1 : FAILURE in 1h 57m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b2f68b235c1204895328969dbbfc321fd9390d9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-09 17:08:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/4d30fca630534409ab1a53286bfdfd2f : SUCCESS in 28m 51s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/0f633885b2984a76a9417b0a14896d2b : SUCCESS in 1h 17m 48s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/fc442666d493405fa4dbb3d00b5db3f2 : SUCCESS in 1h 26m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf86f7c66ca7450e8afd921fabcae15e : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c4ab16183304e3d801425749239f6fd : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c270bf165cd740fb9547e44d573d921a : SUCCESS in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee186280b23f4481bafce859d57329bf : SUCCESS in 5m 24s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9e57757d84d6497b9a43996c4de495e4 : SUCCESS in 24m 26s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8229ef447b3a4eae998a784225081c2d : SUCCESS in 54m 57s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4e3092ee71ea44018ec48f6351c9cf34 : SUCCESS in 1h 04m 40s","accounts_in_message":[],"_revision_number":1},{"id":"cc10b324bdca549dae5db2e4531709ab3baa97a2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-09 17:08:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7c0a83a5159ed885d7fd2377105ffea28ab34150","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-09 17:09:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8a4367fab534bda95f563f56a6c1cf3 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5baeaf498e147e909524ac3ca094e7c987d18af6","revisions":{"5baeaf498e147e909524ac3ca094e7c987d18af6":{"kind":"REWORK","_number":1,"created":"2022-06-09 09:22:16.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/45/845245/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/45/845245/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/845245/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/845245/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/845245/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/45/845245/1"}}},"commit":{"parents":[{"commit":"e90786ab9affe99e3c6986f54a580844f1598e39","subject":"Allow to provide own repos.yaml file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e90786ab9affe99e3c6986f54a580844f1598e39"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-06-07 08:54:43.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-06-09 09:21:54.000000000","tz":180},"subject":"Pin diskimage-builder version due to drop Python3.6","message":"Pin diskimage-builder version due to drop Python3.6\n\nSince Ibdbcf459608711ac64e7fefb1707f6708d68e750 merged to master we\nneed to pin diskimage-builder to latest stable tag with support of\nthe Python3.6 until we move to CentOS9.\n\nChange-Id: I9f9e381716611b71584ecf8b371f8b7f45c694ee\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 1b259ecc5bbbe48e083a08d2e67f42887bc2eec3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5baeaf498e147e909524ac3ca094e7c987d18af6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5baeaf498e147e909524ac3ca094e7c987d18af6"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
