)]}'
{"id":"openstack%2Fkolla~845601","triplet_id":"openstack%2Fkolla~master~I4bd581c3ea422e1c9a075a71721413d47fdbdfe4","project":"openstack/kolla","branch":"master","topic":"openstack-exporter-1.6.0","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-06-22 05:08:37.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"30523":{"account":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"last_update":"2022-06-27 11:44:55.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-06-22 07:46:22.000000000","reason":"removed on reply"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-06-27 11:44:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4bd581c3ea422e1c9a075a71721413d47fdbdfe4","subject":"Bump prometheus-openstack-exporter version to 1.6.0","status":"MERGED","created":"2022-06-13 15:34:32.000000000","updated":"2022-06-27 11:46:32.000000000","submitted":"2022-06-27 11:44:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"845601-openstack-exporter-1.6.0","meta_rev_id":"d227b586e8a6e1c48b6ff01dc6912fcd0846bf85","_number":845601,"virtual_id_number":845601,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-27 11:44:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2022-06-23 08:39:19.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2022-06-23 08:32:01.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2022-06-27 10:19:49.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-13 17:05:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-13 18:01:48.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2022-06-20 08:56:54.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-06-20 21:03:33.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2022-06-23 08:39:19.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"8063e58f1aa66bbca003d9b3099487dee6133104","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-06-13 15:34:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2e2a3f0e56920cfac012cce5b4ad63d884af9532","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-06-13 15:40:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7be2667e3b3c557d475dbb8f4de400b41d5dc8a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-13 17:05:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f7ba1bda75be4634b01f2edec82559a3 : FAILURE in 20m 05s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ec33d5ced97e4889ac376033ffa5f258 : SUCCESS in 59m 09s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/f3fdbb0ed6ef427baf1133748e5701cd : SUCCESS in 1h 19m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fa9a2fbe2ec4e13b4f1b342c75866ee : SUCCESS in 3m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/41f78f975a464bb38476a2ce4baa8dec : SUCCESS in 2m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a54c36706a50443085445042669df48d : SUCCESS in 3m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/46002075c96d498384cf809f85986623 : SUCCESS in 3m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8125c4c41e1947fdb3396c8afda96c99 : SUCCESS in 4m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1b96b89575c54fb1a64de1b741f8bf3e : SUCCESS in 3m 18s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/640113715bad482d9fd974fcf8b23584 : FAILURE in 24m 55s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f516ef7e70d1497db5082be9c680fda7 : SUCCESS in 1h 15m 51s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/32b33e55af43456db9a488f8b2df25e3 : SUCCESS in 1h 00m 10s","accounts_in_message":[],"_revision_number":2},{"id":"0955af73b0597ac2359cc0799005d64b05813da0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-13 17:17:09.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ee7ed433c67d4d3bba4c2fd42c6dc67b : FAILURE in 30m 10s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/656f17a7740f480da293b37f4cc2e333 : SUCCESS in 1h 29m 51s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0bb3e00b10c24e9a8d325417841400db : SUCCESS in 4m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/43e4d3ecaeb740a0bb2089c596d0d849 : SUCCESS in 4m 11s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/5eddb9ebb335437b8f90a814e06f4a52 : FAILURE in 22m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"312b7fbc21e9bbbeca44331ae87273f6f78251ce","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-13 18:01:48.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"af216ed23f24dd1a8660e2e91095dc1780acc06a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-06-16 16:12:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"859503eae6cc08908df31e7095f4e806915fca34","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-06-16 16:13:35.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"4e5beaa34b4423e3c1a2fe6b54bd9c0de2c92b59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 17:33:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/10a89c9658d74f65844e0ce72dfa63a1 : SUCCESS in 21m 37s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ed70a22d96f0483aadb91de2ef2c2f8d : SUCCESS in 59m 41s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/a20a27cdd93646769aea2e30bcd726bd : SUCCESS in 1h 16m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d694c761693847d79fd8ebe6bc6b5163 : SUCCESS in 3m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/917e2532555d4a6cb1791a331038b357 : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/41bbe8224c224cb3b0790562e0be7a5c : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/89655d4162a54718b6c48b4d58c84685 : SUCCESS in 3m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d7f94ffe14f443bb0d486a34eb20fe2 : SUCCESS in 5m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df5bfe889917479c900337e43c9e6d2b : SUCCESS in 3m 59s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d8653bb593c74c54a5d62e1b8ca81b7b : SUCCESS in 21m 17s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/fdc246126a3f4dc4b9a2bbb36a32da75 : SUCCESS in 56m 16s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3c504a0b84ee49d3ba4dbd9b6532f5f0 : SUCCESS in 1h 06m 39s","accounts_in_message":[],"_revision_number":3},{"id":"5280ea65d3b481ccb88f464b043c1fbb48bd7d56","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 17:43:07.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ed37fb1b6b3246cdac04ab9563a9e76b : SUCCESS in 29m 36s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4a10b78bed3544c2b6d244388a0a7c0c : SUCCESS in 1h 22m 11s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7aa4cdfecbaf4a0a8e408e2a31a9f424 : SUCCESS in 4m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/64ffc23065904d059b363746a9d155b1 : SUCCESS in 4m 14s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/52edc36cde334508a0b86784caae20ad : FAILURE in 29m 03s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"879e4f9c9fd75a87e5a07e8201e8d5e085a9cf34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-06-16 21:37:19.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"d059614eadf802f8ee2e0929fc3ea6906c4d5eea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 22:42:09.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1c0fc88e1d1849e782c6975a8c4a7e56 : SUCCESS in 17m 21s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2fb0de7281f64cb0b72c946ea98d5282 : SUCCESS in 1h 01m 23s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/c73de59661894e378eef3fc94657d86d : SUCCESS in 1h 03m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b39161f132724543bc46188f67aa1d05 : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1306c5ab48c24374ad05201fe3e549ef : SUCCESS in 5m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7f3f51b57edc4095bcc71d777cc5b7d0 : SUCCESS in 5m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cff483c99285433ca44514ba43117e16 : SUCCESS in 3m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2640d63161d4c17b23c525662b34214 : SUCCESS in 5m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e57c2f0d15a04ad8a2dea9e5b5c064ea : SUCCESS in 5m 15s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c6f33a2127e74ee8bbd132213bde593b : SUCCESS in 24m 16s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/99e27e932e214d1b95a47b51a3569fec : SUCCESS in 52m 17s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/719df3c4f9574bb1ad6f02f960f997f6 : SUCCESS in 1h 00m 12s","accounts_in_message":[],"_revision_number":4},{"id":"1c33b3953884fb19bb75b0fe43e5b6b1b9b7094f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 23:17:12.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1ce34c333a97419e950a8967309d86f3 : SUCCESS in 1h 11m 06s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/72876bb92ece486782788a25dddfcec7 : SUCCESS in 1h 37m 53s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b043cbb3a3914d5e9b78b240fc0faf6d : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ec813678f90b4ea5953bfe0557155c1c : SUCCESS in 6m 53s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/ea9744008f6b4da49613ea5e6ffaaeb2 : FAILURE in 54m 13s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e66f8021808631ecdbdf75de8a063fe5c8969942","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-06-20 08:56:54.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4357c505085dab7ef20e004095a5328f2fbe52a0","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-06-20 10:12:53.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7680f39690c3f33259d70c688b7d843f9d3a679e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-22 05:08:37.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3901c41637af26512ad1bff928fa29988b9c719f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-06-22 07:46:22.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"de70233e0351874259c0801846d17feb5072aff4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-06-22 14:57:47.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"9bd26e4aa9d65806b7dbc887448f0e0249aee420","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 16:14:40.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4c369dd85daf4045ad83038f4ff4d288 : SUCCESS in 20m 36s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/43df080ad1a24eaebf16041ff6385a08 : SUCCESS in 49m 22s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/db7f8fa6d9ce4fb2ac6b78dc6de8c0fb : SUCCESS in 1h 08m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7205d42384984ffd92efc74061413907 : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/744d41995e034e598b8a849fa264e32e : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cd3f15e899cb4d8e9d81a4305ca9279e : SUCCESS in 3m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c1b6180a3a8e4dc29cdb82a8e07b0190 : SUCCESS in 3m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd2cd3da27c0435dbe61587a33f2822e : SUCCESS in 6m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/72258c11f70b4de0b829239dd3dd1247 : SUCCESS in 5m 16s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ed6de57f0b7d4e27857a4ec37fa19f48 : SUCCESS in 30m 13s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/50bbd27a04be40ca9d424ff2893d34f3 : SUCCESS in 55m 42s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1eb1e517db2342d890f285aac454c25f : SUCCESS in 1h 11m 47s","accounts_in_message":[],"_revision_number":5},{"id":"3dc72d9d99b0bc7b49b15fa6cb2a9ef08b2f4622","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 16:54:19.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/614cd311f0614919864902c4da14c2a8 : SUCCESS in 30m 23s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4540eda07599447d893de6fc45456bca : SUCCESS in 1h 54m 29s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/495e2b44990a4fb880e9b394e75f086d : SUCCESS in 4m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a17d2974cf37472d90677147530afd66 : SUCCESS in 4m 15s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/09260331bc4d4674bd4c3b1c2e45e246 : FAILURE in 29m 04s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e523d9118d857ef288f78b80307160284d2026c3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-06-23 08:32:01.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"049acd6be5d2019038065737135aa94098c4e675","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-06-23 08:39:19.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"07e47cf6b5f28d9333724534be03eb6c408ce502","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-06-27 09:56:25.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b53c2bee28acd298cdb7aa56554f8dc18631010f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-06-27 10:19:49.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"cfa1bc1ef31e78e32c05ac8ddc34d6af1a40976a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-27 10:20:17.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"9afe4baf17778693cdd61b9c0564e44f287bbb5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-27 11:44:55.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8ad1b4c42486403ea167f95f8c8876d9 : SUCCESS in 20m 17s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9f0bd2b8e1a04412812dc25c0cb1aef6 : SUCCESS in 1h 11m 12s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/825ce4d5c79046209615a17ee8a5478a : SUCCESS in 1h 18m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f928df277374d13803359b1177c6789 : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/742d1c8f071042f7a351e7267c208c82 : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9cda88d850d440e18a0b99126bb73407 : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2acae331abbb4dcfb5a2c387572e9400 : SUCCESS in 7m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/523d82edbbd549699c8d22f37ef316d8 : SUCCESS in 5m 39s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/53f5049f26b44c9ba84c76aba8472dcc : SUCCESS in 20m 37s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/77153347785d4a5599992218ffea8a42 : SUCCESS in 54m 27s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0f7c1d7956834be7a074522fbf4f9849 : SUCCESS in 1h 08m 55s","accounts_in_message":[],"_revision_number":5},{"id":"dad6d2dd2e15a803e1f2bb26656474d07fd8ffaa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-27 11:44:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"d227b586e8a6e1c48b6ff01dc6912fcd0846bf85","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-27 11:46:32.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c667643d2f964d06a06da35db1126921 : SUCCESS in 1m 11s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/269c0696cda4412abf508d404788d0e4 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"5f94e53bea9fb81786595cb8b452879afe436e26","revisions":{"19fb24c9c8313f1b0304c6d57c9b165fc152e1c8":{"kind":"REWORK","_number":1,"created":"2022-06-13 15:34:32.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/01/845601/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/01/845601/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/845601/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/845601/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/845601/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/01/845601/1"}}},"commit":{"parents":[{"commit":"09f6999cdcbae9f773cb7a7d84ece3d50becf2ad","subject":"Merge \"Add documentation to create monthly stable releases\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/09f6999cdcbae9f773cb7a7d84ece3d50becf2ad"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-06-13 15:33:47.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-06-13 15:33:47.000000000","tz":120},"subject":"Bump prometheus-openstack-exporter version to 1.6.0","message":"Bump prometheus-openstack-exporter version to 1.6.0\n\nChange-Id: I4bd581c3ea422e1c9a075a71721413d47fdbdfe4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/19fb24c9c8313f1b0304c6d57c9b165fc152e1c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/19fb24c9c8313f1b0304c6d57c9b165fc152e1c8"}]},"branch":"refs/heads/master"},"6d08f9ea5ef900ffbe068a94501073c1324e8986":{"kind":"REWORK","_number":2,"created":"2022-06-13 15:40:54.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/01/845601/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/01/845601/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/845601/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/845601/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/845601/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/01/845601/2"}}},"commit":{"parents":[{"commit":"09f6999cdcbae9f773cb7a7d84ece3d50becf2ad","subject":"Merge \"Add documentation to create monthly stable releases\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/09f6999cdcbae9f773cb7a7d84ece3d50becf2ad"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-06-13 15:40:07.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-06-13 15:40:07.000000000","tz":120},"subject":"Bump prometheus-openstack-exporter version to 1.6.0","message":"Bump prometheus-openstack-exporter version to 1.6.0\n\nChange-Id: I4bd581c3ea422e1c9a075a71721413d47fdbdfe4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6d08f9ea5ef900ffbe068a94501073c1324e8986"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6d08f9ea5ef900ffbe068a94501073c1324e8986"}]},"branch":"refs/heads/master"},"e7af71e8f04a0735a95e427944dc1fe68a739fcb":{"kind":"REWORK","_number":3,"created":"2022-06-16 16:12:27.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/01/845601/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/01/845601/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/845601/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/845601/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/845601/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/01/845601/3"}}},"commit":{"parents":[{"commit":"e8083e1def34294aa455c4d820ca7e9c9ec44fc9","subject":"Merge \"Switch OPENSTACK_RELEASE back to master\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e8083e1def34294aa455c4d820ca7e9c9ec44fc9"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-06-16 16:11:25.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-06-16 16:11:25.000000000","tz":120},"subject":"Bump prometheus-openstack-exporter version to 1.6.0","message":"Bump prometheus-openstack-exporter version to 1.6.0\n\nChange-Id: I4bd581c3ea422e1c9a075a71721413d47fdbdfe4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e7af71e8f04a0735a95e427944dc1fe68a739fcb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e7af71e8f04a0735a95e427944dc1fe68a739fcb"}]},"branch":"refs/heads/master"},"127bfde537b86aa55be814b19d1f59eef552ba88":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2022-06-16 21:37:19.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/01/845601/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/01/845601/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/845601/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/845601/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/845601/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/01/845601/4"}}},"commit":{"parents":[{"commit":"e8083e1def34294aa455c4d820ca7e9c9ec44fc9","subject":"Merge \"Switch OPENSTACK_RELEASE back to master\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e8083e1def34294aa455c4d820ca7e9c9ec44fc9"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-06-16 21:36:56.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-06-16 21:36:56.000000000","tz":120},"subject":"Bump prometheus-openstack-exporter version to 1.6.0","message":"Bump prometheus-openstack-exporter version to 1.6.0\n\nChange-Id: I4bd581c3ea422e1c9a075a71721413d47fdbdfe4\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/828902\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/127bfde537b86aa55be814b19d1f59eef552ba88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/127bfde537b86aa55be814b19d1f59eef552ba88"}]},"branch":"refs/heads/master"},"5f94e53bea9fb81786595cb8b452879afe436e26":{"kind":"REWORK","_number":5,"created":"2022-06-22 14:57:47.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/01/845601/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/01/845601/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/845601/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/845601/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/845601/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/01/845601/5"}}},"commit":{"parents":[{"commit":"9931f839136c28cd8cdd7a0e649b0efe1eb561ea","subject":"disable monasca for AArch64 on all distros","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9931f839136c28cd8cdd7a0e649b0efe1eb561ea"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-06-22 14:57:32.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-06-22 14:57:32.000000000","tz":120},"subject":"Bump prometheus-openstack-exporter version to 1.6.0","message":"Bump prometheus-openstack-exporter version to 1.6.0\n\nChange-Id: I4bd581c3ea422e1c9a075a71721413d47fdbdfe4\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/828902\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5f94e53bea9fb81786595cb8b452879afe436e26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5f94e53bea9fb81786595cb8b452879afe436e26"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
