)]}'
{"id":"openstack%2Fkolla~846257","triplet_id":"openstack%2Fkolla~stable%2Fyoga~I177949b9319a977c9cd9121eb28b710256b72a5a","project":"openstack/kolla","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"27446":{"account":{"_account_id":27446,"name":"Daniel Stoye","email":"d.stoye@cms.hu-berlin.de","username":"dasTor"},"last_update":"2022-06-20 21:03:27.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-06-20 21:03:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I177949b9319a977c9cd9121eb28b710256b72a5a","subject":"Bump skydive version to 0.28","status":"MERGED","created":"2022-06-17 16:46:05.000000000","updated":"2022-06-20 21:04:39.000000000","submitted":"2022-06-20 21:03:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"846257","cherry_pick_of_change":846257,"cherry_pick_of_patch_set":2,"meta_rev_id":"338c953fd75561a11edad700062279ebb4f589e6","_number":846257,"virtual_id_number":846257,"owner":{"_account_id":27446,"name":"Daniel Stoye","email":"d.stoye@cms.hu-berlin.de","username":"dasTor"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-20 21:03:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":2,"date":"2022-06-20 10:38:43.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2022-06-20 19:35:29.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-06-20 10:15:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":1,"date":"2022-06-20 19:35:29.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-17 16:54:52.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2022-06-17 18:24:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-20 09:42:03.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-06-20 19:35:29.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"}],"messages":[{"id":"c662996278ca09d006c251887ed32322f6efb231","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27446,"name":"Daniel Stoye","email":"d.stoye@cms.hu-berlin.de","username":"dasTor"},"date":"2022-06-17 16:46:05.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a8194b6ae13b6e6aa858c1cf11f6027275136807","tag":"autogenerated:gerrit:abandon","author":{"_account_id":27446,"name":"Daniel Stoye","email":"d.stoye@cms.hu-berlin.de","username":"dasTor"},"date":"2022-06-17 16:47:25.000000000","message":"Abandoned\n\nwait till merged in master","accounts_in_message":[],"_revision_number":1},{"id":"9938c9f4474cf00c3bf9acfc1b9a97508326e81e","tag":"autogenerated:gerrit:restore","author":{"_account_id":27446,"name":"Daniel Stoye","email":"d.stoye@cms.hu-berlin.de","username":"dasTor"},"date":"2022-06-17 16:54:12.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"f7f7a5930734d1ed5d44c05deeeb8bc399f8b1a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-06-17 16:54:30.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"c4a59e71b76209e4e268d0e2422de741061eab1b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-06-17 16:54:52.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dd55b6bafa9fb70ddd5dc10ff2176d0b5c589a8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 18:24:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/8efb9f87706e40e6b818b222967262fb : SUCCESS in 26m 54s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/03dc0a6824b9428ba28d14cc6882f5ed : SUCCESS in 33m 22s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/99c7dcf82c3c44e88e11b83d0004a079 : SUCCESS in 1h 13m 49s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/820975680aa8436083410d3ff9d6c391 : SUCCESS in 1h 11m 36s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/fb07e7d9579c4920b65def00c72bff0a : SUCCESS in 1h 23m 41s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e76dc6939ee04f2f9885d213f2b5e2ad : SUCCESS in 21m 25s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/44b0908c542e4e5893a6df661e9a23e5 : SUCCESS in 18m 06s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8c95e2e88b0146b5a14045902a217cd2 : SUCCESS in 1h 08m 31s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/fde40176864348fdbfcf430bdfd75f60 : SUCCESS in 1h 12m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/043d00164d684025ac2943a5ffea3cb9 : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de882b2f10214bc5bab544819b1a78c6 : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a747ee35df58499c8548cbb38a5ec43e : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/589948f2dcf841e3bc3ed451d51052ed : SUCCESS in 6m 08s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cf3072942884434f8e6b0218b62f26b5 : SUCCESS in 18m 23s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2ef8c19ef15b4a08ab9b6b7a5861fc18 : SUCCESS in 24m 42s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/02e6cffdeb8e4dbab4fac7a5a15da797 : SUCCESS in 51m 13s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/28b58f2fef554c478ac25d1caaae8767 : SUCCESS in 59m 51s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d52d37a595a341178e0683fa09127b73 : SUCCESS in 1h 02m 00s","accounts_in_message":[],"_revision_number":2},{"id":"0e1edbf73ce96d13cf00ddc9b53a69a0477d9fe7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 19:27:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/1d317f2ed2b341a695ec1275ea7f6d9a : SUCCESS in 2h 30m 49s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/749f7172b20946ca9c30fe9db5a28a06 : SUCCESS in 1h 59m 46s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/652ea82113e94aaf9823af61b3cfecd5 : SUCCESS in 2h 26m 45s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9eee6c0abd514c8c8dbcd9a7fb52873c : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c8f0f88b91b340569104baced659c654 : SUCCESS in 4m 27s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/d11394d9f5904f1986d4e7141c3c4752 : FAILURE in 30m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"924b584e0867ccdd1c78e09cfb09b28213976dee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27446,"name":"Daniel Stoye","email":"d.stoye@cms.hu-berlin.de","username":"dasTor"},"date":"2022-06-20 09:35:50.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6906e4d42ea85c36bc0d6af030b07909d347b3d8","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-06-20 09:42:03.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"73caf95dc79782f65b213a50d6d3fe518d10fa13","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-06-20 09:42:14.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ebe4621415c6d9d77f13116760d51dcdfd6782d8","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-06-20 10:14:16.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"38aa51efc6fc5daea0ed82f1cddd89daf64d5204","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-06-20 10:15:09.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3a1ad13ff07f902b0e20a83feb0f0f0e5b4e86cf","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-06-20 10:15:39.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e18d38a7feb4f67fdb2692e4866fe5b6357b8621","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-06-20 10:38:43.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"eb90b75fd3372982ad3949c3ec920195f681df14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 11:39:25.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/efb8d2c2538449f1a58aa8966beedf02 : SUCCESS in 25m 07s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/bbd1da20a26b464dac66b706315bd505 : SUCCESS in 28m 32s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/60ff3aa50d264832b268a330f78e6a02 : SUCCESS in 1h 15m 01s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/140c1e264d744e46a31c34a8b9f75b82 : SUCCESS in 1h 06m 45s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/6c49f4b63a9743a18831bf447ebddda9 : SUCCESS in 1h 21m 42s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f0a22782c7754affa0cd81ce7869f736 : SUCCESS in 23m 14s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b111d35d3d4c41e39aafbdc3981066da : SUCCESS in 17m 47s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/921f71e3424643e5a18990b566a0889d : SUCCESS in 56m 37s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/64622f8ec55743dfad4de79dff0a0ea5 : SUCCESS in 1h 11m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18fefecd0b904c0ba90dc6251929452f : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c469ab4cf1647e9aa10f619077ddc5f : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8f39b28db5c347edb657ff6a7eb7a3ac : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d6b48096c4b41ffb6b0f738e746330f : SUCCESS in 6m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c0f2fd66f374270894dcd18eacf0435 : SUCCESS in 4m 49s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f27ee77dacd34527bfaada45cba703f6 : SUCCESS in 19m 46s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1b1860f898954d9e8ce123dec996c7dc : SUCCESS in 21m 43s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/04fd1f76b6ab4296924b87b1c68e7fc3 : SUCCESS in 53m 38s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d0f54256fd5a459fbbdd2702c663b39c : SUCCESS in 56m 42s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c23e3e8a085b454db908c4121396f1fa : SUCCESS in 1h 03m 17s","accounts_in_message":[],"_revision_number":4},{"id":"1176dec17a53ca12c20ef76d20b94a3ba4824917","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 12:23:32.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/735191de86724e97ae1993fae075aac4 : SUCCESS in 2h 06m 07s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e98e89e7a3b541c4a51d5d4ea3c64c8f : SUCCESS in 51m 37s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2ac3453323bd447e871af23526d8f83f : SUCCESS in 1h 45m 07s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/61561f6c14f64682a02cff9a7210295d : SUCCESS in 6m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/48dacd0704974a97a6939d097ca32e90 : SUCCESS in 9m 57s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/8d792e999a7f428fa5fbc0c6f537b33b : FAILURE in 49m 04s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1b1a73ea123b4b83cc6668cc7c9a490c0ee8e55e","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2022-06-20 19:35:29.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"05aad273ac021706cd9257b79e530a87415f0040","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 19:35:58.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ace1a4d44dcd7de62721423d0ee0bbb9ef09d39c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 21:03:27.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/609501620dde4e3ba2abd9216e489c28 : SUCCESS in 27m 40s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/92e77577bbc243c894c56fb1fbfd3a56 : SUCCESS in 1h 04m 31s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/43d09d2146954f0fb755dd5d74ac48d4 : SUCCESS in 1h 26m 40s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/8bf414bbd8ee457498d0b49ddc9b8c35 : SUCCESS in 28m 05s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c6407421d716437e915bf5e5164cbcc9 : SUCCESS in 1h 04m 34s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/6d3987e934ef4ab587cf77cc70445f37 : SUCCESS in 1h 17m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f8d72b5958d451085c50b6fa88b8793 : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a12a53cb2a75438e8b01ea050530e93a : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/935c38975e214bed9cd23e51b1deece1 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b61f07ab91040819eee19841fae247b : SUCCESS in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/20911e298d9c4289829f21dfdf92d10b : SUCCESS in 3m 37s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8a892da0c3bb4708982824369ba1b12a : SUCCESS in 20m 33s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/295688512ea54e35995982f3fd9f7a56 : SUCCESS in 48m 06s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/65ff5bb4643142569b93fb9f054da11a : SUCCESS in 1h 00m 53s","accounts_in_message":[],"_revision_number":4},{"id":"f7bbb37e29b9cef859a6225f935bf7807bb3a15b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 21:03:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"338c953fd75561a11edad700062279ebb4f589e6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 21:04:39.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e211748a9b9f4858905395603817b1cc : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/31faa869163f45b5b69d6a7f9cc7878b : SUCCESS in 51s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"66eec33571fdcc6e6d95a284e11925d626329bf1","revisions":{"bb1122a4222e2c0b8b67f8262ca686e9b7d99f46":{"kind":"REWORK","_number":1,"created":"2022-06-17 16:46:05.000000000","uploader":{"_account_id":27446,"name":"Daniel Stoye","email":"d.stoye@cms.hu-berlin.de","username":"dasTor"},"ref":"refs/changes/57/846257/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/846257/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/846257/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/846257/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/846257/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/846257/1"}}},"commit":{"parents":[{"commit":"34cc46a4d04db0b4988b52b25287c62e36843d6e","subject":"Don\u0027t ignore errors of \u0027apt-get update\u0027 command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/34cc46a4d04db0b4988b52b25287c62e36843d6e"}]}],"author":{"name":"Daniel Stoye","email":"d.stoye@cms.hu-berlin.de","date":"2022-06-17 14:23:11.000000000","tz":120},"committer":{"name":"Daniel Stoye","email":"d.stoye@cms.hu-berlin.de","date":"2022-06-17 16:46:05.000000000","tz":0},"subject":"Bump skydive version to 0.28","message":"Bump skydive version to 0.28\n\nSkydive versions prior to 0.28.0 panic on newer versions of libc.\nFixed upstream in 0.28, see: https://github.com/skydive-project/skydive/issues/2329\nThis should be backported to at least xena and yoga, as skydive is\ncurrently not working with centos 8 on these releases.\n\nCloses-Bug: #1940862\nChange-Id: I177949b9319a977c9cd9121eb28b710256b72a5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bb1122a4222e2c0b8b67f8262ca686e9b7d99f46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bb1122a4222e2c0b8b67f8262ca686e9b7d99f46"}]},"branch":"refs/heads/stable/yoga"},"46bdc5c9c0e4af9ed0a9851456a07e6bf2ee1b44":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-06-17 16:54:30.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/57/846257/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/846257/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/846257/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/846257/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/846257/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/846257/2"}}},"commit":{"parents":[{"commit":"34cc46a4d04db0b4988b52b25287c62e36843d6e","subject":"Don\u0027t ignore errors of \u0027apt-get update\u0027 command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/34cc46a4d04db0b4988b52b25287c62e36843d6e"}]}],"author":{"name":"Daniel Stoye","email":"d.stoye@cms.hu-berlin.de","date":"2022-06-17 14:23:11.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-06-17 16:52:28.000000000","tz":120},"subject":"Bump skydive version to 0.28","message":"Bump skydive version to 0.28\n\nSkydive versions prior to 0.28.0 panic on newer versions of libc.\nFixed upstream in 0.28, see: https://github.com/skydive-project/skydive/issues/2329\nThis should be backported to at least xena and yoga, as skydive is\ncurrently not working with centos 8 on these releases.\n\nCloses-Bug: #1940862\nChange-Id: I177949b9319a977c9cd9121eb28b710256b72a5a\n(cherry picked from commit 88421554e58bf4410d5a93992e178a9f2728c073)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/46bdc5c9c0e4af9ed0a9851456a07e6bf2ee1b44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/46bdc5c9c0e4af9ed0a9851456a07e6bf2ee1b44"}]},"branch":"refs/heads/stable/yoga"},"a3dddb0536776c82699b7fe8ea6032068f6dc90f":{"kind":"NO_CHANGE","_number":3,"created":"2022-06-20 09:35:50.000000000","uploader":{"_account_id":27446,"name":"Daniel Stoye","email":"d.stoye@cms.hu-berlin.de","username":"dasTor"},"ref":"refs/changes/57/846257/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/846257/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/846257/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/846257/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/846257/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/846257/3"}}},"commit":{"parents":[{"commit":"34cc46a4d04db0b4988b52b25287c62e36843d6e","subject":"Don\u0027t ignore errors of \u0027apt-get update\u0027 command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/34cc46a4d04db0b4988b52b25287c62e36843d6e"}]}],"author":{"name":"Daniel Stoye","email":"d.stoye@cms.hu-berlin.de","date":"2022-06-17 14:23:11.000000000","tz":120},"committer":{"name":"Daniel Stoye","email":"d.stoye@cms.hu-berlin.de","date":"2022-06-20 09:35:50.000000000","tz":0},"subject":"Bump skydive version to 0.28","message":"Bump skydive version to 0.28\n\nSkydive versions prior to 0.28.0 panic on newer versions of libc.\nFixed upstream in 0.28, see: https://github.com/skydive-project/skydive/issues/2329\nThis should be backported to at least xena and yoga, as skydive is\ncurrently not working with centos 8 on these releases.\n\nCloses-Bug: #1940862\nChange-Id: I177949b9319a977c9cd9121eb28b710256b72a5a\n(cherry picked from commit 88421554e58bf4410d5a93992e178a9f2728c073)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a3dddb0536776c82699b7fe8ea6032068f6dc90f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a3dddb0536776c82699b7fe8ea6032068f6dc90f"}]},"branch":"refs/heads/stable/yoga"},"66eec33571fdcc6e6d95a284e11925d626329bf1":{"kind":"REWORK","_number":4,"created":"2022-06-20 10:15:09.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/57/846257/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/846257/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/846257/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/846257/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/846257/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/846257/4"}}},"commit":{"parents":[{"commit":"34cc46a4d04db0b4988b52b25287c62e36843d6e","subject":"Don\u0027t ignore errors of \u0027apt-get update\u0027 command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/34cc46a4d04db0b4988b52b25287c62e36843d6e"}]}],"author":{"name":"Daniel Stoye","email":"d.stoye@cms.hu-berlin.de","date":"2022-06-17 14:23:11.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-06-20 10:15:02.000000000","tz":120},"subject":"Bump skydive version to 0.28","message":"Bump skydive version to 0.28\n\nSkydive versions prior to 0.28.0 panic on newer versions of libc.\nFixed upstream in 0.28, see: https://github.com/skydive-project/skydive/issues/2329\nThis should be backported to at least xena and yoga, as skydive is\ncurrently not working with centos 8 on these releases.\n\nCloses-Bug: #1940862\nChange-Id: I177949b9319a977c9cd9121eb28b710256b72a5a\n(cherry picked from commit 1f1214d4fe496ca015497d7e2fb5aa994c805736)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/66eec33571fdcc6e6d95a284e11925d626329bf1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/66eec33571fdcc6e6d95a284e11925d626329bf1"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
