)]}'
{"id":"openstack%2Fkolla~846342","triplet_id":"openstack%2Fkolla~master~If95a8a41d2caf6738d4ccb9ddb651a06018743f4","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-06-20 10:42:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If95a8a41d2caf6738d4ccb9ddb651a06018743f4","subject":"Fix error message when git clone fails","status":"MERGED","created":"2022-06-17 11:44:07.000000000","updated":"2022-06-20 10:44:18.000000000","submitted":"2022-06-20 10:42:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"846342","meta_rev_id":"633baccff3219499e5d14f4230c2f2b2cfc2a31c","_number":846342,"virtual_id_number":846342,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-20 10:42:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2022-06-20 08:18:50.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2022-06-20 08:41:15.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2022-06-20 08:41:15.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2022-06-20 08:41:15.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-17 13:09:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-17 13:11:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-20 08:18:50.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2022-06-20 08:41:15.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"1a31afe07fca38e1a9d1d4606affe1c6fb2ebf54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-06-17 11:44:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7fb94c7074baeaadf3971e1317820967235af7f7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 13:09:43.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3e6712fbeb6c4e44bbea5d4fed25b81b : FAILURE in 31m 04s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0d2b7fcecd354108b8c582e6843ef50f : SUCCESS in 1h 17m 16s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cc1f77e19db4428d8e808a8a64580187 : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/76663dcb5cf34af898a4ae9a6cdeca29 : SUCCESS in 4m 56s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/26af8b15b51c4c40b6ae69597273de68 : FAILURE in 33m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b3f24de895bf6408b672850969dcc5db26f9c421","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 13:11:25.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/688dcb1ca6824b378d70c0ee5e47a57f : SUCCESS in 22m 41s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3bbcc5cbcdf14bb0bb498ed216906f08 : SUCCESS in 1h 05m 09s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/d4107c1c0d244347b43a17f1bcb5d569 : SUCCESS in 1h 22m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0664ff1238af4669bf960b52e780e22c : FAILURE in 2m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea332d7131504ce6a19129c27db32466 : SUCCESS in 3m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a0befd5fe3b04698ad31afdda31eaf41 : SUCCESS in 3m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/418fd68d5f6a44c8b04cc72ed4b0ddc1 : SUCCESS in 3m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b2c351e8a3d40faa7946270ff12225a : SUCCESS in 5m 30s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/bc92d4421cf640e58a1c6bd963f1beb6 : SUCCESS in 21m 03s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2db7f10f037a4f9091db91dce3524bf0 : SUCCESS in 51m 05s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1e86512620c34da3bb6d3b4e76fa1b56 : SUCCESS in 1h 01m 38s","accounts_in_message":[],"_revision_number":1},{"id":"9f54942f3fb3fe36caef466b925fa36b05f1f4d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-06-20 08:08:11.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"a4b0cda5dec0e82d972579b29446114e8efa5955","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-06-20 08:18:50.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"56516b664704d071c960c61a3fa15c6a8820b72b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-06-20 08:41:15.000000000","message":"Patch Set 2: Backport-Candidate+2 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5bdf52636d9e786877d4b3cf1eaf51bf11e44ce0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 09:27:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/be8e06fb9dcb4e6796765acaf2406e00 : SUCCESS in 26m 27s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2359a71df0514973b9830f5733068a35 : SUCCESS in 58m 06s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/433cdef625aa4cc8b8c9fb770ca7b37f : SUCCESS in 1h 14m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f12b09626db4dc7b103b09d7dad0388 : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5fc4c0fae06b4ea3995905013870eb18 : SUCCESS in 3m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b2793eef52ca4fa5887b6ef542b706b8 : SUCCESS in 3m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cb81cddbb46c4f3eba7af4cd69d26340 : SUCCESS in 3m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b78ed7c5dec8499ab948711e344a765f : SUCCESS in 6m 24s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ad0c591e3dca4797afb9ab09cc04c0f9 : SUCCESS in 26m 22s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1e0a5f8692404526bba649cd13c1f6db : SUCCESS in 58m 04s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f18cc93d4d24455b926f3209cc02b350 : SUCCESS in 1h 03m 55s","accounts_in_message":[],"_revision_number":2},{"id":"b66ff24b7fbf94f4712808e6e4a81b17ecd00fac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 09:27:39.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"825c77cc5460047b6d64aae98be0d11cb643a405","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 09:54:04.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/355404d9dd2843feb004ca00964c2f34 : FAILURE in 40m 36s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1c194502db6e43b198dbb887f54f4363 : SUCCESS in 1h 28m 33s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/000dfc49e8d24a4ba721dd96de2f4423 : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b03bb6f7ca5d4b389bcb36f545f23a87 : SUCCESS in 4m 50s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/b6fd933d263f41778562081174b0b9e9 : FAILURE in 41m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"096f006701d60eff921acaeec50daf26421fcd7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 10:42:44.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9e2027ee7b7e48fd8c35101f72f6f6b3 : SUCCESS in 25m 05s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f93a22c9d5814f1fac9e1a9561d470b8 : SUCCESS in 59m 50s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/1f26184b39d747218c279ff24e731585 : SUCCESS in 1h 13m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/872b98713fe74bc1b95b120bd5087017 : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed451de14984464cb393c700875d676a : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/afe6f27d763441a093d1f70c4fdd0258 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed513cb44bf646e8a72cb18ae1192863 : SUCCESS in 6m 14s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ffba627398704bc1938e85894f52dd12 : SUCCESS in 20m 36s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3d635380d57d4d4d811cc05c46c9623a : SUCCESS in 52m 10s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2b57bf20d4224c859a0906ad911cf0a5 : SUCCESS in 1h 03m 45s","accounts_in_message":[],"_revision_number":2},{"id":"0623fc83a972114b997032c2c686f13d35287068","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 10:42:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"633baccff3219499e5d14f4230c2f2b2cfc2a31c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 10:44:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdb0904cfb13423daec1de8ef2c729d7 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"cdee63e415c6ccd70a3ebcf58cf4afc9bba28537","revisions":{"433025950a333d099d32d658e5608e7c6f891e18":{"kind":"REWORK","_number":1,"created":"2022-06-17 11:44:07.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/42/846342/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/846342/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/846342/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/846342/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/846342/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/846342/1"}}},"commit":{"parents":[{"commit":"3e35fede9adc622610451b7b5a7f8a45b933473b","subject":"Merge \"doc: get rid of remainders of binary images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e35fede9adc622610451b7b5a7f8a45b933473b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-06-17 11:43:50.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-06-17 11:43:52.000000000","tz":60},"subject":"Fix error message when git clone fails","message":"Fix error message when git clone fails\n\n1. There was no %s for the image.name\n2. The image name is in the logger, so the source is more useful (which\n   may be a plugin or addition)\n\nTrivialFix\n\nChange-Id: If95a8a41d2caf6738d4ccb9ddb651a06018743f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/433025950a333d099d32d658e5608e7c6f891e18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/433025950a333d099d32d658e5608e7c6f891e18"}]},"branch":"refs/heads/master"},"cdee63e415c6ccd70a3ebcf58cf4afc9bba28537":{"kind":"REWORK","_number":2,"created":"2022-06-20 08:08:11.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/42/846342/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/846342/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/846342/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/846342/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/846342/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/846342/2"}}},"commit":{"parents":[{"commit":"3e35fede9adc622610451b7b5a7f8a45b933473b","subject":"Merge \"doc: get rid of remainders of binary images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e35fede9adc622610451b7b5a7f8a45b933473b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-06-17 11:43:50.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-06-20 08:08:11.000000000","tz":0},"subject":"Fix error message when git clone fails","message":"Fix error message when git clone fails\n\n1. There was no %s for the image.name\n2. The image name is in the logger, so the source is more useful (which\n   may be a plugin or addition)\n\nTrivialFix\n\nChange-Id: If95a8a41d2caf6738d4ccb9ddb651a06018743f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cdee63e415c6ccd70a3ebcf58cf4afc9bba28537"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cdee63e415c6ccd70a3ebcf58cf4afc9bba28537"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
