)]}'
{"id":"openstack%2Fkolla~847278","triplet_id":"openstack%2Fkolla~master~I4db7346b7b8da330f5140e0b3ca660a5cea6acf3","project":"openstack/kolla","branch":"master","topic":"jammy","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-07-19 10:19:02.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I4db7346b7b8da330f5140e0b3ca660a5cea6acf3","subject":"ubuntu: add tzdata","status":"ABANDONED","created":"2022-06-23 08:54:12.000000000","updated":"2022-07-19 10:19:02.000000000","total_comment_count":1,"unresolved_comment_count":1,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"5ea076c581822613e887476aaed3a66d29d2d5bb","_number":847278,"virtual_id_number":847278,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-06-23 10:12:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":-1,"date":"2022-06-23 09:24:49.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"reviewer_updates":[{"updated":"2022-06-23 09:24:49.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2022-06-23 10:12:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7c574294e807dbab7d964e7a4ea2769451d2a068","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-06-23 08:54:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c384d8b7035180400da1e4a34d5cc5cd17b99b5f","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-06-23 09:24:49.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"574e151daf551d98d103f7315aec33b1eb25c518","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-06-23 09:48:30.000000000","message":"Topic set to ubuntu-jammy","accounts_in_message":[],"_revision_number":1},{"id":"b5507d533e0b5925dbaaf9247cd5cb15d8a88228","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-06-23 10:00:44.000000000","message":"Topic ubuntu-jammy removed","accounts_in_message":[],"_revision_number":1},{"id":"908054d82b41a2a15fd58c5aa87cf412182d4719","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-06-23 10:00:47.000000000","message":"Topic set to jammy","accounts_in_message":[],"_revision_number":1},{"id":"46fcfc8e315c564cc4d3cbf5bf31c0085431ef82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-23 10:12:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5b5531f597444e17b911dbf38eed52f2 : SUCCESS in 23m 33s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/fd05dc262119487997ca5b4a7dfe5bea : SUCCESS in 1h 01m 40s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/cc1b5b2e6e9049789dd47664ff3fff27 : SUCCESS in 1h 13m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4cb8c24e7c44e22b4e95c5abdffb551 : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c48be80eaad417e90b760411bb82f86 : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be60cdede295408984e14606d97cb930 : SUCCESS in 3m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f912bdfa18fa4286a41926873c48f14b : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d401b49124a240ca915116c891c88344 : SUCCESS in 5m 33s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e71edfacb2db4fcf9834b4100766513c : SUCCESS in 26m 47s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9b71b043acd64591a06b538adce6c7ce : SUCCESS in 50m 52s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/580afbe767194e0aaaab9d85c84390e9 : SUCCESS in 1h 10m 34s","accounts_in_message":[],"_revision_number":1},{"id":"8e3cb496d0443eb4b8dfd87c529e136a8de4e340","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-23 11:25:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/28b6311c92574abc9140ed620e9eb5d8 : SUCCESS in 2h 28m 53s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9e2ab837be764f9ba7a29e9cdbad3734 : SUCCESS in 1h 27m 03s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1956f11cd6f940518b214b143b8c8ddc : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2364e85efc914232be8e24f4183289eb : SUCCESS in 6m 24s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/afcead562558472db25d3439a2f81246 : FAILURE in 1h 44m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5ea076c581822613e887476aaed3a66d29d2d5bb","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-07-19 10:19:02.000000000","message":"Abandoned\n\nnot needed anymore","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cfc72c1c790ee11a8295a3bb929e580ca2844bbf","revisions":{"cfc72c1c790ee11a8295a3bb929e580ca2844bbf":{"kind":"REWORK","_number":1,"created":"2022-06-23 08:54:12.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/78/847278/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/78/847278/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/847278/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/847278/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/847278/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/78/847278/1"}}},"commit":{"parents":[{"commit":"9931f839136c28cd8cdd7a0e649b0efe1eb561ea","subject":"disable monasca for AArch64 on all distros","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9931f839136c28cd8cdd7a0e649b0efe1eb561ea"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-06-23 08:46:53.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-06-23 08:54:08.000000000","tz":120},"subject":"ubuntu: add tzdata","message":"ubuntu: add tzdata\n\nIn Jammy /etc/localtime is a symbolic link to /usr/share/zoneinfo\n(which is in tzdata package) - and bindmount /etc/localtime fails.\n\nChange-Id: I4db7346b7b8da330f5140e0b3ca660a5cea6acf3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cfc72c1c790ee11a8295a3bb929e580ca2844bbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cfc72c1c790ee11a8295a3bb929e580ca2844bbf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
