)]}'
{"id":"openstack%2Fkolla~847368","triplet_id":"openstack%2Fkolla~master~I62a34256fb7395d0448af169ac3cf036f60cd290","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2022-07-26 17:25:49.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-07-26 17:25:49.000000000","reason":"Change was submitted"},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-07-21 12:48:53.000000000","reason":"\u003cGERRIT_ACCOUNT_24072\u003e replied on the change","reason_account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-07-20 17:53:41.000000000","reason":"\u003cGERRIT_ACCOUNT_30491\u003e replied on the change","reason_account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}}},"hashtags":[],"change_id":"I62a34256fb7395d0448af169ac3cf036f60cd290","subject":"Install etcd binary from github","status":"MERGED","created":"2022-06-23 15:54:44.000000000","updated":"2022-07-26 17:26:52.000000000","submitted":"2022-07-26 17:25:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":25,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"847368","meta_rev_id":"40f113708d71daff1dd99b85d1cb7c6b28865f15","_number":847368,"virtual_id_number":847368,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-26 17:25:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-07-21 12:48:53.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2022-07-20 19:01:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2022-07-20 19:42:00.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2022-07-21 12:48:53.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-23 16:05:45.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2022-06-23 17:01:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-23 17:14:38.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-06-24 07:54:59.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-07-18 08:05:23.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2022-07-20 17:54:27.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"CC"},{"updated":"2022-07-20 19:01:38.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"9bc5cc6639ea25cae1f9f0af7aa949a60c4a7889","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2022-06-23 15:54:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e5e65a1be23c1946811a3fc9379a4a0fd025f498","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-06-23 16:05:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fcc34e0ba74c84de1c8887ef3cfd929afc4378e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-23 17:01:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/31c9993527134aad9cbc1852bcb29b7d : FAILURE in 21m 14s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c7bd435ba481400295debccec4392603 : SUCCESS in 47m 30s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/189d32351bef461cbcb1a6f22337dc9b : SUCCESS in 59m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da4cfe69ef4044778322d5f305fb9908 : SUCCESS in 3m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d33192c9e7a4f28b1e1fc027258fe3b : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bbcd8d604b534caeaaac46dff373e6e7 : SUCCESS in 3m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5d5411a4340e4fd2ac4f70669a22cc15 : SUCCESS in 3m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de72cd63b32b4c109c11f5fa15aad3a4 : SUCCESS in 5m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae341bb9c5f147f292d8abb2545c61ba : SUCCESS in 4m 27s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/513eae8bb182492d8df05aa2102d48fb : FAILURE in 27m 16s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0171101f52ab4acc8c73fd6b9c6e9231 : SUCCESS in 50m 30s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/65f9d598a88a4d5b95ea65bb0da535cb : SUCCESS in 1h 05m 10s","accounts_in_message":[],"_revision_number":1},{"id":"f4ae152fbc6754d2ef84afc6c072b7653297a8f3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-06-23 17:14:38.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"c560bb611ae2f433082d3259a3d6f422d961407c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-23 17:55:06.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/66b58408b130467180561295c1975e6a : FAILURE in 1h 30m 37s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/05c5d4ecee1647efae6e8689a4f0aa5b : SUCCESS in 1h 58m 29s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ba8e512fff7d4923b2dba01841fdfc27 : SUCCESS in 7m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5ca1ed1939b9464bbb3ad9e9b0f6eb5a : SUCCESS in 7m 10s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/ab44d36d260441ea90d7e0d94c272a49 : FAILURE in 1h 23m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bdfba0b1bce67ed456efdd6f27476883a07e2360","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-06-24 07:54:59.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a1fa3f9c22c29f3ec95a519b17e53af9baf52f04","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-07-18 08:05:23.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e1c89d78e4a6033fa62578b48fc5d74729b2d6bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2022-07-20 10:20:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9e52135aac6da029dc45ffc807a3add68f4754af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2022-07-20 10:22:36.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1f64aaf9a9f30b966ac8e8358004a63f8fb3e78f","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2022-07-20 10:25:23.000000000","message":"Patch Set 2:\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"b09819cb90da11b9d61b811368cd819511944d17","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-07-20 10:44:49.000000000","message":"Patch Set 3: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"d0dedbd2679ce88c478e7397f1f5f4b022ea49c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2022-07-20 11:20:12.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a669978c68789bf681626d0fabfb9a86fde40824","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2022-07-20 11:20:26.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"1379c98b7f3b2625d36b8499515d4ce9fa6bc1af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2022-07-20 11:21:24.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"7b8eeff076daf7b5fc784c0c394d4e94425472b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 12:45:02.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63df5e886ee54b11baedd4d5c38dccf7\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/01a0cad518bb42f690d0f0a017348243 : SUCCESS in 23m 55s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6925b226ff784f838ece33df5f24565b : SUCCESS in 1h 06m 06s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/475c973a5e6241cea863a934c94b7e05 : SUCCESS in 1h 17m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b271f79934da4588832dddf90b5e84ee : SUCCESS in 2m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f966bc8ba066487d8de2a180f5313299 : SUCCESS in 3m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/35c81d8b2e184ca4ab0659c874515d2e : SUCCESS in 3m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/da2146e71ee14f20ac2444da98640de5 : SUCCESS in 3m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47f5bab8dce444b48a9e4001d116475a : SUCCESS in 5m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2eefb4e6a4644939a38a633fa2226c04 : SUCCESS in 3m 22s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7633db2c8476408fa724c4ed918bfbbb : SUCCESS in 21m 42s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/632a3033b52140258157505b17eae76d : SUCCESS in 57m 01s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6664e2a0c0424f2f83a8f888bb321b10 : SUCCESS in 1h 05m 30s","accounts_in_message":[],"_revision_number":5},{"id":"32274f6a411f61bc5e6065a79dfe5bc5a20d00ea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 14:12:18.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e74b1797c728462183731626dffb449d\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/af758111b93445d59117fb2df3ab3fb8 : SUCCESS in 1h 49m 50s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e2b6dfe442ff496180317a7c7c2f348c : SUCCESS in 2h 38m 44s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/032f73c319234dac8525f13ae0486929 : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5d119b1132c34eb0a0fce0d8382bac0b : SUCCESS in 6m 28s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/e2bd0dbe5d074c70a01dfbd943485560 : FAILURE in 1h 17m 00s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3ee5733523da7abd5604ce9f4705d9e9443b2dd9","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-07-20 17:53:41.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"62014d621aa6057d5465d074959133ae70e8588c","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 18:48:30.000000000","message":"Patch Set 5:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fc60ea909e740d3bf9c327e23a8679d\n\n- kolla-build-no-infra-wheels-centos8s https://zuul.opendev.org/t/openstack/build/a550dca6dcb64623af4b9e54ce9ac1d9 : FAILURE in 14m 50s\n- kolla-ansible-centos8s-source-zun https://zuul.opendev.org/t/openstack/build/bf13c1be5c4a42c5aedf07894738c107 : FAILURE in 13m 41s (non-voting)\n- kolla-build-no-infra-wheels-debian https://zuul.opendev.org/t/openstack/build/2af35bb2fbe04a778a9a4980cf7b4fce : SUCCESS in 20m 54s\n- kolla-build-no-infra-wheels-ubuntu https://zuul.opendev.org/t/openstack/build/ee3f16554eae436db5aa4d262184bd18 : SUCCESS in 20m 02s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/725474a2747c4bf89a0b4505b8a1f939 : FAILURE in 47m 15s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"52e195bbcd6492bf428abe41a23f23596700d9f6","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-07-20 19:01:38.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"17a1a1cd5dfb9f2ce9fda1bdef56d0956b8dead3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-07-20 19:42:00.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c7533979e5fb3d3e8b0c21d9c8b4584627571d53","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-07-21 12:48:53.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"8952e32644412fd47f281f7a0ed48f3b45adbcba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 12:49:17.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"81aa12d6d2064212ebae7537b8d6488ef1f65230","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 13:26:06.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e3b60b4c17b943ca85af9711cd38d6d4\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/710a4bf678174214ab167b4c18f497ac : FAILURE in 10m 54s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f92a35cb81c3448d83922f3ada805c7c : FAILURE in 19m 17s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/b9fb35c21435414792b8dfa2bdc06498 : FAILURE in 19m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a317aa3b27f34fea905e4ff7c330a027 : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b3ba68f4c3c45168d92403e91c16223 : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf932ae1416e487b9a1dec885433834e : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9850421f5c042b18f6dd1ef50028b2e : SUCCESS in 5m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/80a2898418af4cdab2053c3716307003 : SUCCESS in 4m 32s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/799d747907aa4b07abd9b1eba854a748 : FAILURE in 11m 02s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a379ed124892464abc1da2b779c14e6d : FAILURE in 17m 20s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/bb61fac7e25f473dac7a8f92795d8610 : FAILURE in 15m 42s","accounts_in_message":[],"_revision_number":5},{"id":"66cd084c80d8aada44effe5a275e8753be6342d1","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-07-21 15:17:39.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"cedf80ded173f80bdf2646bfcb96b8c378396687","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 15:47:30.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/926ddf70e86f40f9a9521410b9f90030\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ba8e60e1adc44c78ba5a3fd636448b14 : FAILURE in 9m 52s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f72bde5451cc4d7fb8af01dd576808f0 : FAILURE in 23m 01s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/dc084ff4cfc048f380faa9653d52bab3 : FAILURE in 21m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58c114455b5b411899a1455642fcee27 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80ed77453e7d40939eedec422eb71827 : SUCCESS in 3m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f9e49b6a17cd4954bb36e67aa66de2ba : SUCCESS in 4m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3b0254e6f969419ab5c8615613bffbd8 : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fe03b26885f48379596777ee76fcdef : SUCCESS in 5m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b6cac1e6fb9d4e50bf5c01c4365f14dd : SUCCESS in 4m 25s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b9c7fba6e12b4fd586ee77500883dae2 : FAILURE in 10m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/83ecdf30e5a44c6f8bae4cbdc040099f : FAILURE in 17m 03s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8a027c2240564826b19218be6c1d929f : FAILURE in 16m 06s","accounts_in_message":[],"_revision_number":5},{"id":"830663354f9ec309bb33f696b44b84cc9facb823","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 15:48:51.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6527ae0f67204ed78016f218239d578f\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1ecce3dcb7b24a49a8a44d5e816f2d06 : FAILURE in 28m 04s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6c57231aba114076b310ee1cd8c3042c : FAILURE in 30m 31s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/179de829ddf44eeba9618182cb951d0d : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2f6fb45dc7cb480ba3c470d0d1e8f13f : SUCCESS in 4m 22s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/e56cffcc1c0b4171bbeb701103328086 : FAILURE in 25m 14s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"09a347a8ee092537ad00c764ed348f57a83323a3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-07-21 15:49:31.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"9a7593937b24a86847f714a0a850efc9bd8b2cde","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-07-21 16:39:07.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"22b0b9a9a2e9e1ffb999a4c09abb44e04e8e6c5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 17:05:59.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/08e6724942724b9c9ee0ea7b19115f08\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8cf0a8f6c5fa4bddb2e10e3a191b198d : FAILURE in 10m 47s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9e19d1415f5a4b66a436d6c9f9bcdb7a : FAILURE in 21m 28s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/1e11b5c13d94401db3cc80f149f162c3 : FAILURE in 20m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8793937814504659b44c3b9aa635400b : SUCCESS in 3m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ba7bbdb321142459f4419a79fae8c66 : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/92c48f374cc04bc8a0da60220f879960 : SUCCESS in 3m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4abf4fabfdf64ce1b88dfd65b5ddc8da : SUCCESS in 4m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9234ad9423594a0e96e670a531ba77aa : SUCCESS in 5m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4d40c29f7dfa4f5aa488e2e91468849c : SUCCESS in 5m 13s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b493d5cff8494867a6f2bc38c12d2af0 : FAILURE in 9m 11s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/048d50721bf14379ba5e08ca207eff75 : FAILURE in 19m 23s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c7c6b327d07041ab992ce2577cbc12e8 : FAILURE in 15m 02s","accounts_in_message":[],"_revision_number":5},{"id":"b7dce68b591c0b21ae56442aed3308fc557ddd7d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 17:49:22.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d10f72c947e14f8fbd783c0b897266a7\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c000f5f898c14afdb409c2b06645b248 : FAILURE in 1h 08m 00s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/06bc226844bc4fd0b3fa858492c28a44 : FAILURE in 31m 29s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9789d91c78d149cb96153ade809308a1 : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4d56db17896148baae8603eac165d925 : SUCCESS in 4m 49s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/baf9ffd1715c400c8dfac3306ea0774f : FAILURE in 25m 46s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"6ec15a087dd9b030f22d06db9e12f6e60ff37a7f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-07-26 07:57:28.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"cf62887cf84793bf3c6c82ce7af5b9ddaac203f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 09:22:34.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d984d85417f64950a6a1bad4b0a2b7ed\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d741bbc17864456c971b3cf354599230 : SUCCESS in 20m 04s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9d4cf61c33124149b382e808bbed5360 : SUCCESS in 53m 23s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/99a66d498f4540ef9868317760778bd4 : SUCCESS in 1h 18m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf1906a3d1714cc7b72efb68f9f27284 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b2245bfe6f3d44e88d5a93db175ba745 : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2723334a762f428d9ae659bc5db6286a : SUCCESS in 4m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/463404e927e8417cb3e05a8745c12821 : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5e494d9cbb043b0bb9823440fc5bad6 : SUCCESS in 5m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/66ba10e3cfbb4cf38de80b5d5a7ce8bf : SUCCESS in 4m 03s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/957ca78f99094c5fab3daf94c2db7e73 : SUCCESS in 22m 36s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/95154165ebe24674a9227660e162063e : SUCCESS in 59m 52s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9499d941df44443697099254fd0adc71 : SUCCESS in 1h 12m 33s","accounts_in_message":[],"_revision_number":5},{"id":"f89d00667da062dc37fb05c3b34d0aec77f17e63","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 09:23:19.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"5f10c0af3a347755efd33c6b5b011f7cff405ec8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 09:23:20.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3cf2f8068fdc40ab919e097e07054e37\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/dcd5722fb3f645dda36ef721e293a034 : SUCCESS in 57m 37s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1b7590d6583e4b3c817a14ef361d4c6a : SUCCESS in 1h 21m 20s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e67bd1d106a947999902d65ba5704687 : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2b1857a845b0492e985747c4bde1c9f2 : SUCCESS in 3m 59s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/03618b1a84ed41a5a3d2417fa885f9d5 : FAILURE in 57m 41s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"96b6a0e3b133dc66d9402f988c0d58c4a2fccf21","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 12:41:54.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f61cc7c5174247a0ad856ee98f27c44b\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5010a418624f4202b39a2e543cf2aeef : SUCCESS in 22m 37s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1b4772229e98453dbca627b1932f4a55 : FAILURE in 30m 02s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/9b75e9bc2e8b4917a3fc95d90d6636f6 : SUCCESS in 59m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29481992380e4d0d9552b109eeff712d : SUCCESS in 2m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b7da7347dd042c1b30589e3fd3fc586 : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2c8a5639b3bb4b9387807e37f8141817 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c183cbffdec34516ba69215337d2aa1e : SUCCESS in 5m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/98f95badd69449a08fe1d1cc801fa8a8 : SUCCESS in 3m 12s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/70ae7090f6a349739eed16f1797a5b75 : SUCCESS in 17m 15s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bbfde3fdcd334682897086012bcdd3f4 : SUCCESS in 54m 28s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a13208ad479f4b65969fff60048f7da4 : SUCCESS in 1h 08m 08s","accounts_in_message":[],"_revision_number":5},{"id":"733fc15a1d55d53895a396381ffee862d6fe0673","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-07-26 13:09:09.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d4debd4c47ed31d23a8188601e63e30f9221d2c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 14:20:12.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56c24c37403543d7876857c992298cf1\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/73d2367c3f6d450b87029e7e65574348 : SUCCESS in 22m 20s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8a6e6bf7466e4c798565205b7035bb79 : SUCCESS in 58m 33s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/bd14f198c39c443984bdd4213601453e : SUCCESS in 1h 08m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db608e6ad3b148db861888c49ac3ef0b : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4771388c7130405aa6c1119f30c2c974 : SUCCESS in 6m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c11fe8a321c944809505f56d874784e1 : SUCCESS in 6m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f35c1a76ffba40e2813f34f37e113cc1 : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8415729f6e24935853db039126aa48e : SUCCESS in 6m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/13b9f6e44f514bc288db641e507638e5 : SUCCESS in 6m 07s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c716ecda81da4443b2aa0bbd830fad9c : SUCCESS in 23m 02s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/780b47bb2b0046a398fa95099f98b1c0 : SUCCESS in 56m 03s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/462e3bce83d2421090ed02ed9c7165a3 : SUCCESS in 1h 03m 04s","accounts_in_message":[],"_revision_number":5},{"id":"8489d447e1c12dbe7ac2da04ea57f0fe15add461","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 14:20:39.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"a375367ec8fe5aa2664d1dfce5779ced83bdb0c8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 16:42:22.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3fb0e6faec94d489481874c89d51255\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/03fcfaa9dac447daaf31ddd0e423a4fe : SUCCESS in 1h 59m 08s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/07dd8a90d5dd457cb1b1b8ee82f07689 : SUCCESS in 2h 24m 44s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0297034a7555480ba3bba14d70ddc450 : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e23b60326d0a4af1ad3d42a9d6381c99 : SUCCESS in 7m 03s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/68edd4bc9c9e46748eb37dde359e0b70 : FAILURE in 2h 01m 23s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"0d9b9ba822415c62f7a9eba9154add6a027a3c16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 17:25:49.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8bd6d3ce4e54fa389f2f3aaf694de35\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8d5dd80c754742b48bc8492548f02e6c : SUCCESS in 19m 06s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8ad9c144089f44749c706c5363221167 : SUCCESS in 55m 06s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/a1b32c3cb39240a9ae3be5a83dc4109e : SUCCESS in 1h 04m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49ccb9fd66a544bbb3f194489b8d902c : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8504795458524494a63fb2812806c7f1 : SUCCESS in 3m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e4bb85a00c3425a9c6c035eae1cffe4 : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2809b8e293649d7a73a931fa676311b : SUCCESS in 6m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6958de3ce6494541a26ce999c93f5c88 : SUCCESS in 4m 04s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c35933880a914f6f989a449724414883 : SUCCESS in 35m 27s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4be893521d9e486993e44fef2907814e : SUCCESS in 56m 55s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c4ac999be20d4c9a83e8b1c7a376799d : SUCCESS in 1h 08m 32s","accounts_in_message":[],"_revision_number":5},{"id":"fcdb21d71c8e0f3b674081d0bd38bd31e696e93c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 17:25:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"40f113708d71daff1dd99b85d1cb7c6b28865f15","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 17:26:52.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/563bb889f278405893cc4c0f2468b78c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af7d240fc3a641f2a9ed34b970fe4ba0 : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec8838316f644dfd963ef781baddb394 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"91d22dd40fab3b410709bb7cabcf0b0f830214bc","revisions":{"3d60ba91038a4d63d7d69323eaa48e1eb52100ca":{"kind":"REWORK","_number":1,"created":"2022-06-23 15:54:44.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/68/847368/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/68/847368/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/847368/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/847368/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/847368/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/68/847368/1"}}},"commit":{"parents":[{"commit":"9931f839136c28cd8cdd7a0e649b0efe1eb561ea","subject":"disable monasca for AArch64 on all distros","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9931f839136c28cd8cdd7a0e649b0efe1eb561ea"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2022-06-23 15:43:26.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2022-06-23 15:51:51.000000000","tz":60},"subject":"Install etcd binary from github","message":"Install etcd binary from github\n\nRemoves a dependency on RDO for install etcd in CentOS images.\n\nAligns version of etcd across all distributions, see below:\n\n```\n(venv-kolla) [will@juno kolla]$ docker run --rm -it kolla/ubuntu-source-etcd:yoga bash\n()[etcd@730a011b670a /]$ etcd --version\netcd Version: 3.2.26\nGit SHA: Not provided (use ./build instead of go build)\nGo Version: go1.13.7\nGo OS/Arch: linux/amd64\n```\n\n```\n(venv-kolla) [will@juno kolla]$ sudo docker run -it kolla/centos-source-etcd:yoga bash\n()[etcd@9aa486c9e94f /]$ etcd --version\netcd Version: 3.2.21\nGit SHA: 3ac81f3\nGo Version: go1.12.8\nGo OS/Arch: linux/amd64\n```\n\n```\n(venv-kolla) [will@juno kolla]$ docker run --rm -it kolla/debian-source-etcd:yoga bash\n()[etcd@b49cb5fcdf87 /]$ etcd --version\netcd Version: 3.3.25\nGit SHA: Not provided (use ./build instead of go build)\nGo Version: go1.15.9\nGo OS/Arch: linux/amd64\n```\n\nAllows us to run a more up to data version.\n\nChange-Id: I62a34256fb7395d0448af169ac3cf036f60cd290\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3d60ba91038a4d63d7d69323eaa48e1eb52100ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3d60ba91038a4d63d7d69323eaa48e1eb52100ca"}]},"branch":"refs/heads/master"},"885cbf771a1f33031aff05435f50a92dbc235875":{"kind":"REWORK","_number":2,"created":"2022-07-20 10:20:24.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/68/847368/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/68/847368/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/847368/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/847368/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/847368/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/68/847368/2"}}},"commit":{"parents":[{"commit":"9931f839136c28cd8cdd7a0e649b0efe1eb561ea","subject":"disable monasca for AArch64 on all distros","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9931f839136c28cd8cdd7a0e649b0efe1eb561ea"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2022-06-23 15:43:26.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2022-07-20 10:06:58.000000000","tz":60},"subject":"Install etcd binary from github","message":"Install etcd binary from github\n\nRemoves a dependency on RDO for install etcd in CentOS images.\n\nAligns version of etcd across all distributions, see below:\n\n```\n(venv-kolla) [will@juno kolla]$ docker run --rm -it kolla/ubuntu-source-etcd:yoga bash\n()[etcd@730a011b670a /]$ etcd --version\netcd Version: 3.2.26\nGit SHA: Not provided (use ./build instead of go build)\nGo Version: go1.13.7\nGo OS/Arch: linux/amd64\n```\n\n```\n(venv-kolla) [will@juno kolla]$ sudo docker run -it kolla/centos-source-etcd:yoga bash\n()[etcd@9aa486c9e94f /]$ etcd --version\netcd Version: 3.2.21\nGit SHA: 3ac81f3\nGo Version: go1.12.8\nGo OS/Arch: linux/amd64\n```\n\n```\n(venv-kolla) [will@juno kolla]$ docker run --rm -it kolla/debian-source-etcd:yoga bash\n()[etcd@b49cb5fcdf87 /]$ etcd --version\netcd Version: 3.3.25\nGit SHA: Not provided (use ./build instead of go build)\nGo Version: go1.15.9\nGo OS/Arch: linux/amd64\n```\n\nThis will also allow us to update to a more recent version in the\nfuture. We have to be careful not to jump by more than one point release\nto ensure compatability:\n\nhttps://etcd.io/docs/v3.5/upgrades/upgrading-etcd/#upgrading-an-etcd-v3x-cluster\n\nChange-Id: I62a34256fb7395d0448af169ac3cf036f60cd290\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/885cbf771a1f33031aff05435f50a92dbc235875"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/885cbf771a1f33031aff05435f50a92dbc235875"}]},"branch":"refs/heads/master"},"b4af63434761bc8cccc738abd007e6aafcbff98e":{"kind":"REWORK","_number":3,"created":"2022-07-20 10:22:36.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/68/847368/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/68/847368/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/847368/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/847368/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/847368/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/68/847368/3"}}},"commit":{"parents":[{"commit":"9931f839136c28cd8cdd7a0e649b0efe1eb561ea","subject":"disable monasca for AArch64 on all distros","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9931f839136c28cd8cdd7a0e649b0efe1eb561ea"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2022-06-23 15:43:26.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2022-07-20 10:22:17.000000000","tz":60},"subject":"Install etcd binary from github","message":"Install etcd binary from github\n\nRemoves a dependency on RDO for install etcd in CentOS images.\n\nAligns version of etcd across all distributions, see below:\n\n```\n(venv-kolla) [will@juno kolla]$ docker run --rm -it kolla/ubuntu-source-etcd:yoga bash\n()[etcd@730a011b670a /]$ etcd --version\netcd Version: 3.2.26\nGit SHA: Not provided (use ./build instead of go build)\nGo Version: go1.13.7\nGo OS/Arch: linux/amd64\n```\n\n```\n(venv-kolla) [will@juno kolla]$ sudo docker run -it kolla/centos-source-etcd:yoga bash\n()[etcd@9aa486c9e94f /]$ etcd --version\netcd Version: 3.2.21\nGit SHA: 3ac81f3\nGo Version: go1.12.8\nGo OS/Arch: linux/amd64\n```\n\n```\n(venv-kolla) [will@juno kolla]$ docker run --rm -it kolla/debian-source-etcd:yoga bash\n()[etcd@b49cb5fcdf87 /]$ etcd --version\netcd Version: 3.3.25\nGit SHA: Not provided (use ./build instead of go build)\nGo Version: go1.15.9\nGo OS/Arch: linux/amd64\n```\n\nThis will also allow us to update to a more recent version in the\nfuture. We have to be careful not to jump by more than one point release\nto ensure compatability:\n\nhttps://etcd.io/docs/v3.5/upgrades/upgrading-etcd/#upgrading-an-etcd-v3x-cluster\n\nChange-Id: I62a34256fb7395d0448af169ac3cf036f60cd290\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b4af63434761bc8cccc738abd007e6aafcbff98e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b4af63434761bc8cccc738abd007e6aafcbff98e"}]},"branch":"refs/heads/master"},"818fbb6f480b6c1ebcc3639a947874bcf81bf73d":{"kind":"REWORK","_number":4,"created":"2022-07-20 11:20:12.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/68/847368/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/68/847368/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/847368/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/847368/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/847368/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/68/847368/4"}}},"commit":{"parents":[{"commit":"9931f839136c28cd8cdd7a0e649b0efe1eb561ea","subject":"disable monasca for AArch64 on all distros","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9931f839136c28cd8cdd7a0e649b0efe1eb561ea"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2022-06-23 15:43:26.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2022-07-20 11:18:49.000000000","tz":60},"subject":"Install etcd binary from github","message":"Install etcd binary from github\n\nRemoves a dependency on RDO for install etcd in CentOS images.\n\nAligns version of etcd across all distributions, see below:\n\n```\n(venv-kolla) [will@juno kolla]$ docker run --rm -it kolla/ubuntu-source-etcd:yoga bash\n()[etcd@730a011b670a /]$ etcd --version\netcd Version: 3.2.26\nGit SHA: Not provided (use ./build instead of go build)\nGo Version: go1.13.7\nGo OS/Arch: linux/amd64\n```\n\n```\n(venv-kolla) [will@juno kolla]$ sudo docker run -it kolla/centos-source-etcd:yoga bash\n()[etcd@9aa486c9e94f /]$ etcd --version\netcd Version: 3.2.21\nGit SHA: 3ac81f3\nGo Version: go1.12.8\nGo OS/Arch: linux/amd64\n```\n\n```\n(venv-kolla) [will@juno kolla]$ docker run --rm -it kolla/debian-source-etcd:yoga bash\n()[etcd@b49cb5fcdf87 /]$ etcd --version\netcd Version: 3.3.25\nGit SHA: Not provided (use ./build instead of go build)\nGo Version: go1.15.9\nGo OS/Arch: linux/amd64\n```\n\nThis will also allow us to update to a more recent version in the\nfuture. We have to be careful not to jump by more than one point release\nto ensure compatability:\n\nhttps://etcd.io/docs/v3.5/upgrades/upgrading-etcd/#upgrading-an-etcd-v3x-cluster\n\nChange-Id: I62a34256fb7395d0448af169ac3cf036f60cd290\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/818fbb6f480b6c1ebcc3639a947874bcf81bf73d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/818fbb6f480b6c1ebcc3639a947874bcf81bf73d"}]},"branch":"refs/heads/master"},"91d22dd40fab3b410709bb7cabcf0b0f830214bc":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2022-07-20 11:21:24.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/68/847368/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/68/847368/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/847368/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/847368/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/847368/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/68/847368/5"}}},"commit":{"parents":[{"commit":"9931f839136c28cd8cdd7a0e649b0efe1eb561ea","subject":"disable monasca for AArch64 on all distros","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9931f839136c28cd8cdd7a0e649b0efe1eb561ea"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2022-06-23 15:43:26.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2022-07-20 11:20:49.000000000","tz":60},"subject":"Install etcd binary from github","message":"Install etcd binary from github\n\nRemoves a dependency on RDO for install etcd in CentOS images.\n\nAligns version of etcd across all distributions, see below:\n\n```\n(venv-kolla) [will@juno kolla]$ docker run --rm -it kolla/ubuntu-source-etcd:yoga bash\n()[etcd@730a011b670a /]$ etcd --version\netcd Version: 3.2.26\nGit SHA: Not provided (use ./build instead of go build)\nGo Version: go1.13.7\nGo OS/Arch: linux/amd64\n```\n\n```\n(venv-kolla) [will@juno kolla]$ sudo docker run -it kolla/centos-source-etcd:yoga bash\n()[etcd@9aa486c9e94f /]$ etcd --version\netcd Version: 3.2.21\nGit SHA: 3ac81f3\nGo Version: go1.12.8\nGo OS/Arch: linux/amd64\n```\n\n```\n(venv-kolla) [will@juno kolla]$ docker run --rm -it kolla/debian-source-etcd:yoga bash\n()[etcd@b49cb5fcdf87 /]$ etcd --version\netcd Version: 3.3.25\nGit SHA: Not provided (use ./build instead of go build)\nGo Version: go1.15.9\nGo OS/Arch: linux/amd64\n```\n\nThis will also allow us to update to a more recent version in the\nfuture. We have to be careful not to jump by more than one point release\nto ensure compatibility:\n\nhttps://etcd.io/docs/v3.5/upgrades/upgrading-etcd/#upgrading-an-etcd-v3x-cluster\n\nChange-Id: I62a34256fb7395d0448af169ac3cf036f60cd290\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/91d22dd40fab3b410709bb7cabcf0b0f830214bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/91d22dd40fab3b410709bb7cabcf0b0f830214bc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
