)]}'
{"id":"openstack%2Fkolla~847773","triplet_id":"openstack%2Fkolla~stable%2Fyoga~I4bd581c3ea422e1c9a075a71721413d47fdbdfe4","project":"openstack/kolla","branch":"stable/yoga","topic":"openstack-exporter-1.6.0-stable/yoga","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-07-01 14:06:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4bd581c3ea422e1c9a075a71721413d47fdbdfe4","subject":"Bump prometheus-openstack-exporter version to 1.6.0","status":"MERGED","created":"2022-06-27 11:54:54.000000000","updated":"2022-07-01 14:07:51.000000000","submitted":"2022-07-01 14:06:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"847773-openstack-exporter-1.6.0-stable/yoga","meta_rev_id":"a19f21ca9cba99d777dfe3ea6bc1f797d7fbdeff","_number":847773,"virtual_id_number":847773,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-01 14:06:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-07-01 08:53:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-07-01 12:51:09.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2022-07-01 12:51:09.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-27 12:13:54.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-06-27 13:06:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-27 13:51:27.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"2b8e2dd9cb4863a673cc32607932fbcf00b04520","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-06-27 11:54:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"55eb1116a3f0eb317ada0097a6ee66f9801f8533","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-06-27 12:13:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"334267ff7bdf6a4b98eb8e378d4e3734e08b2751","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-27 13:06:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/df7ea1bcd4de434eac8aaab5b37d27c3 : SUCCESS in 28m 24s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/fcd3b0574f7346c5a9091bca88cb2ba9 : FAILURE in 26m 56s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/df647b7344d94e17bbb348558af8ec45 : FAILURE in 27m 18s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/42fa101e5f0342538d7e5b1cf2faf363 : SUCCESS in 1h 09m 54s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/073875f6be4f4f669758bbbb063dbf7c : FAILURE in 21m 08s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/dc0589d0d92b44db820bb9195b04eb51 : FAILURE in 23m 41s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/10f0dec1bea948f29af416b5e75b2ff1 : SUCCESS in 18m 14s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a54b39362332416ca6848451a9857c0e : FAILURE in 31m 20s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/f7bcf65cd5104a19b78c543da3e88168 : FAILURE in 22m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ccc857c9ead4db8b64308ce8be2717f : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/18a77c886d8b4881b9592a98048f1a29 : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/494d645f440e453ab4eaae9baff7a880 : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fc2fae366ba478c8d5315ae7f63d18b : SUCCESS in 5m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b1a8e8df608944fe93de37187390bc8a : SUCCESS in 5m 02s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b68da1a6c33440cbb0ac2e83e495c1ec : SUCCESS in 25m 05s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/57de6f5f41b44e21a13e21bb00e69b90 : FAILURE in 30m 22s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7e0176e2993e4874a58073b997d29229 : FAILURE in 14m 15s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a65ee4062628458599f24819a30d5a01 : SUCCESS in 45m 05s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c621cb06efa542c08ab323fd1dafc584 : FAILURE in 21m 35s","accounts_in_message":[],"_revision_number":1},{"id":"43411201cb1c7fd8ebccafad640d4fc4b060d26d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-27 13:47:57.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/2ce781ac4d814326b5fa2a8f543bd455 : FAILURE in 48m 06s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/dc6fc6051f184c6091d7851f4194679a : FAILURE in 1h 50m 52s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a701d906b99c44f0919bf8cd1978dbd6 : FAILURE in 58m 43s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/345732e1053d40f7b6a6aa2b73ddd7d7 : SUCCESS in 8m 41s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/814cc26361dc48e68b9f3557b47345ed : SUCCESS in 5m 48s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/40e884f273274249ae54788cf755531c : FAILURE in 1h 44m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b67094cf9a2fe9732cc86d23eec9dbc212295b51","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-06-27 13:51:27.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4862d88b7a3d57aa72860045825291b57fba2496","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-07-01 08:53:07.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"680b698ce8388bbed7723839ae1c3f7cbcaf23ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-01 10:31:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/34aef3d1596a41829e1ac4ebe37d6436 : SUCCESS in 29m 09s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/ccb0c2ea75a54b46b8d268e2b3acff3f : SUCCESS in 36m 00s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/a996d9e749c8495aace95c73223df5eb : SUCCESS in 1h 19m 53s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/b062c2f655a746c69093d25a60e09140 : SUCCESS in 1h 16m 08s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/4eaf9f9127a14cf0ac11bcaaf1232333 : SUCCESS in 1h 33m 29s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5213fe1e067548f5afebb08cebeb504e : SUCCESS in 20m 49s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/574bd9b8d14c4decad615a011867524b : FAILURE in 15m 45s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7bcf43b69b934950a1b9fcb6a91a61ed : SUCCESS in 54m 45s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/990588c3a7a54457ae452fa7d6d5c3cd : SUCCESS in 1h 20m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e561787e18a14da1b2a07e42bcc738ed : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d737b54ed8444541a723b5c9e2328492 : SUCCESS in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dda9463fca574d84a61e2a46cc91000d : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c23404ba8ec45a6901ba0443d344d1a : SUCCESS in 6m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9965a85918844f0e8f34d34d9006ac89 : SUCCESS in 3m 34s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/32cfcd957593431990342565c2e06514 : SUCCESS in 17m 52s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1328e824616d44b7977755a85509deeb : SUCCESS in 22m 32s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/19e0184aa1924062b84aa3d4889bb0f7 : SUCCESS in 46m 44s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/61922c92c2d4453da5a9f3daa93c13ea : SUCCESS in 53m 05s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f4aae0bff3fa429388aab8781ea23bc3 : SUCCESS in 1h 01m 58s","accounts_in_message":[],"_revision_number":2},{"id":"d9ca9f421e23a01178e407882d1ea0bf5e38c36d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-01 11:22:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/28f07ebbe255496cb11ed0bae3cb1556 : SUCCESS in 1h 49m 28s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b500c0701d0c48ffb08a8514472f68ab : SUCCESS in 38m 30s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/39f76c1e27384ff39098e5cee81642a8 : SUCCESS in 2h 26m 50s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/543d8a6d3f7c42f0a0756061c1a41d80 : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ab6f6eae320a46a4937759565bde1506 : SUCCESS in 5m 42s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/da48d45da3984c8181f73e398470b7e4 : FAILURE in 34m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4f9accaaac8cbd256d674d2b016e98abcf206b9b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-07-01 12:51:09.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7662d44ad7d9f3db6d8df7d8e43634c1da1399b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-01 12:51:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6e2e5c9ed7645be49b61b055f0833609ea0a6d6c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-01 14:06:31.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/acfd36b983d246d7b265f4325e22740f : SUCCESS in 29m 07s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/ce5ea9317a67428199f56f8638c2cb04 : SUCCESS in 1h 13m 19s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/20f6f145cbeb4a85b28c484bc2a1fc11 : SUCCESS in 1h 12m 46s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/63eb1db614954a92a958e544b66bae25 : SUCCESS in 27m 14s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/fe9f9c3371c143969b9f7c355d300219 : SUCCESS in 55m 26s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/70d6fe089c2141b08ca02f3008426ef4 : SUCCESS in 1h 04m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/962838ed5d144db7835faf366ffd7ce6 : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ace5b9262fbc47af9ae1ff0a3d5505eb : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2018dc01bb2742cd95b95d6bcb6915ee : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22c06fc4694a42dbad8819ba3feddb8a : SUCCESS in 6m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eeabb0abccf446eb8ca29327d60da976 : SUCCESS in 3m 54s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0791f01744324c149f2235566c9b68eb : SUCCESS in 32m 44s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/71533d84fbc44c4589fd0073ef89d580 : SUCCESS in 44m 26s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5728c9187f8c46f28e07ca24c491cd40 : SUCCESS in 1h 05m 14s","accounts_in_message":[],"_revision_number":2},{"id":"d0418e522a0a199053000fdc359246886c4bf596","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-01 14:06:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"a19f21ca9cba99d777dfe3ea6bc1f797d7fbdeff","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-01 14:07:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59510e486f0648859c28c044d7e1b8aa : SUCCESS in 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ede8450e10e450b92f5dd622da231ee : SUCCESS in 53s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8253acdea048cd52470b0ad2f09a0c87185af193","revisions":{"d4f0e217bddf48ff3045c0bad12af68d0c095f91":{"kind":"REWORK","_number":1,"created":"2022-06-27 11:54:54.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/73/847773/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/73/847773/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/847773/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/847773/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/847773/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/73/847773/1"}}},"commit":{"parents":[{"commit":"66eec33571fdcc6e6d95a284e11925d626329bf1","subject":"Bump skydive version to 0.28","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/66eec33571fdcc6e6d95a284e11925d626329bf1"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-06-27 11:54:28.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-06-27 11:54:28.000000000","tz":120},"subject":"Bump prometheus-openstack-exporter version to 1.6.0","message":"Bump prometheus-openstack-exporter version to 1.6.0\n\nChange-Id: I4bd581c3ea422e1c9a075a71721413d47fdbdfe4\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/847538\n(cherry picked from commit 5f94e53bea9fb81786595cb8b452879afe436e26)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d4f0e217bddf48ff3045c0bad12af68d0c095f91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d4f0e217bddf48ff3045c0bad12af68d0c095f91"}]},"branch":"refs/heads/stable/yoga"},"8253acdea048cd52470b0ad2f09a0c87185af193":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-07-01 08:53:07.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/73/847773/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/73/847773/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/847773/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/847773/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/847773/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/73/847773/2"}}},"commit":{"parents":[{"commit":"47c64433a09460cd34d26a8a752a81428e0c4ed9","subject":"upper-constraints: remove neutron","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/47c64433a09460cd34d26a8a752a81428e0c4ed9"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-06-27 11:54:28.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-07-01 08:53:07.000000000","tz":0},"subject":"Bump prometheus-openstack-exporter version to 1.6.0","message":"Bump prometheus-openstack-exporter version to 1.6.0\n\nChange-Id: I4bd581c3ea422e1c9a075a71721413d47fdbdfe4\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/847538\n(cherry picked from commit 5f94e53bea9fb81786595cb8b452879afe436e26)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8253acdea048cd52470b0ad2f09a0c87185af193"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8253acdea048cd52470b0ad2f09a0c87185af193"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
