)]}'
{"id":"openstack%2Fkolla~848459","triplet_id":"openstack%2Fkolla~stable%2Fwallaby~I4bd581c3ea422e1c9a075a71721413d47fdbdfe4","project":"openstack/kolla","branch":"stable/wallaby","topic":"openstack-exporter-1.6.0-stable/wallaby","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-07-04 10:05:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4bd581c3ea422e1c9a075a71721413d47fdbdfe4","subject":"Bump prometheus-openstack-exporter version to 1.6.0","status":"MERGED","created":"2022-07-01 14:47:48.000000000","updated":"2022-07-04 10:06:51.000000000","submitted":"2022-07-04 10:05:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"848459-openstack-exporter-1.6.0-stable/wallaby","meta_rev_id":"6c72a7eefad8f4de81e5f397aef3d8257c55ddb0","_number":848459,"virtual_id_number":848459,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-04 10:05:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-07-04 08:19:36.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2022-07-01 14:58:43.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2022-07-04 08:19:36.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-01 14:58:43.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-07-01 16:13:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-04 08:19:36.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"2c6cc9e921a2e172773628445270b0f6937c3c49","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-07-01 14:47:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a066f8bf3f3172e3efccf0334696d1d4da3e6274","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-07-01 14:58:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"944782ab612446e80f724620f1cb6c283f381e24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-01 16:13:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/e4bee278d286475082efdeb061a33989 : SUCCESS in 35m 49s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/3cc7dae875fa4740abb8c9dc4480ec08 : SUCCESS in 36m 50s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/66e448c0a309470090c600b8186be99a : SUCCESS in 1h 10m 01s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/96497c220b7c4d0085098ec938edfb43 : SUCCESS in 1h 07m 54s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/65927f22612d42ceb5d42b04a3b5759e : SUCCESS in 1h 23m 56s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f79431c24f514f7a8def4135501ad64f : SUCCESS in 52m 25s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/984a6d5121f94f90b8be62b3dfd2d898 : SUCCESS in 21m 40s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c6975c947b19415ebd5bf5e4dcfcadb0 : SUCCESS in 1h 08m 24s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f79dbfb2cf15410bb91481a363f76730 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4dac2232d1064467b18ee0b5e1ec70a9 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b58f4d2c71d4483591d830c1c2bf0395 : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e6b94c7e60b448108842cc9cfc50521a : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cff7e4aab98d4491b85d969aa649483d : SUCCESS in 6m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/80c621358d164765844f9b81f294ea76 : SUCCESS in 4m 20s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b642555d317c4af5a2bd65b4a9eb3f0f : SUCCESS in 17m 27s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d6bad745a49e42b7b36456b05b28053a : SUCCESS in 53m 13s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/60ba023200f841009d70858fa6b98883 : SUCCESS in 41m 15s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/85e9ef9f8c6145e19da37f023f4781a4 : SUCCESS in 40m 47s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/19c8027875c34b9faf2c15efdc9111ac : SUCCESS in 51m 20s","accounts_in_message":[],"_revision_number":1},{"id":"bf87222dbec094ff2183057d0be2ab6bba1dc4d3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-01 17:00:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/0ef51331faab45be8ecb14298ca26d58 : SUCCESS in 2h 02m 12s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5a71ef3b241e4da49de6d8eb7be9b759 : SUCCESS in 2h 11m 53s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/45af84e96f634aeb8e7d4126efd95920 : SUCCESS in 1h 58m 36s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/c24981da1f5e4bc5866fadd65c7e4e5e : FAILURE in 1h 26m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0232039425913b6566f66baa1d77e4d0b6d52962","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-07-04 08:19:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"96fffbd8db17aba5244187a95fad2992561db661","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 08:21:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4edf0dd06e1ac5012cd1499a429ef641e8bfee7d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 10:05:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/afb5b0717d124c41b941e35dc9e6c947\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/66958bc9b9dc4aa4b0c1064e97904815 : SUCCESS in 36m 45s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/264245273b994f86aad30f87c92cecc2 : SUCCESS in 1h 19m 15s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/d4ed82c1051d4fdd8b4b2e96773471df : SUCCESS in 1h 39m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44744f3e418d4dfb9ae621fa9db3c15a : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aaa7d72a078b40809177066335503080 : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/467309389c0845f791265695a4aac4b5 : SUCCESS in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77b4bcf0298c42bd88514968124c9c64 : SUCCESS in 5m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9cf43018f5104b50ade69d5cdaaaea1d : SUCCESS in 4m 05s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/542bdd23aff64d2d9aa1d882bf88ba60 : SUCCESS in 38m 45s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4c171b2bb8944a8f82e3494f8765f306 : SUCCESS in 49m 06s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7068d766e42240b4a62867dd482c66d6 : SUCCESS in 51m 17s","accounts_in_message":[],"_revision_number":1},{"id":"4ff6e128ef8ae9698418581c2b39306affc8f85f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 10:05:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6c72a7eefad8f4de81e5f397aef3d8257c55ddb0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 10:06:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6216e88308dc4907842f9fac01a4e52c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dd21c076ffc433eb4943a0f07252769 : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/942c13ce4b67457aab72751656d44a4a : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0541e933286b2ec3e0b4342870cec1c945da59b1","revisions":{"0541e933286b2ec3e0b4342870cec1c945da59b1":{"kind":"REWORK","_number":1,"created":"2022-07-01 14:47:48.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/59/848459/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/848459/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/848459/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/848459/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/848459/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/848459/1"}}},"commit":{"parents":[{"commit":"836e5945a37e0053f853bd75184b3e344866afae","subject":"Don\u0027t ignore errors of \u0027apt-get update\u0027 command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/836e5945a37e0053f853bd75184b3e344866afae"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-06-22 14:57:32.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-07-01 14:47:25.000000000","tz":120},"subject":"Bump prometheus-openstack-exporter version to 1.6.0","message":"Bump prometheus-openstack-exporter version to 1.6.0\n\nChange-Id: I4bd581c3ea422e1c9a075a71721413d47fdbdfe4\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/848457\n(cherry picked from commit 5f94e53bea9fb81786595cb8b452879afe436e26)\n(cherry picked from commit 51d5e442c105bd5dd74c983ef277812b8915eb52)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0541e933286b2ec3e0b4342870cec1c945da59b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0541e933286b2ec3e0b4342870cec1c945da59b1"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
