)]}'
{"id":"openstack%2Fkolla~849816","triplet_id":"openstack%2Fkolla~stable%2Fvictoria~I78da7eb2b6b5eae2b04fd95b24615bd4a572a1d6","project":"openstack/kolla","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"12640":{"account":{"_account_id":12640,"name":"Arun S A G","email":"sagarun@gmail.com","username":"sagarun"},"last_update":"2022-07-18 09:12:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I78da7eb2b6b5eae2b04fd95b24615bd4a572a1d6","subject":"neutron: Use update-alternatives --display instead of --query","status":"MERGED","created":"2022-07-14 17:37:16.000000000","updated":"2022-07-18 09:13:42.000000000","submitted":"2022-07-18 09:12:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849816","cherry_pick_of_change":815781,"cherry_pick_of_patch_set":2,"meta_rev_id":"e3b6ea1afe78b13f5ce3af33e599685a2c1031ba","_number":849816,"virtual_id_number":849816,"owner":{"_account_id":12640,"name":"Arun S A G","email":"sagarun@gmail.com","username":"sagarun"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-18 09:12:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2022-07-18 07:57:55.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2022-07-15 08:13:01.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2022-07-18 07:57:55.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-14 17:37:16.000000000","updated_by":{"_account_id":12640,"name":"Arun S A G","email":"sagarun@gmail.com","username":"sagarun"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2022-07-14 19:47:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-14 20:43:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-15 08:13:01.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-07-18 07:57:55.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"}],"messages":[{"id":"3cae8d207d8f6d7d2fc4b405ca6b8d7ae9affdcc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12640,"name":"Arun S A G","email":"sagarun@gmail.com","username":"sagarun"},"date":"2022-07-14 17:37:16.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"3bd5a597156f62872d0575bb9bb980fa8f2f3d3b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 19:47:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4194d5c984214e0f80a99e28572716c7\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/9359c0fe3c5043e59f900304f4e11050 : SUCCESS in 1h 07m 56s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/71a733fe2ab34549a05d48d02d31bd1c : FAILURE in 2h 08m 17s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8ac3417d30584cac8aa8cf4c9104cd01 : FAILURE in 1h 10m 55s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/f603431aeae94b6d9e095cdb805740fb : FAILURE in 49m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d19d95f3644b6aa513ee49b2615fa939f1e74b08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 20:43:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b578c12fecd4b18903ea4106439e306\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/1e1e6c387b2148b39d1ad33c4d92f256 : TIMED_OUT in 3h 01m 25s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/19da8524d234498b8a0ee75c1fcbcc63 : SUCCESS in 44m 59s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/2339d1b194da47e08aefbebf6b0033a7 : SUCCESS in 1h 07m 09s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/fb1a85a1791442cd92dc25fc754d69c4 : SUCCESS in 1h 11m 12s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/1b9508cad3c14edc9115b80f8c770a53 : SUCCESS in 1h 14m 00s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/fd07246306154d03839c00fda7108c24 : SUCCESS in 28m 07s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a37cdefedf824b278a501fa21f59e38b : SUCCESS in 20m 05s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b40728d4e92140699351134b7d5ccdca : SUCCESS in 59m 18s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/980b2159dc614a6f85f02fcc881e5a3f : SUCCESS in 7m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6ce2c0605f424eb48f009ba14b5561ca : SUCCESS in 6m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c2ab57a98b6f4a37903090e630433245 : SUCCESS in 6m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77fa95e87b964d75bf88d1ded484456d : SUCCESS in 9m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a969c5ff4bad4e8f8f777bf20f9952b5 : SUCCESS in 6m 57s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c984233899ee4d8f859824d95d7339c2 : SUCCESS in 37m 06s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5100d4e55e364d9794fd6b9ee03dd8db : SUCCESS in 41m 03s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ac765f680d7f4d5b899be38a9d229e57 : SUCCESS in 1h 00m 54s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6c8267393d44434e97209b59c5971537 : SUCCESS in 51m 28s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b59bbb195f9442b698ecf6b5f5ccedac : SUCCESS in 1h 02m 41s","accounts_in_message":[],"_revision_number":1},{"id":"b3abed40f60ff5d6e41d7260b85870273fecd97f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-07-15 08:13:01.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bf4b4be96f707a4bab108d4be715ff87b58f9cfc","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-07-18 07:57:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bd4e90de1d67f2bef2040839269dbfb999982c15","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 07:58:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b49442edaeb27e1579cc1a8a33c5804fe064f5f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 09:12:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5063e12d50854cb99ec9fd644128bb64\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/d422938548354b00935ca9255b0e7434 : SUCCESS in 32m 54s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/27fc2094bd6e4207a362f1f12ebc3b9c : SUCCESS in 58m 02s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/54734abc4cd34504a87b9755c7a00093 : SUCCESS in 59m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e47d84d8497a401a9cf6fb7036795b5a : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bcc1fd7065b941c79c0c7293f8eef054 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aae5c9003abd46b081454f8090b643c6 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be3067edc21745a89f6bf8264d335c4c : SUCCESS in 6m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f571b4939864568a91b0bd4085a55a7 : SUCCESS in 3m 52s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/334699e0c5fa47fbae75d9bf188b2814 : SUCCESS in 28m 02s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3abff61b85594b759b299790db69788f : SUCCESS in 45m 04s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f80ce261579f4db1b7ac0c614ea09a10 : SUCCESS in 54m 29s","accounts_in_message":[],"_revision_number":1},{"id":"cff9111e0d69ef524d4ffe3d2ef2b6ebc0ab4b37","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 09:12:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e3b6ea1afe78b13f5ce3af33e599685a2c1031ba","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 09:13:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b3d4cf154ea49a9b44c781854c9da1d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/871c9155099040d1a35ffea29af37168 : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/298545de61554cb1b2f7270f618b599c : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"62d7ef9e86f146c99f4fdea6d19827a6375b12e4","revisions":{"62d7ef9e86f146c99f4fdea6d19827a6375b12e4":{"kind":"REWORK","_number":1,"created":"2022-07-14 17:37:16.000000000","uploader":{"_account_id":12640,"name":"Arun S A G","email":"sagarun@gmail.com","username":"sagarun"},"ref":"refs/changes/16/849816/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/849816/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/849816/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/849816/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/849816/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/849816/1"}}},"commit":{"parents":[{"commit":"5f6a0ecab338fd0f07506e53a58d1ad3732ef3e8","subject":"Merge \"masakari: add Cyrus SASL packages to monitors image\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5f6a0ecab338fd0f07506e53a58d1ad3732ef3e8"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-10-28 07:17:17.000000000","tz":120},"committer":{"name":"Arun S A G","email":"sagarun@gmail.com","date":"2022-07-14 17:37:16.000000000","tz":0},"subject":"neutron: Use update-alternatives --display instead of --query","message":"neutron: Use update-alternatives --display instead of --query\n\n--query is not supported before update-alternatives 1.15\nand CentOS uses 1.13\n\nCloses-Bug: #1936947\nChange-Id: I78da7eb2b6b5eae2b04fd95b24615bd4a572a1d6\n(cherry picked from commit 2d1bbe72579f9209e151133352d4d6761f764f33)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/62d7ef9e86f146c99f4fdea6d19827a6375b12e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/62d7ef9e86f146c99f4fdea6d19827a6375b12e4"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
