)]}'
{"id":"openstack%2Fkolla~849817","triplet_id":"openstack%2Fkolla~stable%2Fwallaby~I78da7eb2b6b5eae2b04fd95b24615bd4a572a1d6","project":"openstack/kolla","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"12640":{"account":{"_account_id":12640,"name":"Arun S A G","email":"sagarun@gmail.com","username":"sagarun"},"last_update":"2022-07-18 09:22:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I78da7eb2b6b5eae2b04fd95b24615bd4a572a1d6","subject":"neutron: Use update-alternatives --display instead of --query","status":"MERGED","created":"2022-07-14 17:37:56.000000000","updated":"2022-07-18 09:23:38.000000000","submitted":"2022-07-18 09:22:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"849817","cherry_pick_of_change":815781,"cherry_pick_of_patch_set":2,"meta_rev_id":"f6efd7d6a7d14457501593cd21e0a56a92a2689b","_number":849817,"virtual_id_number":849817,"owner":{"_account_id":12640,"name":"Arun S A G","email":"sagarun@gmail.com","username":"sagarun"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-18 09:22:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2022-07-18 07:58:05.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2022-07-15 08:12:53.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2022-07-18 07:58:05.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-14 17:37:56.000000000","updated_by":{"_account_id":12640,"name":"Arun S A G","email":"sagarun@gmail.com","username":"sagarun"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2022-07-14 19:22:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-15 08:12:53.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-07-18 07:58:05.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"}],"messages":[{"id":"0199db6499d3a638d349dbfa4fad8c713359b103","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12640,"name":"Arun S A G","email":"sagarun@gmail.com","username":"sagarun"},"date":"2022-07-14 17:37:56.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"f89da0036ef5bf33588953970ef4e8ece2b1070f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 19:22:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9e7bef1fabc4412a977e7a32336fecb5\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/27a68121181545d8b7dd0c4f2b149439 : SUCCESS in 36m 12s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/67cbd8e7e7094325bc55931451592bb4 : SUCCESS in 37m 37s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/818e14f4562445bfbfa6252b0b8fd1d6 : SUCCESS in 1h 14m 04s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/3ddb4598c7b84419a09236526827749c : SUCCESS in 1h 14m 09s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/a80c2e1f3f5345fcb948af1479069c71 : SUCCESS in 1h 26m 19s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f090281a683543ab9092fb2ad1f35054 : SUCCESS in 57m 56s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0b516f162c9a42859ec86f008f980655 : SUCCESS in 21m 56s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/49da7f2735804168b1933f20e689e34c : SUCCESS in 1h 02m 02s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0800aff55154a2cb69c8b4defb3cbbf : SUCCESS in 7m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a3ef5ed4f134d649c2e1b219f2118a0 : SUCCESS in 7m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6532d49d313048938417855e8925cee8 : SUCCESS in 7m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a35b91845c0a4210a10da0f8bf650185 : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa3f7f10357d45b8ad567bef5aa14e97 : SUCCESS in 8m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5bb664d20257471581f54cd0d7d69a5e : SUCCESS in 6m 48s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6699b7f59f794197b6e0b3b418754404 : SUCCESS in 18m 55s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c8ac2cccb4a14b92aaee9e9d5586e3af : FAILURE in 1h 38m 50s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1fe126184f5f448480a2d0ed5739eecb : SUCCESS in 1h 10m 04s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9c54a095a3364f40be2777be24c1ce37 : SUCCESS in 59m 13s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e9d2140daa354d3d9d0f01b6b8b41091 : SUCCESS in 1h 02m 48s","accounts_in_message":[],"_revision_number":1},{"id":"74efb345fc7ac20736d69a9fa869a9c90363f90d","author":{"_account_id":12640,"name":"Arun S A G","email":"sagarun@gmail.com","username":"sagarun"},"date":"2022-07-14 19:31:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b2921f1cd03e63af851031430108d65217126ea0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 20:11:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/726448b64d6b4ecaae0d20da29040cb3\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/298db0db2efa4d01a24e0f0da792a364 : SUCCESS in 1h 08m 00s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/08bb48edd34642c6b5b06b9a53a1f854 : FAILURE in 2h 31m 08s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/af4f6075b8a94d0baf5113814c3e288c : SUCCESS in 1h 30m 11s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/c2c0c986a3be4a8f92ca30ade9595204 : FAILURE in 2h 04m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ca101d8f71a6bf0f82421aaf3a8a43b4e8c7d74e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 20:40:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c37d7261f8b4f05861f7b9d1f877de9\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/9bc572e347724cf589a1ee1a42b2afd8 : SUCCESS in 30m 17s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/13251f6f344e486ea9715da7c7156185 : SUCCESS in 29m 47s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/c069f4ce4e4f4aceb3d564dd58ad0583 : SUCCESS in 1h 00m 31s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/73246d0b8c3b44aa917ddff7808db422 : SUCCESS in 59m 50s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/b7fd12af17a0416c82c76b7d3c53f8be : SUCCESS in 1h 08m 11s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/542aaf34b7ce45c98d39a64a304d39e3 : SUCCESS in 38m 54s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/7512be7b75394ba395430ee0bf2d5049 : SUCCESS in 22m 10s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c12b0f0febe048a296c5140a183f6655 : SUCCESS in 43m 24s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4b051738f5a4a47b0e47d4d9252b157 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/95272361ba63453db7c3384de8e570cc : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e5a987a666ad4e6a9a426aba2819c8ef : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4f02a848c69e4ccba6b1abb5daefdf13 : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47fc46a0c97945fea3b301f7df9c47db : SUCCESS in 7m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7793dcbcfd55486cb0f2dba22adcf613 : SUCCESS in 3m 51s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/74167b95613243e19887119c25ed6f40 : SUCCESS in 21m 29s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/79c1ddec8a6c45189d2c6e2cb48fdaf4 : SUCCESS in 44m 11s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6c8e1f51f0ab4dfba32b7bf6f5af686c : SUCCESS in 57m 51s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4eee25ab23164f79a57d5f92639bf4d2 : SUCCESS in 49m 05s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c981b93b785d4e53ad3839e23bb86580 : SUCCESS in 1h 07m 15s","accounts_in_message":[],"_revision_number":1},{"id":"21093803ee9af80dedd17f491e1c2d37ce8b4e06","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-07-15 08:12:53.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2e861f9780dc8956541514c4b42555ac22c7413c","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-07-18 07:58:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"21d5988609e0d6ab8dd1b9ba0d7e571fc0e77b30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 07:58:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e16cfa38397ddc203ac71947a097b0b3516d6778","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 09:22:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc242ed280fb4ddf8d284818597309a4\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/37cc13ec3f19498ba9ed3f499b0e23a5 : SUCCESS in 31m 40s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/661406bf6e9f42efb26540ccd6507fd2 : SUCCESS in 1h 07m 55s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/56981e23c89243c39ba890acc4a4efdb : SUCCESS in 1h 23m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66983fd3f88c4c84b13daf0aced88bd1 : SUCCESS in 6m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a7c94bef8e484f06876c3ae67f1a4ceb : SUCCESS in 8m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/921a6d1418cf44dfafbb34b4a2a95242 : SUCCESS in 6m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/601c72e99f964172bfd84878c9655d8f : SUCCESS in 8m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aae997ad369441eba2dd0c6bb3756ea0 : SUCCESS in 6m 16s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1b46cd209a974e91acf12559d6a50c6f : SUCCESS in 45m 42s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/876399ea0d93413f83232b28eaeb95bd : SUCCESS in 52m 35s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/09de85e3bdae492d84d3dde1aebd5aa5 : SUCCESS in 59m 29s","accounts_in_message":[],"_revision_number":1},{"id":"1e035ac5e145cc3efb2db56e9d5a69cd3eb115d8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 09:22:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f6efd7d6a7d14457501593cd21e0a56a92a2689b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 09:23:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e5da9486a5a48bd9a6de74267b7c0ee\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71fdeaa20b61491587d0e031d6494afe : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/75c5af66c9514856b184ad624196aee4 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1cc710d00b2c9bcbbb8d444caa2f8cc360b76bae","revisions":{"1cc710d00b2c9bcbbb8d444caa2f8cc360b76bae":{"kind":"REWORK","_number":1,"created":"2022-07-14 17:37:56.000000000","uploader":{"_account_id":12640,"name":"Arun S A G","email":"sagarun@gmail.com","username":"sagarun"},"ref":"refs/changes/17/849817/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/849817/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/849817/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/849817/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/849817/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/849817/1"}}},"commit":{"parents":[{"commit":"0541e933286b2ec3e0b4342870cec1c945da59b1","subject":"Bump prometheus-openstack-exporter version to 1.6.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0541e933286b2ec3e0b4342870cec1c945da59b1"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-10-28 07:17:17.000000000","tz":120},"committer":{"name":"Arun S A G","email":"sagarun@gmail.com","date":"2022-07-14 17:37:56.000000000","tz":0},"subject":"neutron: Use update-alternatives --display instead of --query","message":"neutron: Use update-alternatives --display instead of --query\n\n--query is not supported before update-alternatives 1.15\nand CentOS uses 1.13\n\nCloses-Bug: #1936947\nChange-Id: I78da7eb2b6b5eae2b04fd95b24615bd4a572a1d6\n(cherry picked from commit 2d1bbe72579f9209e151133352d4d6761f764f33)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1cc710d00b2c9bcbbb8d444caa2f8cc360b76bae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1cc710d00b2c9bcbbb8d444caa2f8cc360b76bae"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
