)]}'
{"id":"openstack%2Fkolla~849818","triplet_id":"openstack%2Fkolla~stable%2Fxena~I78da7eb2b6b5eae2b04fd95b24615bd4a572a1d6","project":"openstack/kolla","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"12640":{"account":{"_account_id":12640,"name":"Arun S A G","email":"sagarun@gmail.com","username":"sagarun"},"last_update":"2022-07-18 09:12:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I78da7eb2b6b5eae2b04fd95b24615bd4a572a1d6","subject":"neutron: Use update-alternatives --display instead of --query","status":"MERGED","created":"2022-07-14 17:42:05.000000000","updated":"2022-07-18 09:13:45.000000000","submitted":"2022-07-18 09:12:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849818","cherry_pick_of_change":815781,"cherry_pick_of_patch_set":2,"meta_rev_id":"eaada1ec1c4c0f6cbebca5fe09e83f52b3ea321a","_number":849818,"virtual_id_number":849818,"owner":{"_account_id":12640,"name":"Arun S A G","email":"sagarun@gmail.com","username":"sagarun"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23266,"name":"lixuehai","email":"li.xuehai@99cloud.net","username":"lixuehai"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-18 09:12:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"recommended":{"_account_id":23266,"name":"lixuehai","email":"li.xuehai@99cloud.net","username":"lixuehai"},"all":[{"value":1,"date":"2022-07-15 08:22:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23266,"name":"lixuehai","email":"li.xuehai@99cloud.net","username":"lixuehai"},{"value":2,"date":"2022-07-18 07:57:47.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2022-07-15 08:12:48.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":23266,"name":"lixuehai","email":"li.xuehai@99cloud.net","username":"lixuehai"},{"value":1,"date":"2022-07-18 07:57:47.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23266,"name":"lixuehai","email":"li.xuehai@99cloud.net","username":"lixuehai"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":23266,"name":"lixuehai","email":"li.xuehai@99cloud.net","username":"lixuehai"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23266,"name":"lixuehai","email":"li.xuehai@99cloud.net","username":"lixuehai"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-14 17:42:05.000000000","updated_by":{"_account_id":12640,"name":"Arun S A G","email":"sagarun@gmail.com","username":"sagarun"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2022-07-14 19:08:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-15 08:12:48.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-07-15 08:22:33.000000000","updated_by":{"_account_id":23266,"name":"lixuehai","email":"li.xuehai@99cloud.net","username":"lixuehai"},"reviewer":{"_account_id":23266,"name":"lixuehai","email":"li.xuehai@99cloud.net","username":"lixuehai"},"state":"REVIEWER"},{"updated":"2022-07-18 07:57:47.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"}],"messages":[{"id":"7f80d96b047062142822782604f360a0a06c2161","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12640,"name":"Arun S A G","email":"sagarun@gmail.com","username":"sagarun"},"date":"2022-07-14 17:42:05.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"5542162c7275315219eced604f2a4322fdcc2d73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 19:08:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a346683203df4d4591bfc610415912d7\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/b73dbb9157c046e896b051deb3bb7e4a : SUCCESS in 24m 07s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/945e2c3c826e4d408639b048d1b6f116 : SUCCESS in 28m 49s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/e818f6cd716e4951a878831a999f173e : SUCCESS in 1h 17m 45s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/d0b3afe0c19e4f5082407b4a815e0314 : SUCCESS in 1h 16m 16s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/298f429a55c64bb78b67506e169b3840 : SUCCESS in 1h 25m 29s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/0b34f8aa5ea1454291c3847d09e7c4a8 : SUCCESS in 16m 39s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d3d888d62f504b57832b31566fd35baa : SUCCESS in 16m 46s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/bd0d6904f51e4f25a5cc4ed659c1f20c : SUCCESS in 1h 05m 22s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebbc48a1c3ac4a41957c1bc628434528 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4124292ed2384f30bd44fd9aadadb08f : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d16e5821af084145bc45a09f911ad52a : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9aa1aa54322040b0b1222abf6ddaa965 : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/775bc8344cd54ee0b3d5a326836ebdd9 : SUCCESS in 6m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/997fb258662c4d97b7ad3ff3e200fa3e : SUCCESS in 5m 33s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0e4564c652064344bc4e46640b618634 : SUCCESS in 23m 52s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/23211e60c96746c8a8f33df2b3a1e607 : SUCCESS in 25m 28s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8f319df9171c42688b297097dddde212 : SUCCESS in 56m 42s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d37ea95c0d4a462c8cbd9d3179378c70 : SUCCESS in 55m 25s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e1723b3d4a0242a0aa5d2b19b8024f3d : SUCCESS in 1h 04m 40s","accounts_in_message":[],"_revision_number":1},{"id":"36cff7dee2cdcd25f566f441523494d3a81fd083","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 19:26:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67796101ed3f4dbe947667509a19ffcd\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/3245585f28f14a7c98b38b050b2b6d4a : SUCCESS in 1h 00m 35s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2e6cbf7a32d7470d8c4226f24d24e84c : SUCCESS in 1h 29m 27s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/525c1f679b6b4c4c8832c91eb9e94e97 : SUCCESS in 1h 36m 43s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/355a3f33a86e4687901c4904d56e9d98 : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c51a9255eb594b7baf61bf6f47d8d4d3 : SUCCESS in 4m 13s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/9641853d8e034cc2b1bdc4b403e3835e : FAILURE in 1h 37m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1420898a97dbaa133751a89dd4a20e709b6ce4e8","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-07-15 08:12:48.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a67e0201dee1714cf95aade0364cf8a109e5ce2d","author":{"_account_id":23266,"name":"lixuehai","email":"li.xuehai@99cloud.net","username":"lixuehai"},"date":"2022-07-15 08:22:33.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"848f825815ed63b13e83e611158612a3f43c23b8","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-07-18 07:57:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"293417a5e5c7060210f546322e2b1b9630ef4255","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 07:58:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1aebbca672d92c1cffdfc3de30afece15c1f0b97","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 09:12:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae2a98aaaa114cd390332ece854a96e5\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/d9ef662211684900b15541a6c9bb13d8 : SUCCESS in 30m 53s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/f29ee8e72af14f8ba174f866453101ac : SUCCESS in 57m 04s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/de6ee6f0340147d2acc8d977c679ae43 : SUCCESS in 1h 04m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a008727ba1264c08814964bc48aee408 : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f894ff686074925a7403123f58fb8e5 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8b9b4dbf288b43b3a664bbf2d3a17cf7 : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ceca084c135b4e318c9fd2b60b49530a : SUCCESS in 6m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c46c22366dd841a6b4d2e06913c42b22 : SUCCESS in 4m 11s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6cc7aa0aa55f4cddb2ca28f473992a8b : SUCCESS in 21m 42s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3c8272cfe5a843588d606493f7a3d56a : SUCCESS in 44m 46s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/26e3b67d887348f39251486d9a5f6eea : SUCCESS in 56m 34s","accounts_in_message":[],"_revision_number":1},{"id":"a9c34697494c415b53c8f5bc3d7f3931c87eb1e4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 09:12:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"eaada1ec1c4c0f6cbebca5fe09e83f52b3ea321a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 09:13:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e090bd329b64433f9b464c5af03c5c4e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a826275294049bc8303537072bd37d8 : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9eb6030fd20849458a39b7013cdedef7 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"22ce46488b4cef8b35f328948a80ba9ba6fad8d9","revisions":{"22ce46488b4cef8b35f328948a80ba9ba6fad8d9":{"kind":"REWORK","_number":1,"created":"2022-07-14 17:42:05.000000000","uploader":{"_account_id":12640,"name":"Arun S A G","email":"sagarun@gmail.com","username":"sagarun"},"ref":"refs/changes/18/849818/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/18/849818/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/849818/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/849818/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/849818/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/18/849818/1"}}},"commit":{"parents":[{"commit":"51d5e442c105bd5dd74c983ef277812b8915eb52","subject":"Bump prometheus-openstack-exporter version to 1.6.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/51d5e442c105bd5dd74c983ef277812b8915eb52"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-10-28 07:17:17.000000000","tz":120},"committer":{"name":"Arun S A G","email":"sagarun@gmail.com","date":"2022-07-14 17:42:05.000000000","tz":0},"subject":"neutron: Use update-alternatives --display instead of --query","message":"neutron: Use update-alternatives --display instead of --query\n\n--query is not supported before update-alternatives 1.15\nand CentOS uses 1.13\n\nCloses-Bug: #1936947\nChange-Id: I78da7eb2b6b5eae2b04fd95b24615bd4a572a1d6\n(cherry picked from commit 2d1bbe72579f9209e151133352d4d6761f764f33)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/22ce46488b4cef8b35f328948a80ba9ba6fad8d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/22ce46488b4cef8b35f328948a80ba9ba6fad8d9"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
