)]}'
{"id":"openstack%2Fkolla~850761","triplet_id":"openstack%2Fkolla~master~Ibc88fa962e37f6313300f2c7b19d0c42948c4376","project":"openstack/kolla","branch":"master","topic":"bug/1982588","attention_set":{},"removed_from_attention_set":{"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-08-03 16:46:30.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_30491\u003e by clicking the attention icon","reason_account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},"32398":{"account":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"last_update":"2023-09-06 09:36:47.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ibc88fa962e37f6313300f2c7b19d0c42948c4376","subject":"Add additional grafana plugins install support:","status":"ABANDONED","created":"2022-07-22 14:51:20.000000000","updated":"2023-09-06 09:36:47.000000000","total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"8823d544bd233753fcf729b79a88596fa33fd8fd","_number":850761,"virtual_id_number":850761,"owner":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-07-22 16:27:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-22 14:53:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-22 14:53:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-28 08:43:29.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2023-03-25 20:27:07.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REMOVED"}],"messages":[{"id":"8439eb696288e6aaadbf250e10b4123e54f3c062","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2022-07-22 14:51:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a26f0889bb52702f96ab77c4c32ec299368a3674","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 14:53:07.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"540c5449f0f281c1a32d6daf991e96541fbc5b69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 14:53:10.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"a1725e80fc7fe2c0fd978fe60a6877c9fb33cb94","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2022-07-22 14:53:22.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"eb41a484508a1092c56d0906b22fedee07898a80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2022-07-22 15:07:05.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2d20514f6bbb3c1f65c70650ee9e147274bd3a37","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2022-07-22 15:09:44.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1389e6651c92ad5d72d5ac28935959c4487b429a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 16:27:07.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cd1ca1d39059402eb2fcd7195c12a211\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/48c09b2df90d43b3a4d489a783acfe85 : FAILURE in 27m 26s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/aca80d76051f453db0b99e3efd36a098 : SUCCESS in 57m 56s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/db86f7c375164797a444529504b9d1b1 : SUCCESS in 1h 12m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/247eb13ab3cb4741b04f3f93e11f88d2 : FAILURE in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb4751a7b2a642ba8f58f5f274de1cdc : SUCCESS in 3m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d5570c8e627244549802f43cd06ddd0f : SUCCESS in 4m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/732f932cca414f78bd623bf2d11669ba : SUCCESS in 3m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ee9995e2c594750b25e826040cb2bf4 : SUCCESS in 6m 37s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/11faa05206f04c97b5c60a8114be7b80 : FAILURE in 19m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bcbd228d64c543abbc458a85d08caa89 : SUCCESS in 50m 17s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a58e9a6f3b8b417081911133b6224c7a : SUCCESS in 1h 05m 49s","accounts_in_message":[],"_revision_number":4},{"id":"ee4febc05ca837b85e47ad9095251bd63e2ccb1e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 17:42:22.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11f12bae626f4b41934954def89f9b33\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d4bf6ae6ad824286ac02bdc4fa89229d : FAILURE in 2h 03m 15s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c5704e9aed634d0cae41761f1f7c0694 : SUCCESS in 2h 30m 56s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/111a0eb1172a4c80b0691cd83e2b5d9b : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d95c4a861e7e40e8ac333e1c89239949 : SUCCESS in 7m 10s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/c8494b3fa5d64c4494032b332231bb6e : FAILURE in 1h 49m 42s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a8c61f3a0366ab57cf88f3066becd8be9047afe1","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-07-28 08:43:29.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9a74eaf3f60503a491d8c9d75f44457c8ca3d45c","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2022-07-28 10:29:18.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"68f442b8dc0c13637a3bde1c92cf3c64eeeca2a3","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2023-03-25 20:27:07.000000000","message":"Removed reviewer \u003cGERRIT_ACCOUNT_30491\u003e with the following votes:\n\n* Code-Review-1 by \u003cGERRIT_ACCOUNT_30491\u003e\n","accounts_in_message":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}],"_revision_number":4},{"id":"8823d544bd233753fcf729b79a88596fa33fd8fd","tag":"autogenerated:gerrit:abandon","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2023-09-06 09:36:47.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"955ef1f8f68f3c3cd6120414170dc4ddad215a0f","revisions":{"30a45204bb95d142385a2fec902e6b79060415a8":{"kind":"REWORK","_number":1,"created":"2022-07-22 14:51:20.000000000","uploader":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"ref":"refs/changes/61/850761/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/61/850761/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/850761/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/850761/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/850761/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/61/850761/1"}}},"commit":{"parents":[{"commit":"ea5d95d3f26c5d95a8084d59233692ce9cdd65f3","subject":"Merge \"Change kolla_version LABEL to git sha-1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ea5d95d3f26c5d95a8084d59233692ce9cdd65f3"}]}],"author":{"name":"Gaël THEROND (Fl1nt)","email":"gael.therond@bitswalk.com","date":"2022-07-22 14:51:03.000000000","tz":120},"committer":{"name":"Gaël THEROND (Fl1nt)","email":"gael.therond@bitswalk.com","date":"2022-07-22 14:51:03.000000000","tz":120},"subject":"Add additional grafana plugins install support:","message":"Add additional grafana plugins install support:\n\n  - Add a way to install additional grafana plugins.\n  - Add a way to install them offline.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/850760\nChange-Id: Ibc88fa962e37f6313300f2c7b19d0c42948c4376\nClose-Bug: #1982588\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/30a45204bb95d142385a2fec902e6b79060415a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/30a45204bb95d142385a2fec902e6b79060415a8"}]},"branch":"refs/heads/master"},"030ae447b830f4068096fa03f3af3e73d174f559":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-07-22 14:53:22.000000000","uploader":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"ref":"refs/changes/61/850761/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/61/850761/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/850761/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/850761/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/850761/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/61/850761/2"}}},"commit":{"parents":[{"commit":"ea5d95d3f26c5d95a8084d59233692ce9cdd65f3","subject":"Merge \"Change kolla_version LABEL to git sha-1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ea5d95d3f26c5d95a8084d59233692ce9cdd65f3"}]}],"author":{"name":"Gaël THEROND (Fl1nt)","email":"gael.therond@bitswalk.com","date":"2022-07-22 14:51:03.000000000","tz":120},"committer":{"name":"Gaël THEROND","email":"gael.therond@bitswalk.com","date":"2022-07-22 14:53:22.000000000","tz":0},"subject":"Add additional grafana plugins install support:","message":"Add additional grafana plugins install support:\n\n  - Add a way to install additional grafana plugins.\n  - Add a way to install them offline.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/850760\nChange-Id: Ibc88fa962e37f6313300f2c7b19d0c42948c4376\nCloses-Bug: #1982588\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/030ae447b830f4068096fa03f3af3e73d174f559"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/030ae447b830f4068096fa03f3af3e73d174f559"}]},"branch":"refs/heads/master","description":"Edit commit message"},"2c7408184260348ec835daf1a6df1269a5c84114":{"kind":"REWORK","_number":3,"created":"2022-07-22 15:07:05.000000000","uploader":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"ref":"refs/changes/61/850761/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/61/850761/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/850761/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/850761/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/850761/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/61/850761/3"}}},"commit":{"parents":[{"commit":"ea5d95d3f26c5d95a8084d59233692ce9cdd65f3","subject":"Merge \"Change kolla_version LABEL to git sha-1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ea5d95d3f26c5d95a8084d59233692ce9cdd65f3"}]}],"author":{"name":"Gaël THEROND (Fl1nt)","email":"gael.therond@bitswalk.com","date":"2022-07-22 15:06:50.000000000","tz":120},"committer":{"name":"Gaël THEROND (Fl1nt)","email":"gael.therond@bitswalk.com","date":"2022-07-22 15:06:50.000000000","tz":120},"subject":"Add additional grafana plugins install support:","message":"Add additional grafana plugins install support:\n\n  - Add a way to install additional grafana plugins.\n  - Add a way to install them offline.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/850760\nChange-Id: Ibc88fa962e37f6313300f2c7b19d0c42948c4376\nCloses-Bug: #1982588\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2c7408184260348ec835daf1a6df1269a5c84114"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2c7408184260348ec835daf1a6df1269a5c84114"}]},"branch":"refs/heads/master"},"955ef1f8f68f3c3cd6120414170dc4ddad215a0f":{"kind":"REWORK","_number":4,"created":"2022-07-22 15:09:44.000000000","uploader":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"ref":"refs/changes/61/850761/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/61/850761/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/850761/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/850761/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/850761/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/61/850761/4"}}},"commit":{"parents":[{"commit":"ea5d95d3f26c5d95a8084d59233692ce9cdd65f3","subject":"Merge \"Change kolla_version LABEL to git sha-1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ea5d95d3f26c5d95a8084d59233692ce9cdd65f3"}]}],"author":{"name":"Gaël THEROND (Fl1nt)","email":"gael.therond@bitswalk.com","date":"2022-07-22 15:09:34.000000000","tz":120},"committer":{"name":"Gaël THEROND (Fl1nt)","email":"gael.therond@bitswalk.com","date":"2022-07-22 15:09:34.000000000","tz":120},"subject":"Add additional grafana plugins install support:","message":"Add additional grafana plugins install support:\n\n  - Add a way to install additional grafana plugins.\n  - Add a way to install them offline.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/850760\nChange-Id: Ibc88fa962e37f6313300f2c7b19d0c42948c4376\nCloses-Bug: #1982588\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/955ef1f8f68f3c3cd6120414170dc4ddad215a0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/955ef1f8f68f3c3cd6120414170dc4ddad215a0f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
