)]}'
{"id":"openstack%2Fkolla~850911","triplet_id":"openstack%2Fkolla~stable%2Fxena~I232fe835f5a650c258243f1ee4720480f2e6e7b1","project":"openstack/kolla","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-07-26 09:56:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I232fe835f5a650c258243f1ee4720480f2e6e7b1","subject":"Fix container image publish to Docker Hub","status":"MERGED","created":"2022-07-25 15:12:12.000000000","updated":"2022-07-26 09:57:55.000000000","submitted":"2022-07-26 09:56:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"850911","cherry_pick_of_change":850873,"cherry_pick_of_patch_set":2,"meta_rev_id":"ef7acfe3f1d70c8847ba250a75c7284f77c5b525","_number":850911,"virtual_id_number":850911,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-26 09:56:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2022-07-26 08:06:43.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2022-07-25 17:10:43.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2022-07-26 08:06:43.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-25 16:47:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-25 17:10:43.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-07-26 08:06:43.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"}],"messages":[{"id":"13b301a7314815c0d085fef3220ef79dfc58a8d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-07-25 15:12:12.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c12050e2bd39658023e2b942e0a1c84396ae7fd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 16:47:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac7c1378229a4cc9818ad275e595d60a\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/e58726eee9244c429f1364fea19ec2e6 : SUCCESS in 43m 22s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/0c18b95168c44c5c907deebe0ddbfa0e : SUCCESS in 1h 05m 12s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/1a221669feb94e00a196d6941ec2de90 : SUCCESS in 1h 27m 54s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/44ec95e11c7847acab39cfe30b50e873 : SUCCESS in 1h 32m 18s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/31ffdcd01f59429283e23905b027dcd6 : SUCCESS in 1h 33m 08s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/033e88fc0d4544639d1c37da182985d1 : SUCCESS in 26m 15s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d4b4d5f776cb4a54a669d5a3aeb9a278 : SUCCESS in 19m 19s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2e28efce7bfa4c40ab1e3f58857cf9be : SUCCESS in 1h 05m 29s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03056100e75d42b0b5216fbe2ee32ac8 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/223a7a159463401ca766a02abe0178e8 : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4517db5a5f3e4d59b3b88c8f1bb5f5ad : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f6a39e76daf4a33ba2a244c46acb3f3 : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e77758a8c4f4e54b5b1dbeb2ac917be : SUCCESS in 7m 40s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/603a5f674b7b4514b6ece9207ac8e723 : SUCCESS in 23m 15s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/fdf93f864bc4438d93a33ada41dab042 : SUCCESS in 21m 02s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/09508b61d5924b868553c1a7a10ae028 : SUCCESS in 1h 01m 32s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f7602478f0714cbfb61c1fad96bcb642 : SUCCESS in 55m 07s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f526d2cdb11c44eab6a75bf6e96f0105 : SUCCESS in 1h 13m 19s","accounts_in_message":[],"_revision_number":1},{"id":"3a24ffd9545a4ef8835688a73e7519176b0e51f5","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-07-25 17:10:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"02607ab44a3345b788feaeb03fe4d24b3b0a4279","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 17:54:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c99a8d4cb3a446ff85767bdd94ce1833\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/1348f28fb1084723aa4c101c34ecdd8f : SUCCESS in 2h 20m 08s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4cea8a3d527843598c8a64e367374920 : SUCCESS in 2h 03m 53s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/52c93d5beb264e3cb8b95b8cfdb7b245 : SUCCESS in 2h 22m 31s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/94512af267004302a582c4c324545087 : SUCCESS in 11m 27s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9390b81c56694afab1df01d0603ff6cb : SUCCESS in 5m 42s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/110b6ccd676e44a0bb662e3a37996eae : FAILURE in 2h 21m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4c4cb565fcee4eaddb29c410348597f0f444ffda","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-07-26 08:06:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1b8a505429c8d0ceba43fb1e17e6c63ef70a3f2f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 08:07:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7d0fb22a81e79e2677833f79036d41207dd6d177","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 09:56:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2cc203cafd8a41d885eb6360e932fb96\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/bc3c16c1609c43ba8be74c35e62ea1bb : SUCCESS in 34m 11s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/3d07f601547b4b1a8f9acf228a5f6f17 : SUCCESS in 1h 19m 51s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/8cbc99c00e0947d5accb16946f0d4dee : SUCCESS in 1h 29m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5430617436fd4985a8852c244db80a6c : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/63dd5dde16974a5bb63020a641fb617c : SUCCESS in 3m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eb152f964ffa46d3ba51c479f4e50d36 : SUCCESS in 3m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4563df3148041c4b6100de27234af0d : SUCCESS in 5m 07s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f291f5c42494403597bf3e44b997e894 : SUCCESS in 21m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0a1420213f59456a8bd0d8e6d598652e : SUCCESS in 53m 27s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/da60c65c370d4f28b23ef3a97f667fb3 : SUCCESS in 1h 08m 41s","accounts_in_message":[],"_revision_number":1},{"id":"12da9f4c2476f23921ad5da99ce0f67919f60898","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 09:56:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ef7acfe3f1d70c8847ba250a75c7284f77c5b525","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 09:57:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53f8a5af2cdf46e8b9be2a24f7089788\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14a477fa621a4b419fe5307a2774addc : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bc4890ba283407f9097ec86924a991c922eb868b","revisions":{"bc4890ba283407f9097ec86924a991c922eb868b":{"kind":"REWORK","_number":1,"created":"2022-07-25 15:12:12.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/11/850911/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/11/850911/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/850911/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/850911/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/850911/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/11/850911/1"}}},"commit":{"parents":[{"commit":"4cf7c7c7a1f9db81c2b5a220e29f88f668b5aefc","subject":"Merge \"Unify curl use (customisation options)\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4cf7c7c7a1f9db81c2b5a220e29f88f668b5aefc"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-07-25 11:10:15.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-07-25 15:12:12.000000000","tz":0},"subject":"Fix container image publish to Docker Hub","message":"Fix container image publish to Docker Hub\n\nPublishing container images to Docker Hub was failing with:\n\n    state is present but all of the following are missing: source\n\nThis is caused by the recent Ansible upgrade by Zuul [1]: the source\nparameter is mandatory since Ansible 2.12.\n\n[1] https://review.opendev.org/c/openstack/project-config/+/849120\n\nChange-Id: I232fe835f5a650c258243f1ee4720480f2e6e7b1\n(cherry picked from commit 2da76f78dee53ae84bfe6da051367d9b320af0f6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bc4890ba283407f9097ec86924a991c922eb868b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bc4890ba283407f9097ec86924a991c922eb868b"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
