)]}'
{"id":"openstack%2Fkolla~851211","triplet_id":"openstack%2Fkolla~stable%2Fussuri~I6b0b677067d8b307798b6df113cf5a45bb7525ab","project":"openstack/kolla","branch":"stable/ussuri","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-07-27 13:51:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6b0b677067d8b307798b6df113cf5a45bb7525ab","subject":"Mark monasca-grafana as unbuildable","status":"MERGED","created":"2022-07-27 11:03:50.000000000","updated":"2022-07-27 14:13:30.000000000","submitted":"2022-07-27 13:51:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"851211","meta_rev_id":"51077857ea73e84480da5a6057021b73015d25c5","_number":851211,"virtual_id_number":851211,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-27 13:51:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2022-07-27 11:25:45.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2022-07-27 12:38:36.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2022-07-27 12:38:36.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-27 11:25:45.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2022-07-27 12:33:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-27 12:38:36.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"7c3279fc99302c04e785e56660d9812d2d53a91c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-07-27 11:03:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6b844f5b959a4966c77e1cf941a2d6d90df6eca0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-07-27 11:24:56.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"2208c0cd2877d5dd703de2a25f0467390ac79d9f","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-07-27 11:25:45.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"601957e175ae5baf5fe306da6e9ae0d2e1d7c949","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 12:33:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6ce77d932c34ef5bedde29afc7845b3\n\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/79f5d99f3e9745d09d77750f9c05280f : SUCCESS in 1h 04m 21s\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/fe296f94539f4c0ba1ba34e491d7257e : SUCCESS in 36m 14s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1ee5f37feaa1416ea763841ecfb3a9d4 : SUCCESS in 23m 21s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/864a9a4e672c4036b9b06b3e9a61c36b : SUCCESS in 19m 38s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/dc62e928e1da47c697b486e6b2aded23 : SUCCESS in 54m 57s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef6739ad6d1943d4b43791828ad4eaa0 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/64e1d9e380854b0a81c4fe8ed104a33f : SUCCESS in 3m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/de86551bb0994302a38572ec0ff9f4fd : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e9d01c3a1bc4c3fa4dc04eae752a5d0 : SUCCESS in 6m 24s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/199e83b8aa724d5f87b6c2ba45aeb5cd : SUCCESS in 23m 58s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/def3af63f234474ea6b2cbadfb191218 : SUCCESS in 25m 54s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8d2e911037e84798babcf49498ef574e : SUCCESS in 43m 14s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f22b5885417744649ee41956a8a8d527 : SUCCESS in 55m 27s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/718154caf4e441cf8c692f2e182393b6 : SUCCESS in 1h 05m 31s","accounts_in_message":[],"_revision_number":2},{"id":"ec71145bd64e7b83011b807c606f941c51b31741","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-07-27 12:38:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f5cdd7230df92dca65e25f8b199acbe39db0bf3f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 12:39:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fbaade15d6200f96d63e977f5e821bbf897fba5c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 13:51:20.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c534fc9845bd431eb16df696d493ba2e\n\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/3b539204cfe34d729ff04b06e53e086f : SUCCESS in 1h 09m 20s\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/a20b68c8e82640c689949fdcee58b73b : SUCCESS in 32m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f2e598bab1946018c0145e44b433d38 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a3217e844094d518dca549627fa50ed : SUCCESS in 5m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/22a044ad3166439da32b690bed74e6a3 : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/714618cf823142f6aa1a361d8ea10ae4 : SUCCESS in 6m 18s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c7256195bd164f85b8fb3dd4cb0e29df : SUCCESS in 30m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/662a399b4f5c433e92eb8e8fca670543 : SUCCESS in 56m 36s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/85eb7ed5b2fa4994b27e23489c2f6dfa : SUCCESS in 1h 03m 41s","accounts_in_message":[],"_revision_number":2},{"id":"bb3dfb42b6ec0dcf52a8f6b74fb003619dacd17a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 13:51:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"23e6df6efccd2a8489facb90292241bd6228febd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 13:52:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c57ba1bd55c4d6491dc1da98f687997\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f60ce342dfb428aab2772cae0cc4ea9 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":2},{"id":"51077857ea73e84480da5a6057021b73015d25c5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 14:13:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1484e9a3f0f74543b9d793aad35ba14c\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/468024cf9773456d87c7cccbd2a1b9ef : SUCCESS in 2h 45m 42s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9d0699dc7f3041eea22f9ba3541acf7f : FAILURE in 29m 44s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/562f29818668460fbc4ac6de8e897f12 : FAILURE in 1h 04m 33s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/fd56c78bf24c47dbab9347b6739f2226 : FAILURE in 2h 19m 05s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"780180088d010745a3b3a4158229b1d0c1026d3b","revisions":{"e319319170d2a6457d694077035ef81fa4cdd365":{"kind":"REWORK","_number":1,"created":"2022-07-27 11:03:50.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/11/851211/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/11/851211/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/851211/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/851211/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/851211/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/11/851211/1"}}},"commit":{"parents":[{"commit":"d926c35f0f2e33d728d2390b93e6f5cd3d56ecb2","subject":"Fix Ubuntu image builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d926c35f0f2e33d728d2390b93e6f5cd3d56ecb2"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-07-27 11:03:27.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-07-27 11:03:27.000000000","tz":120},"subject":"Mark monasca-grafana as unbuildable","message":"Mark monasca-grafana as unbuildable\n\nChange-Id: I6b0b677067d8b307798b6df113cf5a45bb7525ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e319319170d2a6457d694077035ef81fa4cdd365"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e319319170d2a6457d694077035ef81fa4cdd365"}]},"branch":"refs/heads/stable/ussuri"},"780180088d010745a3b3a4158229b1d0c1026d3b":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-07-27 11:24:56.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/11/851211/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/11/851211/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/851211/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/851211/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/851211/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/11/851211/2"}}},"commit":{"parents":[{"commit":"d926c35f0f2e33d728d2390b93e6f5cd3d56ecb2","subject":"Fix Ubuntu image builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d926c35f0f2e33d728d2390b93e6f5cd3d56ecb2"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-07-27 11:24:37.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-07-27 11:24:37.000000000","tz":120},"subject":"Mark monasca-grafana as unbuildable","message":"Mark monasca-grafana as unbuildable\n\nImage build is failing with the following error:\n\n    fpm -t deb -s dir --description Grafana -C /tmp/grafana-linux-pack475276015 --vendor Grafana --url https://grafana.com --license \"Apache 2.0\" --maintainer contact@grafana.com --config-files /etc/init.d/grafana-server --config-files /etc/default/grafana-server --config-files /usr/lib/systemd/system/grafana-server.service --after-install packaging/deb/control/postinst --name grafana --version 4.2.0 -p ./dist --deb-no-default-config-files --iteration 1658863511 --depends adduser --depends libfontconfig .\n    /usr/lib/ruby/2.5.0/rubygems/core_ext/kernel_require.rb:59:in `require\u0027: /var/lib/gems/2.5.0/gems/dotenv-2.8.0/lib/dotenv/parser.rb:87: syntax error, unexpected \u0027]\u0027 (SyntaxError)\n          !line.split[1..].all? { |var| @hash.member?(v...\n                         ^\n    \tfrom /usr/lib/ruby/2.5.0/rubygems/core_ext/kernel_require.rb:59:in `require\u0027\n    \tfrom /var/lib/gems/2.5.0/gems/pleaserun-0.0.32/lib/pleaserun/platform/base.rb:1:in `\u003ctop (required)\u003e\u0027\n    \tfrom /usr/lib/ruby/2.5.0/rubygems/core_ext/kernel_require.rb:59:in `require\u0027\n    \tfrom /usr/lib/ruby/2.5.0/rubygems/core_ext/kernel_require.rb:59:in `require\u0027\n    \tfrom /var/lib/gems/2.5.0/gems/pleaserun-0.0.32/lib/pleaserun/cli.rb:8:in `\u003ctop (required)\u003e\u0027\n    \tfrom /usr/lib/ruby/2.5.0/rubygems/core_ext/kernel_require.rb:59:in `require\u0027\n    \tfrom /usr/lib/ruby/2.5.0/rubygems/core_ext/kernel_require.rb:59:in `require\u0027\n    \tfrom /var/lib/gems/2.5.0/gems/fpm-1.14.2/lib/fpm/package/pleaserun.rb:6:in `\u003ctop (required)\u003e\u0027\n    \tfrom /usr/lib/ruby/2.5.0/rubygems/core_ext/kernel_require.rb:59:in `require\u0027\n    \tfrom /usr/lib/ruby/2.5.0/rubygems/core_ext/kernel_require.rb:59:in `require\u0027\n    \tfrom /var/lib/gems/2.5.0/gems/fpm-1.14.2/lib/fpm/command.rb:18:in `block in \u003ctop (required)\u003e\u0027\n    \tfrom /var/lib/gems/2.5.0/gems/fpm-1.14.2/lib/fpm/command.rb:15:in `each\u0027\n    \tfrom /var/lib/gems/2.5.0/gems/fpm-1.14.2/lib/fpm/command.rb:15:in `\u003ctop (required)\u003e\u0027\n    \tfrom /usr/lib/ruby/2.5.0/rubygems/core_ext/kernel_require.rb:59:in `require\u0027\n    \tfrom /usr/lib/ruby/2.5.0/rubygems/core_ext/kernel_require.rb:59:in `require\u0027\n    \tfrom /var/lib/gems/2.5.0/gems/fpm-1.14.2/bin/fpm:5:in `\u003ctop (required)\u003e\u0027\n    \tfrom /usr/local/bin/fpm:23:in `load\u0027\n    \tfrom /usr/local/bin/fpm:23:in `\u003cmain\u003e\u0027\n\nChange-Id: I6b0b677067d8b307798b6df113cf5a45bb7525ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/780180088d010745a3b3a4158229b1d0c1026d3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/780180088d010745a3b3a4158229b1d0c1026d3b"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
