)]}'
{"id":"openstack%2Fkolla~851502","triplet_id":"openstack%2Fkolla~master~Icce201677dceaab8ca6f045158bbf565eef3a324","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-07-29 12:18:52.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-07-29 13:35:57.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-07-29 13:35:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icce201677dceaab8ca6f045158bbf565eef3a324","subject":"mariadb-clustercheck: do not install xinetd","status":"MERGED","created":"2022-07-29 10:11:07.000000000","updated":"2022-07-29 13:37:02.000000000","submitted":"2022-07-29 13:35:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"851502","meta_rev_id":"fbdd09d14bb61d2aef312e5c8aa15a06be1a407e","_number":851502,"virtual_id_number":851502,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-29 13:35:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-07-29 12:18:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-07-29 10:14:52.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-07-29 12:18:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-29 10:14:52.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-07-29 10:15:00.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-07-29 10:15:00.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-07-29 11:27:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9f45a48a1fa03ea06eabaaf50a3d5a4e36b3c3bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-07-29 10:11:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e3274ade8103328c561eafa959691f5315f299a0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-07-29 10:14:52.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9b3042df73195461ca2f4aba59572b4c30909170","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 11:27:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15880b11a49c4eb09cdbdb513c6e53d5\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f05e0078640240bd942975d6a32e8322 : SUCCESS in 22m 49s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a69515987f9d444695cd06c895be65e2 : SUCCESS in 1h 00m 54s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/886c67ddb466479cbffaf3274df6c3bf : SUCCESS in 1h 11m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cf214e8e741414f9a8277cdf58e00de : SUCCESS in 2m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea98c668218046198fd4bdcecc8860c2 : SUCCESS in 3m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/27b17cc97d7140f49a76c7f71d0d42bd : SUCCESS in 2m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0908e838372b493d9f17b6a28373ce4c : SUCCESS in 3m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5b170ec0f204ae9b5f7e3a09e07d3bc : SUCCESS in 4m 54s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/43517fa47820495ea4406819b8387f7b : SUCCESS in 21m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f241c9cf9cd142899e3e8fd4a9d498b2 : SUCCESS in 53m 56s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b2e097e1515143b7ab8f151c044f9190 : SUCCESS in 1h 00m 39s","accounts_in_message":[],"_revision_number":1},{"id":"ae00f586b392fadc436a34b2b779f19d5c4fea61","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 12:18:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1bd21696e8bb4835937b278e257d78cc\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/caaedb8932d64f9689fa1bc59257aa97 : SUCCESS in 1h 36m 03s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/af0335bf4b7a47109139f8217ed1de0a : SUCCESS in 2h 05m 03s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1953a6792d11499688fb310d00e6e861 : SUCCESS in 7m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b2c2faa4671c44348a9f8cbae17d9e06 : SUCCESS in 7m 33s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/7a1a7b40069645829de0acb904fc2468 : SUCCESS in 1h 45m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ab77820b371dfdff8bf234a011f53fe54585cdee","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-07-29 12:18:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"985d5fb47e804da878702f8c60910071c38832dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 12:19:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bee0d4bb37e42fc668f8a0e3da59a371b80ff85d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 13:35:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0b4cd39d3e740c9847ffbbdfcb8f110\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ae9a0671dc1542fbaa504229bc581838 : SUCCESS in 21m 58s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2241703fe02543eca1328c291f37fb87 : SUCCESS in 56m 23s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/a26e385a2ceb4019a760a830fa873e68 : SUCCESS in 1h 12m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a19a561ec2fe4efc85bc500e9be54341 : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df5a34379c0a49abbb190a7dbba99f87 : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f4f67e25b47a4cab8813556859bf01c4 : SUCCESS in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8f8b5c511d04fdb898b2d9c85ff5dd8 : SUCCESS in 5m 29s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/dc5971c915cd455fa4c0d438baa603cd : SUCCESS in 23m 57s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/df593be393ca480eb68358883023b40d : SUCCESS in 56m 03s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b463635d81a4424d9a13416154216986 : SUCCESS in 1h 08m 33s","accounts_in_message":[],"_revision_number":1},{"id":"e8360e234faec494002856057600632f3354ccbd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 13:35:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fbdd09d14bb61d2aef312e5c8aa15a06be1a407e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 13:37:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c2e0227e75f499a961dde8aa4888974\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e7f31c08ed14598b080c1859658b8de : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b8572209e7190fe718610382a3ac77becfd39522","revisions":{"b8572209e7190fe718610382a3ac77becfd39522":{"kind":"REWORK","_number":1,"created":"2022-07-29 10:11:07.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/02/851502/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/02/851502/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/851502/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/851502/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/851502/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/02/851502/1"}}},"commit":{"parents":[{"commit":"d7af870f81f5c298321c364b3748d3262f850e3c","subject":"Merge \"Run proxysql with proxysql user\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d7af870f81f5c298321c364b3748d3262f850e3c"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-07-29 10:10:29.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-07-29 10:10:29.000000000","tz":120},"subject":"mariadb-clustercheck: do not install xinetd","message":"mariadb-clustercheck: do not install xinetd\n\nWe moved to socat on kolla-ansible side.\n\nChange-Id: Icce201677dceaab8ca6f045158bbf565eef3a324\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b8572209e7190fe718610382a3ac77becfd39522"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b8572209e7190fe718610382a3ac77becfd39522"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
