)]}'
{"id":"openstack%2Fkolla~853245","triplet_id":"openstack%2Fkolla~stable%2Fwallaby~I177949b9319a977c9cd9121eb28b710256b72a5a","project":"openstack/kolla","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-08-16 14:20:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I177949b9319a977c9cd9121eb28b710256b72a5a","subject":"Bump skydive version to 0.28","status":"MERGED","created":"2022-08-16 11:11:09.000000000","updated":"2022-08-16 14:21:39.000000000","submitted":"2022-08-16 14:20:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"853245","cherry_pick_of_change":846554,"cherry_pick_of_patch_set":2,"meta_rev_id":"7f2a1837f12bb1413fd17c28094b4bb086aca885","_number":853245,"virtual_id_number":853245,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-16 14:20:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2022-08-16 12:57:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2022-08-16 12:48:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2022-08-16 12:57:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":27446,"name":"Daniel Stoye","email":"d.stoye@cms.hu-berlin.de","username":"dasTor"},{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"}],"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-16 11:11:09.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":27446,"name":"Daniel Stoye","email":"d.stoye@cms.hu-berlin.de","username":"dasTor"},"state":"CC"},{"updated":"2022-08-16 12:44:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-16 12:57:08.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2022-08-16 13:43:59.000000000","updated_by":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"reviewer":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"state":"CC"}],"messages":[{"id":"145b2b4a77b63e0651981985539584d1e972764b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-08-16 11:11:09.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"e46351d0b96f2621a8f7df4dd15cecea6cb04303","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 12:44:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30b97a88128944fd9910463f9991dafa\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/ba6bf45223ba4c46a151a19ac4d7fc97 : SUCCESS in 28m 49s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/b0552717e25c486293c2aa53d15305ef : SUCCESS in 30m 09s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/183e00df16824cf98d489c8545889d02 : SUCCESS in 1h 09m 27s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/6fb21a2b77b446df9dad3f9b5f4d6af5 : SUCCESS in 59m 04s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/04a441dccd5a4d2d9393c8ed10e67e7d : SUCCESS in 1h 20m 15s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b162deeb49ff40d28c0148261174250f : SUCCESS in 20m 02s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/8cbff97498734626866aa375ec4d658c : SUCCESS in 22m 56s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7e41370b88394e06941869ed99cb03db : SUCCESS in 53m 37s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7923a9dbaf6b4327a28ca82553a8653c : SUCCESS in 6m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3895956213544dbb8a86fa9cb0450d01 : SUCCESS in 6m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc075aeb19c04baa833f015b5d87200a : SUCCESS in 7m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6e8523d2ad18477584725619caf6014d : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f508d4dd0356455ea92b4b250e61bd97 : SUCCESS in 9m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/335982c186cf4f819668c9838eff5d76 : SUCCESS in 6m 16s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6591e430116349aeae6e768439f3ed98 : SUCCESS in 20m 06s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/99ef9cc38991474ab5868d0278d012b0 : SUCCESS in 19m 28s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4308182561024c9eaf4cbfad817cd929 : SUCCESS in 55m 01s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d2ea49b3a4d74ed2a56eb2a1077d4469 : SUCCESS in 52m 42s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5d32d47282ec4585872e2881550f841a : SUCCESS in 1h 05m 48s","accounts_in_message":[],"_revision_number":1},{"id":"3f36bc5789a398e2ace736e3964dc702d5edaf68","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-08-16 12:48:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2f995e545fc26de3b36488c919f5cc80d8287883","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-08-16 12:57:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c9015fadfd09ae39a46de95ed82067c3e6d2f92f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 12:57:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9ddbf1160d87cbff784acf1dde593e881da146ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 13:41:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70692b645a194eee9a1fc50b6cd5d676\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/9fddd642ad874bc7a57ea13b9ff4fb05 : SUCCESS in 1h 51m 51s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2dec54584a8643bf9ef3de817a3adba0 : SUCCESS in 50m 47s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/99e9086b14c54fae8149c17a6cb6f0a3 : SUCCESS in 2h 16m 36s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/c2d6df76b19c47279c890146785274df : FAILURE in 49m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bff618b2c2c972cd92adbcd55f3da86bf647252e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 14:20:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b806c82b43e480cb6a09b84a83c218f\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/c8075ae45dc142219cc95fee0e3fc23d : SUCCESS in 30m 28s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/58b7cba413574d41be10011e0a2147e7 : SUCCESS in 1h 08m 45s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/d74b1b8090ec498c8b350280eee50eee : SUCCESS in 1h 21m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c447068e0814dafa79b3f1ff13f946a : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e5ade69c398416588888abb43d10064 : SUCCESS in 7m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f683d289ab1d4156b97e8b1c273b90c2 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ee082b7e8e04d3c96f80226eacb6d2a : SUCCESS in 7m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c9e0a26df8746589f28c1f82fde8248 : SUCCESS in 4m 04s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/cd62814364fd4486bd5c06f490938640 : SUCCESS in 19m 52s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c20c02a01c624abd9f76e86fbdc623ec : SUCCESS in 53m 09s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a562fd34d3434e57bbd5ecbf23e56ab0 : SUCCESS in 1h 06m 22s","accounts_in_message":[],"_revision_number":1},{"id":"8309dad60175390bb684aff6138bfca2c1edf27a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 14:20:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7f2a1837f12bb1413fd17c28094b4bb086aca885","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 14:21:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ffdd08c6ffb40099861118155f594c4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adcd9176b6f441118b87692cdf8d31af : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f287af4fa16c4bcf8deb951f7124c575 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a7b00b08c3dc9488a308c2939bb6dd2c531bc879","revisions":{"a7b00b08c3dc9488a308c2939bb6dd2c531bc879":{"kind":"REWORK","_number":1,"created":"2022-08-16 11:11:09.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/45/853245/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/45/853245/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/853245/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/853245/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/853245/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/45/853245/1"}}},"commit":{"parents":[{"commit":"ae42fe3f983d8a561d14ea26b788a98fc51ed745","subject":"Fix container image publish to Docker Hub","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ae42fe3f983d8a561d14ea26b788a98fc51ed745"}]}],"author":{"name":"Daniel Stoye","email":"d.stoye@cms.hu-berlin.de","date":"2022-06-17 14:23:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-08-16 11:11:09.000000000","tz":0},"subject":"Bump skydive version to 0.28","message":"Bump skydive version to 0.28\n\nSkydive versions prior to 0.28.0 panic on newer versions of libc.\nFixed upstream in 0.28, see: https://github.com/skydive-project/skydive/issues/2329\nThis should be backported to at least xena and yoga, as skydive is\ncurrently not working with centos 8 on these releases.\n\nCloses-Bug: #1940862\nChange-Id: I177949b9319a977c9cd9121eb28b710256b72a5a\n(cherry picked from commit 1f1214d4fe496ca015497d7e2fb5aa994c805736)\n(cherry picked from commit 9adffe329cd296edb8ecc0a49f503e28b1ee5623)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a7b00b08c3dc9488a308c2939bb6dd2c531bc879"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a7b00b08c3dc9488a308c2939bb6dd2c531bc879"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
