)]}'
{"id":"openstack%2Fkolla~853806","triplet_id":"openstack%2Fkolla~master~Ia094cd1e29232c47d1fb26e00de569677bc5b8a1","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-08-23 11:55:27.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-08-23 13:17:31.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-08-23 13:17:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia094cd1e29232c47d1fb26e00de569677bc5b8a1","subject":"mariadb: Fix RPM repository URLs","status":"MERGED","created":"2022-08-19 10:57:09.000000000","updated":"2022-08-23 13:18:35.000000000","submitted":"2022-08-23 13:17:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"853806","meta_rev_id":"ae84fd341f99c1079aca9a75f86097881f82b2cc","_number":853806,"virtual_id_number":853806,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-23 13:17:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-08-23 11:55:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-08-22 15:37:39.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-08-23 11:55:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-19 11:40:48.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2022-08-19 12:19:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-19 12:20:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-22 08:29:32.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-08-23 11:55:27.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"efcbbf736918a4d80cb24e78fad36c9a479319da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-08-19 10:57:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"090a4ed59237e4479e86440fe778007d982a3313","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-08-19 11:40:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9686f65f4a7b7b6e9c03b0e29a161302108e5a48","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-19 12:19:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d07a9ed58b847b38e51494dbc8cc409\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a9588729a3dc44e695e7bc2744774991 : SUCCESS in 30m 37s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3a619b1f03bb4fc599ea6d7d3bcea3f1 : SUCCESS in 1h 17m 08s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6fea0b6c70dc4aaab0d3408f15c3fd6d : SUCCESS in 10m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/20da285406934d89aeb9a208b5f29a2c : SUCCESS in 3m 46s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/08aba677e1ed4cafb1584e23be3bb3a8 : SUCCESS in 28m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ccd285c3bfdf2e7e096b13374c5c1fad9c640800","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-19 12:20:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cdc6aac9f50646ee82dee44bb595b50f\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/89d93d8735da4273b16957ea936b634f : SUCCESS in 22m 57s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/cb3eaceaae8843aca9d172c94bc3316e : SUCCESS in 59m 44s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/e0e452d810074af7bcd7db80659db81f : SUCCESS in 1h 17m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c8e2da4e87c4469b7b2b3f655460f1e : SUCCESS in 3m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/08cf607477444e4e9a89310f84bee21c : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/599bc4df8ca94a12a6b434e55ad07832 : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cc236ceb227b4e3aa6da178446acc5ad : SUCCESS in 3m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce3b535a2e94495d8e49955808dfa48e : SUCCESS in 6m 31s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/2b3e2c2225274ee0a2789bc58f475c3b : SUCCESS in 19m 30s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4182bdf0f59544878e9c2a9eef56aaf2 : FAILURE in 15m 20s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/cb8d028d5d4140dd961f2ee671457419 : FAILURE in 17m 31s","accounts_in_message":[],"_revision_number":1},{"id":"a7187731732956818127ad6eba01d32abe1f32ce","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-08-19 13:13:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3b401154d611e21c3f00c3b7c3db443ada1f3731","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-08-22 08:29:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bcd1becd56a36a0bc137e23fb425dc6376e46504","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-08-22 08:30:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6b8cd4cb2e04d3066473be41b63a91150bc0b0af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 09:47:06.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2099721e5af544ffa0f8892357cb7a12\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/76b6648578ba40fd877c6e5aab8808a9 : SUCCESS in 24m 53s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/11893b8cdd394e3eb852cb665a9d3ff0 : SUCCESS in 1h 07m 36s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/67d841bce485448698037db5dd377f16 : SUCCESS in 1h 14m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f975c137b2344a58a54eeebe27600fc : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c8d4fc0a8e9f4678ac0d8139d2f6a654 : SUCCESS in 3m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6cd82e9f4340415a983a6ec90595e2ee : SUCCESS in 3m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e975ed5eb8ae4527878b497ca601812e : SUCCESS in 2m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3585a9df69e44e3bcecec436f176fa6 : SUCCESS in 6m 21s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/185043d7c8694ef3b34cd93344437806 : SUCCESS in 22m 41s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e25ed9f1bb63436c9d96570d82211f6b : FAILURE in 14m 23s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8b3dfb0460cd43e7a6a0ed2ec5b51100 : FAILURE in 12m 44s","accounts_in_message":[],"_revision_number":1},{"id":"d496969b0470ff68524a11a10b5696e0f6f93460","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 10:05:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/423f73a6cc514e7599d5350a724d5613\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2c540ac3730a4695a385332abbbf915a : SUCCESS in 38m 10s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/191bc459885a443eb916f3ec3aa6269b : SUCCESS in 1h 28m 36s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/689b30fd9b7a414cb6fa1c1ffdba1691 : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6abb6444ac1a48a39889fc4f22ca88a1 : SUCCESS in 5m 23s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/eb83a034d2d84d9889674ed1c39ac8bd : SUCCESS in 39m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1120dad91cea8b825226816709e016605b848d8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-08-22 15:37:39.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"785fd7eab2c96b8dd3306c2a293d52af44067263","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 16:54:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1f8074408a0485aa401c675ac58cd9e\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9595fa6049c944489e1b540ff14a83bd : SUCCESS in 25m 39s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5f3e0dfe62524f78813b65ae5cde52aa : SUCCESS in 55m 50s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/3ee27bfe72974e53ad641e03594c52fe : SUCCESS in 1h 13m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a327f3082784946a12ba3c7d1f07603 : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/237cbb7521dd4997a8f2526b93b3ccc2 : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a2b20bac51e94f85be75896dbeda05dc : SUCCESS in 4m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/81b285a07906470ab20214d7d1c34ee6 : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/579b42faed91423baf1ceb807f6f0d9d : SUCCESS in 7m 01s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/93b928760ef9406e99aa4024e65200d5 : SUCCESS in 24m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/30066188948447a1bb96130f780abebc : SUCCESS in 51m 41s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/26c8249b1d934e37a3344aeee5404020 : SUCCESS in 1h 08m 46s","accounts_in_message":[],"_revision_number":2},{"id":"bdaa17460c642b56d8c2ec3e9e5b535b1a0bc6d7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 17:42:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a33322b7d0524a739b177d3ea6993874\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/90dbb57e3e4246b5b96a6d73627af8d8 : SUCCESS in 1h 41m 45s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/dc681cd6634a450db9aafae26ae83a71 : SUCCESS in 2h 02m 10s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7b8887d654c8474584c8bed9a4ae687d : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/43d471b43323465dafc02ceb278638de : SUCCESS in 6m 42s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/16b1a92422724f55ad3a62675d07541d : SUCCESS in 54m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d288d99d97ab92c925d4c8a5d87e5a11c612f396","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-08-23 11:55:27.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"54dfb1a2a13438217ea303bd979bc5443c3c0fd8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 11:55:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"619236a39ba08e0551e8ce22574268987d8000ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 13:17:31.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c09d514595b42d9a66a1087a395ee0b\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4658b7cb1c0147bbb02b3074ae634d57 : SUCCESS in 20m 15s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6dfc9a486cda4a2285bbb84649c4a0a7 : SUCCESS in 1h 02m 34s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/92264804d15146178f9ed07ee7e5c922 : SUCCESS in 1h 15m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53a3aed41f39423a9b5ae905a692c7f6 : SUCCESS in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b50a688a0b8a4d799373281e7598c260 : SUCCESS in 2m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6fa03626981c4344bf7219c08a188cb3 : SUCCESS in 2m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74df997ba5a3482fa083b1fc4ce14724 : SUCCESS in 4m 53s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/abfab23549d04524849ad10d9c5df1ef : SUCCESS in 20m 45s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/989e4893b4e04289b35034e780fc73f9 : SUCCESS in 59m 42s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/dbd2a5e878d349e6a0fa267d995fb2fb : SUCCESS in 1h 09m 23s","accounts_in_message":[],"_revision_number":2},{"id":"036202200640b1d1fb44c95faf62ecab168eb83b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 13:17:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ae84fd341f99c1079aca9a75f86097881f82b2cc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 13:18:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a8b407d05c446c39f135589a9f1d316\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe8dad7348ab470b9042f75ea54ddb0b : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"94fad9f26bc056e9f049ff63fde5ef163eae43dc","revisions":{"bae9d97ba8327f83c3699fe8600d15eddffcef33":{"kind":"REWORK","_number":1,"created":"2022-08-19 10:57:09.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/06/853806/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/853806/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/853806/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/853806/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/853806/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/853806/1"}}},"commit":{"parents":[{"commit":"3e5369db541c60df6d5cddccc20b04f50a3d45ad","subject":"Merge \"Remove Keystone admin endpoint\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e5369db541c60df6d5cddccc20b04f50a3d45ad"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-08-19 10:37:42.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-08-19 10:57:01.000000000","tz":60},"subject":"mariadb: Fix RPM repository URLs","message":"mariadb: Fix RPM repository URLs\n\nThe MariaDB project changed the location of the repositories, and the\nold URLs no longer work.\n\nChange-Id: Ia094cd1e29232c47d1fb26e00de569677bc5b8a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bae9d97ba8327f83c3699fe8600d15eddffcef33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bae9d97ba8327f83c3699fe8600d15eddffcef33"}]},"branch":"refs/heads/master"},"94fad9f26bc056e9f049ff63fde5ef163eae43dc":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-08-22 15:37:39.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/06/853806/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/853806/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/853806/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/853806/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/853806/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/853806/2"}}},"commit":{"parents":[{"commit":"8dc092858196b6ca599f8cd2accc5d9d04f725ac","subject":"Switch to use Zed binary packages from UCA","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8dc092858196b6ca599f8cd2accc5d9d04f725ac"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-08-19 10:37:42.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-08-22 15:37:39.000000000","tz":0},"subject":"mariadb: Fix RPM repository URLs","message":"mariadb: Fix RPM repository URLs\n\nThe MariaDB project changed the location of the repositories, and the\nold URLs no longer work.\n\nChange-Id: Ia094cd1e29232c47d1fb26e00de569677bc5b8a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/94fad9f26bc056e9f049ff63fde5ef163eae43dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/94fad9f26bc056e9f049ff63fde5ef163eae43dc"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
