)]}'
{"id":"openstack%2Fkolla~854347","triplet_id":"openstack%2Fkolla~master~I3452e8904992bfb977dcae4084a0d98aab82e221","project":"openstack/kolla","branch":"master","topic":"prometheus","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-08-30 10:16:58.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2022-08-31 07:18:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3452e8904992bfb977dcae4084a0d98aab82e221","subject":"Bump prometheus services to latest version","status":"MERGED","created":"2022-08-24 08:25:03.000000000","updated":"2022-08-31 07:19:37.000000000","submitted":"2022-08-31 07:18:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"854347-prometheus","meta_rev_id":"3e8f85a213e8bd8ba8e215578a9c6af378a7bdb9","_number":854347,"virtual_id_number":854347,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":34646,"name":"Mathias Fechner","email":"mfechner@wavecon.de","username":"matfechner"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-31 07:18:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2022-08-31 06:00:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":34646,"name":"Mathias Fechner","email":"mfechner@wavecon.de","username":"matfechner"},{"value":1,"date":"2022-08-30 11:19:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2022-08-30 09:56:10.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2022-08-31 06:00:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":34646,"name":"Mathias Fechner","email":"mfechner@wavecon.de","username":"matfechner"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":34646,"name":"Mathias Fechner","email":"mfechner@wavecon.de","username":"matfechner"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":34646,"name":"Mathias Fechner","email":"mfechner@wavecon.de","username":"matfechner"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":34646,"name":"Mathias Fechner","email":"mfechner@wavecon.de","username":"matfechner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-24 08:48:57.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2022-08-24 09:19:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-24 09:50:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-25 16:48:33.000000000","updated_by":{"_account_id":34646,"name":"Mathias Fechner","email":"mfechner@wavecon.de","username":"matfechner"},"reviewer":{"_account_id":34646,"name":"Mathias Fechner","email":"mfechner@wavecon.de","username":"matfechner"},"state":"REVIEWER"},{"updated":"2022-08-30 05:27:17.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2022-08-30 05:36:10.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2022-08-30 09:56:10.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-08-30 11:19:04.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"}],"messages":[{"id":"bcd49da8690c54818678d83902ecb56bd247cd2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-08-24 08:25:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"612d38e29bc3ea0ccfa36658c220bfbb319e3f33","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-08-24 08:48:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5d618c7b9340d9c7dae56641705825adff8e42e4","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 09:19:44.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7cf1b9de1184bc6b54208bd7acf0313\n\n- kolla-build-no-infra-wheels-centos8s https://zuul.opendev.org/t/openstack/build/2bf17d05ce8545f1b4acfdbfa5f390ac : FAILURE in 21m 42s\n- kolla-ansible-centos8s-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/34dc2498740d42f690d31bcbcd45f36a : FAILURE in 15m 07s (non-voting)\n- kolla-build-no-infra-wheels-debian https://zuul.opendev.org/t/openstack/build/7f4308711fa745828a1e0e0be5c5f4f1 : FAILURE in 23m 46s\n- kolla-build-no-infra-wheels-ubuntu https://zuul.opendev.org/t/openstack/build/10238d3caeaf40b0a2c80d671725a5ba : FAILURE in 24m 14s\n- kolla-ansible-ubuntu-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/9d12b324637b4b10a135a4caf04d82cf : FAILURE in 19m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3dcee990bf3eec4a4fbdd7303a78575f1b70dbd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 09:50:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7764f7df147341b2a535c19f7e276a93\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/fd56896ee76344bf957574f9d4d934a4 : FAILURE in 20m 54s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/fcad5133a3a14c77b6c686e80d23290c : SUCCESS in 51m 39s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/b98b880485d24a92bce488eb6496f0bb : SUCCESS in 1h 20m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fb0de0300b440a4a489ef9559302e01 : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/08087806010c425a86c488e66c06d289 : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1cf17c4f57db4e6fabdd0a763410ecff : SUCCESS in 3m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/377b08262fbb40c18c733c2709d53698 : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4b8d7a490b34bb1ba6f07be8554bf2e : SUCCESS in 5m 32s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d21947392cee4a42b5617932f1979a2d : FAILURE in 21m 13s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bd28c0943b2945d2a4d9d963c12b865f : SUCCESS in 56m 56s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/860f87bd14d741b9aeabff3efc3bad8d : SUCCESS in 1h 08m 35s","accounts_in_message":[],"_revision_number":1},{"id":"df9f7b91ca023ca493c3c5e03dd89dab31db49ec","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-08-24 09:54:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ca7bde7c95c73f03e9c8751328bd0e5643e8041f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 10:24:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f916de1415e54579816e5f32e11cf039\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0de63c1af56b49b288964679c1bfbca3 : SUCCESS in 43m 39s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2b40aa0523d74176a19ba351db4b616e : SUCCESS in 1h 57m 04s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f7f415bb2f014de2b4bf5096c9ab060c : SUCCESS in 3m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/598e641c34ac4633935f1e33b7f61eb6 : SUCCESS in 3m 37s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/504961920544445a877db50e80246204 : SUCCESS in 44m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f1e0f89868c7709ca8a5df37ebb81d17873f6214","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-08-24 10:30:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b41819554d8cbd78c9f5b07a4226ff55f0df1242","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-08-25 02:59:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3ac00de988559f783821736cae88276a5ca8be08","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-08-25 02:59:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ed1f1080f45552895dc6b239c63a0da956fccdbe","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 03:41:51.000000000","message":"Patch Set 2:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61fcf194cfda47a181d1cba23318a1f8\n\n- kolla-build-no-infra-wheels-centos8s https://zuul.opendev.org/t/openstack/build/b62f5a0dd9394e92b7dfac6acca125d7 : FAILURE in 18m 53s\n- kolla-ansible-centos8s-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/f5c37d402b0f4788b7a40d883ad8fbab : FAILURE in 14m 41s (non-voting)\n- kolla-build-no-infra-wheels-debian https://zuul.opendev.org/t/openstack/build/23a014481a6a4eb3b9f192ac2d84689a : SUCCESS in 22m 45s\n- kolla-build-no-infra-wheels-ubuntu https://zuul.opendev.org/t/openstack/build/bfd8d3460a7647208ba43f5a0fe5a2ca : SUCCESS in 21m 24s\n- kolla-ansible-ubuntu-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/6a300056c2d745849c76a1f077247249 : FAILURE in 24m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8857fe09972fc7566a50a68fedfab9246a2cdf97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 04:14:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bdc929ab10b8402c9f9a198f1e3e304c\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d9b79a7d448b486ea8f6cc21575d2202 : SUCCESS in 18m 36s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e63a99d3a8a84282a9eabe17ca2d53a5 : FAILURE in 42m 54s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/d9fa3088449c4b1db8730a3baaddff51 : FAILURE in 1h 08m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e96ded5253614813bc73489473e35c89 : FAILURE in 2m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/96c21b0def4f41e3beb6c51611c91658 : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6a831b5c0e6a43a1a26b9e6c487142b6 : SUCCESS in 3m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/14c6d19c4adb413a958b69da9d32e8c8 : SUCCESS in 4m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8c04989493c4b76bc2089e6084d1f49 : SUCCESS in 5m 45s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b2ce192aab1146d48f8551c72a142c54 : SUCCESS in 21m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6c8896774f3c4d46be3044e9c23d0137 : FAILURE in 27m 16s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/71dfa68239db4b4bb3c96efe634b4107 : FAILURE in 1h 07m 34s","accounts_in_message":[],"_revision_number":2},{"id":"936e8a23cc5065de2270e9605169f21e385dbccd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 04:34:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eef5c2d8ee6d4dafb8b9211722bfaff3\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5fcd20fa39de482caabcd5264d0f827b : SUCCESS in 1h 31m 33s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b5862cb6793142429534d62542ed22f1 : FAILURE in 42m 05s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/023921087a484c8ba45c1cbcf7e0699e : SUCCESS in 3m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6b6ba6d7e7a647e597c1f7c8f4cfee87 : SUCCESS in 8m 31s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/72af75a431304e449fba63e041453515 : SUCCESS in 24m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"73fb0a5717bc74eb34ab60ccaf141fd4b52e0492","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-08-25 05:51:16.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b95e63f21585530e9b4aee8def6dadaca706cd8c","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-08-25 05:55:39.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fb970222a4e6a76a6637bdb8d19c2bce0c17a36c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 06:56:26.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1f3b248c86a34c7fb8fa9dcfee66dc17\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d2295515916145cda616e2dc6b953e29 : SUCCESS in 20m 03s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/39ccf3603d9b4b308c84117e9826339d : FAILURE in 31m 52s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/d97a5f43204c406a9791e75d92ac5153 : FAILURE in 58m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8687c1f2afca4fc2ba950dbf0083ea43 : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c6c4bb51bdd4bff90cb961a7ed4eedc : SUCCESS in 3m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aedbb485aa3440bf9f18c8a7a5eebc76 : SUCCESS in 3m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d572b434e0ef4fc49619c31b2a5fd516 : SUCCESS in 3m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0dbd8e4eb4b4ae58c2dfaaba935121d : SUCCESS in 5m 18s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/289542922eb84b21af7ca392fb3e6efb : SUCCESS in 21m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/dc597dfde9a747fa8baad47a0c7b9519 : FAILURE in 28m 30s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d79be62841ce45a39f4416e194cf8160 : FAILURE in 1h 04m 00s","accounts_in_message":[],"_revision_number":3},{"id":"05b13499fc490967f34668b2e7a5c8f4aae4207d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 07:41:40.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09fa87b1628d4d8f96f87d7b4a4c94a0\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/55d468ff2cdb41fda9c382d2f3a51c95 : SUCCESS in 1h 36m 55s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a5f77e9841d54fa4b396254af593f0f1 : FAILURE in 1h 30m 15s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/81259799ae0d41aea602a5f9a527faa3 : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/abcad5c5cb9247069f8523bda354f3b1 : SUCCESS in 7m 44s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/93d00daf39aa4f9a9003b72c418384df : SUCCESS in 1h 47m 17s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"88404e17c1e45350853aa39efdf9581f0ee71511","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-08-25 14:12:00.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"81e110e528db3fd0d82ab866aa3579ff379d452d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-08-25 14:12:53.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"9248d9a4e343b9e712b2292ce15e05e307ed0800","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 15:33:12.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac25e675af174af19496dc1f4cb73437\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f9e2ecbec24444fd8f6232f3b04c56c6 : SUCCESS in 20m 19s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2d5518f337de42e4914fa6bc49030ea3 : SUCCESS in 1h 03m 32s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/c129fc99ce0243daaad87f9a967a97a4 : SUCCESS in 1h 17m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/186fd3a88f224626ab4a50b32272cba6 : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/57dfa411a44149a6a12de32383559aba : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/56d6af4b820642319428e6abe01389be : SUCCESS in 4m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/876a988cb3d34d36bcbf865c8884bbd6 : SUCCESS in 3m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f306cd3a18d84667acf954bf92d6b7ac : SUCCESS in 6m 22s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/adc18f2203b94e129746954fc9c2e1d9 : SUCCESS in 26m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/08fd0b9bd12846c789a89eefbc23460c : SUCCESS in 57m 32s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4ffaac80557d45d49aef0d00e2986b57 : SUCCESS in 1h 05m 42s","accounts_in_message":[],"_revision_number":4},{"id":"7d31ba848bf83cd5fc4a11145bc8588553c4f446","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 15:55:46.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49b1db3c8425427c92910987fa795ae0\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4ce8daf2391b4e49877225412964cf14 : SUCCESS in 42m 41s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/134f1b8e85654138ae23d4450a7fcfbd : SUCCESS in 1h 29m 03s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6cba659fab2242018af927bfb0b04b58 : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5b67fa67ee1f49a3b1ef995566909074 : SUCCESS in 5m 42s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/d2f8796d2bfa487e82caa61f2fda63a3 : SUCCESS in 40m 12s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b93b480b90e30139556463d1c582875dbbbc7416","author":{"_account_id":34646,"name":"Mathias Fechner","email":"mfechner@wavecon.de","username":"matfechner"},"date":"2022-08-25 16:48:33.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"6bf566f5ef2877d358ec2f2c48520c2d3eaa6fba","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-08-30 05:27:17.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ce548e377bcfc1551c8630f4a45ae9ec4c7db07a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-08-30 05:36:10.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"4286e796df94eb272ad100f4208673b953b1bef9","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 05:53:09.000000000","message":"Patch Set 4:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d72469635a734d478b07f949781e20cd\n\n- kolla-build-no-infra-wheels-centos8s https://zuul.opendev.org/t/openstack/build/98f18445f1f74f66a4158442883d2570 : FAILURE in 16m 04s\n- kolla-ansible-centos8s-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/e1efe005dff240a9b13540cc813a3460 : FAILURE in 14m 00s (non-voting)\n- kolla-build-no-infra-wheels-debian https://zuul.opendev.org/t/openstack/build/9c06f66d760447e6bf67355b83d52caf : SUCCESS in 18m 15s\n- kolla-build-no-infra-wheels-ubuntu https://zuul.opendev.org/t/openstack/build/70d0238dce0a47bf889dc15a8176f63f : SUCCESS in 24m 58s\n- kolla-ansible-ubuntu-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/c1454e526ef34d1ea0587f6a2a9085f1 : FAILURE in 24m 15s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0d2a7708acc13e7970bb9d05a54152df3bc11c77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-08-30 08:25:35.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"fbcabb75eaf502f18f49cd9ee73102d3619fc21f","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-08-30 08:27:01.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"337cd4aca4082fe3033c2d3ac61fcf404a9183ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 09:50:32.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45df115bbfa24f0e8e3cf093667ea539\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/82f8938f046e48c8a3a2c38f8623ff59 : SUCCESS in 21m 14s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/68d88eb88fea4f7d8c03bc0193cccb39 : SUCCESS in 1h 02m 28s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/e827b5970bd84ed29eeeabd92275cd75 : SUCCESS in 1h 19m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07e794a08e504c6f8c3d28068b27c897 : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ccbf9bfe5754113bc6945f6267143e3 : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e113cf4e6dc841bda40208a1716f72bb : SUCCESS in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/64436c1634584e5b9b567962ffe147fb : SUCCESS in 3m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca49ca9d67c14996ae5be6b729ab0e8f : SUCCESS in 5m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/250999dcd5974dbea2c7769c2806bd36 : SUCCESS in 4m 00s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d5b5f449fb694e5b89a5ca8556bf2817 : SUCCESS in 20m 00s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3e43b743ef7f4ab380c7c6312ba14bff : SUCCESS in 55m 05s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9c4faa019e144084baa63921a83282a9 : SUCCESS in 1h 12m 45s","accounts_in_message":[],"_revision_number":5},{"id":"0baad2f946f369fe0ede8f4ff740949965463aeb","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-08-30 09:56:10.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"440985bf25df3b21db2be81cc9890db6e1e4b7ff","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-08-30 10:16:58.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"79a833e8caacc1141e1b9712bf5fbc43f675953f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 10:54:56.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32f0659dabab4c2c8750dae001444ad8\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/909c00d0568d423393999d87d0e1b1c6 : SUCCESS in 1h 37m 31s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/296a9c3fe20d46e6a5271649af6dce74 : SUCCESS in 1h 53m 06s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1c808ad4a97a4dbf97f50a65ffc9b9c7 : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b9e072a1110a49c9843d491edae848db : SUCCESS in 7m 17s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/4a05f2fd26d34d3cafe4aa4d110362f9 : SUCCESS in 1h 10m 29s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d0ec5a6107af3b032b7944612100c313889308d3","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2022-08-30 11:19:04.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"d53c09d9050ed741ef54a1bde99429dafd42952f","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-08-30 13:16:10.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"3e96e5256fcce61143791dd271fd1f49b93442b8","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 13:48:35.000000000","message":"Patch Set 5:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63b546f56de9430aa1bbc25016f93a7c\n\n- kolla-build-no-infra-wheels-centos8s https://zuul.opendev.org/t/openstack/build/40acbb21f4f84de9af7b9fdd58b8d9f6 : FAILURE in 16m 59s\n- kolla-ansible-centos8s-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/f11ed849bdcf4915b6a3e69cb2ad3033 : FAILURE in 14m 21s (non-voting)\n- kolla-build-no-infra-wheels-debian https://zuul.opendev.org/t/openstack/build/ad712fccbf114b238ec6ce3d97e382e2 : SUCCESS in 26m 22s\n- kolla-build-no-infra-wheels-ubuntu https://zuul.opendev.org/t/openstack/build/6ef1b375c6b54645a54532bf532b0274 : SUCCESS in 20m 14s\n- kolla-ansible-ubuntu-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/e1f834d8a99f4f5399690bfca22eaeb0 : FAILURE in 22m 09s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e40c88e0ad067fe6326a026b7bf2aeba6883f512","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-08-31 01:21:34.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"63c39af2295b4eede59141d15d34993acc140211","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 02:02:21.000000000","message":"Patch Set 5:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7509fb40c6c846858e10cd8f09e7ad44\n\n- kolla-build-no-infra-wheels-centos8s https://zuul.opendev.org/t/openstack/build/d34e40b2e24944fcb26ce0297f329574 : FAILURE in 16m 22s\n- kolla-ansible-centos8s-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/da143d7388214e7096a13ea1e985135b : FAILURE in 13m 54s (non-voting)\n- kolla-build-no-infra-wheels-debian https://zuul.opendev.org/t/openstack/build/f4d00b80d529458b855e8a28d125671a : SUCCESS in 18m 25s\n- kolla-build-no-infra-wheels-ubuntu https://zuul.opendev.org/t/openstack/build/9fbe47908370471289b760e90dcb5697 : SUCCESS in 22m 54s\n- kolla-ansible-ubuntu-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/c943e5c97d9747cfa7b914d54356ad92 : SUCCESS in 30m 49s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"6d167a0efcda0334b430bc771cbbbfe32e61ec6b","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-08-31 06:00:56.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"26981b37acec8169ca6b82ba3e073ec69a897370","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 06:01:21.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"db7165910b415178885e13b6ab94ca4a6f8c95e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 07:18:25.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5a0cc3e5f5c44bd8703779fc2b077f1\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c914b98f99e84c6faab6ae66bbb57dfa : SUCCESS in 23m 15s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b03dcde96f544f508cd110bf9b987eb0 : SUCCESS in 1h 02m 23s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/547a445af2244f2e84152db8e1038595 : SUCCESS in 1h 11m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0b468390de74c20ae8c11eb9711ddbc : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ff1e13cc0f14ac6987ec7773713d27b : SUCCESS in 3m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/52ee27eb7a5c4209992d8d0cabb9fbbe : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a1a7cb8c789470ca63e587f64fe9bbc : SUCCESS in 5m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f275c6c36bce41b496822f76cff94163 : SUCCESS in 4m 11s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/2b0a21eddffc447ca080e0dc43bbec0b : SUCCESS in 20m 22s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/71b01b4cddf941a683e7c5425ba1a7e0 : SUCCESS in 59m 53s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7545aca8c0c54171b78a1f8a82ffb894 : SUCCESS in 1h 08m 08s","accounts_in_message":[],"_revision_number":5},{"id":"581094df9da599147202f2a9bb45e7d1974fb2bb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 07:18:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"3e8f85a213e8bd8ba8e215578a9c6af378a7bdb9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 07:19:37.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2e6856f392c413aa611d9646f701afa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5662a3028f4a45258bc36df069e0a038 : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/64b4907ebbc540e1a46afc5acabed267 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"26b027f0b3852ad72c14da07832d55e1cc7dd5bd","revisions":{"9081a6ca8449d748f9263501446aaf23972c9a19":{"kind":"REWORK","_number":1,"created":"2022-08-24 08:25:03.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/47/854347/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/854347/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/854347/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/854347/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/854347/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/854347/1"}}},"commit":{"parents":[{"commit":"37bab095b6650c9787a1fcc473ae907f08c5d124","subject":"Merge \"mariadb: Fix RPM repository URLs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/37bab095b6650c9787a1fcc473ae907f08c5d124"}]}],"author":{"name":"wuchunyang","email":"wuchunyang@yovole.com","date":"2022-08-24 08:22:58.000000000","tz":480},"committer":{"name":"wuchunyang","email":"wuchunyang@yovole.com","date":"2022-08-24 08:24:42.000000000","tz":480},"subject":"Bump prometheus services to latest version","message":"Bump prometheus services to latest version\n\nblackbox_exporter -\u003e  0.22.0\nprometheus_cadvisor -\u003e  0.45.0\nelasticsearch_exporter -\u003e 1.5.0\nhaproxy_exporter -\u003e 0.13.0\nprometheus_libvirt_exporter -\u003e 2.3.2\nmemcached_exporter_version -\u003e 0.10.0\nprometheus_msteams -\u003e 1.5.1\nprometheus_mtail -\u003e v3.0.0-rc50\nmysqld_exporter -\u003e 0.14.0\nnode_exporter -\u003e 1.3.1\nprometheus  -\u003e 2.38.0\n\nChange-Id: I3452e8904992bfb977dcae4084a0d98aab82e221\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9081a6ca8449d748f9263501446aaf23972c9a19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9081a6ca8449d748f9263501446aaf23972c9a19"}]},"branch":"refs/heads/master"},"d73e561b58dd7cfc995111f13bf54bb1ec1077e4":{"kind":"REWORK","_number":2,"created":"2022-08-25 02:59:21.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/47/854347/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/854347/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/854347/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/854347/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/854347/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/854347/2"}}},"commit":{"parents":[{"commit":"37bab095b6650c9787a1fcc473ae907f08c5d124","subject":"Merge \"mariadb: Fix RPM repository URLs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/37bab095b6650c9787a1fcc473ae907f08c5d124"}]}],"author":{"name":"wuchunyang","email":"wuchunyang@yovole.com","date":"2022-08-24 08:22:58.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-08-25 02:57:46.000000000","tz":480},"subject":"Bump prometheus services to latest version","message":"Bump prometheus services to latest version\n\nblackbox_exporter -\u003e  0.22.0\nprometheus_cadvisor -\u003e  0.45.0\nelasticsearch_exporter -\u003e 1.5.0\nhaproxy_exporter -\u003e 0.13.0\nprometheus_libvirt_exporter -\u003e 2.3.2\nmemcached_exporter_version -\u003e 0.10.0\nprometheus_msteams -\u003e 1.5.1\nprometheus_mtail -\u003e v3.0.0-rc50\nmysqld_exporter -\u003e 0.14.0\nnode_exporter -\u003e 1.3.1\nprometheus  -\u003e 2.38.0\n\nChange-Id: I3452e8904992bfb977dcae4084a0d98aab82e221\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d73e561b58dd7cfc995111f13bf54bb1ec1077e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d73e561b58dd7cfc995111f13bf54bb1ec1077e4"}]},"branch":"refs/heads/master"},"e6e673251f02170d5ce5462f8a212952a51b1a94":{"kind":"REWORK","_number":3,"created":"2022-08-25 05:51:16.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/47/854347/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/854347/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/854347/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/854347/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/854347/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/854347/3"}}},"commit":{"parents":[{"commit":"37bab095b6650c9787a1fcc473ae907f08c5d124","subject":"Merge \"mariadb: Fix RPM repository URLs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/37bab095b6650c9787a1fcc473ae907f08c5d124"}]}],"author":{"name":"wuchunyang","email":"wuchunyang@yovole.com","date":"2022-08-24 08:22:58.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-08-25 05:50:59.000000000","tz":480},"subject":"Bump prometheus services to latest version","message":"Bump prometheus services to latest version\n\nblackbox_exporter -\u003e  0.22.0\nprometheus_cadvisor -\u003e  0.45.0\nelasticsearch_exporter -\u003e 1.5.0\nhaproxy_exporter -\u003e 0.13.0\nprometheus_libvirt_exporter -\u003e 2.3.2\nmemcached_exporter_version -\u003e 0.10.0\nprometheus_msteams -\u003e 1.5.1\nprometheus_mtail -\u003e v3.0.0-rc50\nmysqld_exporter -\u003e 0.14.0\nnode_exporter -\u003e 1.3.1\nprometheus  -\u003e 2.38.0\n\nChange-Id: I3452e8904992bfb977dcae4084a0d98aab82e221\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e6e673251f02170d5ce5462f8a212952a51b1a94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e6e673251f02170d5ce5462f8a212952a51b1a94"}]},"branch":"refs/heads/master"},"b2f590f0cfc8803a99bb8a2c5287ba109c92516f":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-08-25 14:12:53.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/47/854347/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/854347/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/854347/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/854347/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/854347/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/854347/4"}}},"commit":{"parents":[{"commit":"5a65f36e9b6a41fb130ea89f34809d27b1ae944c","subject":"Add python-binary-memcached to openstack-base","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5a65f36e9b6a41fb130ea89f34809d27b1ae944c"}]}],"author":{"name":"wuchunyang","email":"wuchunyang@yovole.com","date":"2022-08-24 08:22:58.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-08-25 14:12:53.000000000","tz":0},"subject":"Bump prometheus services to latest version","message":"Bump prometheus services to latest version\n\nblackbox_exporter -\u003e  0.22.0\nprometheus_cadvisor -\u003e  0.45.0\nelasticsearch_exporter -\u003e 1.5.0\nhaproxy_exporter -\u003e 0.13.0\nprometheus_libvirt_exporter -\u003e 2.3.2\nmemcached_exporter_version -\u003e 0.10.0\nprometheus_msteams -\u003e 1.5.1\nprometheus_mtail -\u003e v3.0.0-rc50\nmysqld_exporter -\u003e 0.14.0\nnode_exporter -\u003e 1.3.1\nprometheus  -\u003e 2.38.0\n\nChange-Id: I3452e8904992bfb977dcae4084a0d98aab82e221\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b2f590f0cfc8803a99bb8a2c5287ba109c92516f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b2f590f0cfc8803a99bb8a2c5287ba109c92516f"}]},"branch":"refs/heads/master","description":"Rebase"},"26b027f0b3852ad72c14da07832d55e1cc7dd5bd":{"kind":"REWORK","_number":5,"created":"2022-08-30 08:25:35.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/47/854347/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/854347/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/854347/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/854347/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/854347/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/854347/5"}}},"commit":{"parents":[{"commit":"37bab095b6650c9787a1fcc473ae907f08c5d124","subject":"Merge \"mariadb: Fix RPM repository URLs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/37bab095b6650c9787a1fcc473ae907f08c5d124"}]}],"author":{"name":"wuchunyang","email":"wuchunyang@yovole.com","date":"2022-08-24 08:22:58.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-08-30 08:25:19.000000000","tz":480},"subject":"Bump prometheus services to latest version","message":"Bump prometheus services to latest version\n\nblackbox_exporter -\u003e  0.22.0\nprometheus_cadvisor -\u003e  0.45.0\nelasticsearch_exporter -\u003e 1.5.0\nhaproxy_exporter -\u003e 0.13.0\nprometheus_libvirt_exporter -\u003e 2.3.2\nmemcached_exporter_version -\u003e 0.10.0\nprometheus_msteams -\u003e 1.5.1\nprometheus_mtail -\u003e v3.0.0-rc50\nmysqld_exporter -\u003e 0.14.0\nnode_exporter -\u003e 1.3.1\nprometheus  -\u003e 2.38.0\n\nChange-Id: I3452e8904992bfb977dcae4084a0d98aab82e221\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/26b027f0b3852ad72c14da07832d55e1cc7dd5bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/26b027f0b3852ad72c14da07832d55e1cc7dd5bd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
