)]}'
{"id":"openstack%2Fkolla~854546","triplet_id":"openstack%2Fkolla~stable%2Fyoga~Ia094cd1e29232c47d1fb26e00de569677bc5b8a1","project":"openstack/kolla","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-08-30 19:35:04.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-08-30 17:56:39.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-08-30 19:35:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia094cd1e29232c47d1fb26e00de569677bc5b8a1","subject":"mariadb: Fix RPM repository URLs","status":"MERGED","created":"2022-08-26 08:03:38.000000000","updated":"2022-08-30 19:36:12.000000000","submitted":"2022-08-30 19:35:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":4,"has_review_started":true,"submission_id":"854546","cherry_pick_of_change":853806,"cherry_pick_of_patch_set":2,"meta_rev_id":"1a67be78b692b19a098446c9bebd72be466e851d","_number":854546,"virtual_id_number":854546,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-30 19:35:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-08-30 17:56:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-08-30 17:53:40.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-08-30 17:56:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2022-08-30 17:53:53.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-08-30 17:53:49.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-26 08:06:23.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-08-26 09:40:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-30 08:35:35.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2022-08-30 17:56:39.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"cbec3ee0326506e960a85c58f0a66578718252b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-08-26 08:03:38.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b1acfa8ab997fee68e71720c7c0357b64d9afde7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-08-26 08:05:40.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"31d7c1220f99235a68fb1d82fb2cf1a752d0915a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-08-26 08:06:23.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"6fea9bab1f6183f4b6fcec79e82f95101db64569","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 09:40:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ba2450a876d74ab8b0c838e3231c1c84\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/1462790bbba84326bea993bc62ca0eb8 : SUCCESS in 35m 51s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/6e8f534a845a4c5bb40fe2ac1c4500cf : SUCCESS in 46m 30s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/3055021ab71d4d3e85075728f49fe6af : SUCCESS in 1h 17m 36s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/5d82442b9de540c5a617ce11e9e66da3 : SUCCESS in 1h 09m 58s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/0acbf8e49218430ca34e282357097e79 : SUCCESS in 1h 29m 18s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3a7f0e89420945b380b2638efc08e8e8 : SUCCESS in 20m 26s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/7f6700f6b252433cbccc15d4278d0ecb : FAILURE in 20m 16s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/084ed10fd5df4503a512fa4251e8c579 : SUCCESS in 1h 04m 47s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/4ede3be6df374eeebeb175506b8e69cc : SUCCESS in 1h 15m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b623f0831e6e4a76bf91e4c0058fbd9c : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0dd9761133a1418c9aed1bafc0468434 : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7451c77f858344a990fcf25edfb105ea : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/720abd1739ec4ecba294e6514c48ee40 : SUCCESS in 6m 16s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6a02295671bc421799a8aa4f7ded4c00 : SUCCESS in 17m 29s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/458fb1938e554ffb88ea347b666e3f24 : SUCCESS in 22m 12s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/51efe0f6a90a4f07b2c71511ab00fcd1 : FAILURE in 18m 16s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/34c52da98787473895ac1b1b60399419 : FAILURE in 15m 11s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2b39b1e9f2c5422795a6f4b60fbe6493 : FAILURE in 19m 29s","accounts_in_message":[],"_revision_number":2},{"id":"73a4a053ef047d0e85d31f6a6f912eecc3939775","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 09:44:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fdb7bc96c374645b021383fdae43333\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/cad9d0e5874f433ba70543bd9098a1b9 : SUCCESS in 45m 25s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f471440bb771495190588a81f8edfae1 : SUCCESS in 35m 35s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2997f92c4ebf49dbaa497c303b9c3339 : SUCCESS in 1h 28m 57s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/89234e166b1d4dff82fb549e30576afe : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fa2600c585fd49ab8faf75cbe3ac97dc : SUCCESS in 6m 45s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/7b4e94cea05042e7bd9caa0a3bbc0700 : FAILURE in 35m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"790970dd6c3a08b8ca2f704d22d5d3c5868be18e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-08-30 08:35:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"59b9804d85b63a367cefcffb6a9dc975743b96cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 10:20:18.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/55806500dced4d26b7ddffddb65f8439\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/4217c4bd7c18465f90e27d7034ec81dd : SUCCESS in 28m 39s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/4db64971f8884548b22638f0714c38fa : SUCCESS in 33m 50s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/faf334148b8b4da2a30af340bf7ddf59 : SUCCESS in 1h 16m 18s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/afe93b524bc8430aa95b54a210cf4909 : SUCCESS in 1h 14m 59s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/30013fe142874a209258ea89d6198866 : SUCCESS in 1h 30m 31s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ace30994873d4d3e9f954569793a1bdc : SUCCESS in 25m 57s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/746325b2615041feb10dab3416b7faae : FAILURE in 21m 13s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/446307a48e564507a4dd4046d23e7a13 : SUCCESS in 1h 05m 43s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/eace1a0e43fb45fc82995739f4fd0cb6 : SUCCESS in 1h 09m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59e61baf556344e890178d05331efd05 : SUCCESS in 6m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e8e147e6cd249fbad911a77b0dcf83b : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c8a61ff259504c2ea559e5e43664eaa9 : SUCCESS in 6m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e1aba1c5ffa4898ba0d4594d5261e8e : SUCCESS in 6m 09s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/144a0f9b435a465d9802c3822ca2186a : SUCCESS in 20m 31s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/dfb33b0c35a841089af5280a89e366d9 : SUCCESS in 19m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/384e986f56c141d0b33becf4de80ad0e : FAILURE in 16m 57s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ab22189cceed40d2aabe1f2684c9555d : FAILURE in 21m 12s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/06f814d319e646818923fa748d15b0c6 : FAILURE in 19m 51s","accounts_in_message":[],"_revision_number":2},{"id":"05497232b2f84ef44acf06e93ea693c95c005001","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-08-30 10:22:42.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f572c1c0c0e5416bcd3709dddacf3a15d0160e68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 11:54:17.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/18a7bb1baf0f44c08947dc50dd6ed8b2\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/ab7067ca7fd64b7d9a503ba43efdbadd : SUCCESS in 30m 40s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/32ddcdc7774a4b4880f19a890ab1adf6 : SUCCESS in 33m 16s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/becd4b599a004adf9ad9100c40f2c160 : SUCCESS in 1h 05m 34s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/28f62108340040cbb38ec9f7f613927f : SUCCESS in 1h 11m 08s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/d1d890529663479c91f8e6315e49990d : SUCCESS in 1h 29m 38s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5a243bc1ac32482b87fd8ee9e818b59f : SUCCESS in 23m 49s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d56780d4c397479591800e735f7d7bb2 : FAILURE in 15m 53s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/051394d132ac4f5eb5f7c4192bae5e3c : SUCCESS in 1h 03m 02s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/2b6d648b9a5c4ece97c8be1b774af11f : SUCCESS in 1h 09m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cc4dc5c335242c8ab8ef786ee5d784a : FAILURE in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/01523baf100b4ba3b20f1fe77105db29 : SUCCESS in 3m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/99b9c1865c5d4445903db473272e6c63 : SUCCESS in 4m 26s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d536af2d0ff749fca43875b8029a25cd : FAILURE in 2m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ff078aac1c4476a99e7441f121e8898 : SUCCESS in 6m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/77de98c1e79941bca1a0f20419f6e291 : SUCCESS in 4m 19s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/eacc4fe06426444a97fd7bb5ddb01301 : FAILURE in 19m 06s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4f56c9ce3df7423286f4a9f7a370dcf3 : SUCCESS in 23m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8dc3fdba4583439f9d7b37026c130845 : SUCCESS in 47m 46s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a9750e4ff7164fe0aa74c76f420e475f : SUCCESS in 54m 34s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5fc43106e25a45a6af2c10b0f7e94cd5 : SUCCESS in 1h 07m 01s","accounts_in_message":[],"_revision_number":3},{"id":"ce60e78ad299cebde323581b23ff1c217547f6fd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 12:32:22.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2718469089fc42b3bcf92beae41cd0db\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/e9a8ef94805e4976a0825aadea8e497a : SUCCESS in 34m 29s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1aaf9e748be24383953bb0e6f651b74e : SUCCESS in 27m 07s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4c05f52d7f214632a91993b202a88caa : SUCCESS in 1h 17m 03s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/50c238be2bb843e8a76dc6e98b5964c6 : SUCCESS in 7m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5e29f76b46e1421b98c4709ee5f41d91 : SUCCESS in 8m 11s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/513a687009814039b3bad9ec2495f125 : FAILURE in 1h 55m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"30199f00fe961c1875cc6a25c60511292b8e0917","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-08-30 13:28:06.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"53cda487043c4ad8f3d1ca3a21f907d94d2ef276","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-08-30 13:32:34.000000000","message":"Patch Set 4: Code-Review+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"75b719c9cfec86741acbf1804b040e477e9bba8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-08-30 13:40:27.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"3f60fe8c0ca09d4f012e027cd9c636784276545e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-08-30 13:43:28.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d7586373d2aa02cd35746087c529aa51a3c16c01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 15:00:35.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85746dc8d4d54a5cab6d4c23b44845e3\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/34c0507ee4d64edc83954714c5e3a186 : SUCCESS in 28m 39s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/94294cbdf165459585c310a48997f897 : SUCCESS in 29m 27s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/1401ce884dc7434baca9ed8b572fbf2d : SUCCESS in 1h 04m 26s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/a39575e6f99940a7b44a1d3a2679e221 : SUCCESS in 1h 02m 43s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/15972ff78bec4208b229bcb86ed554b8 : SUCCESS in 1h 12m 50s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d18a89785af24ce193b8705583173702 : SUCCESS in 21m 10s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/7645206d5ff5423098cee0098df62fb9 : FAILURE in 15m 13s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ac4ab365b1ff448cbf2c1dceb04238fc : SUCCESS in 1h 02m 29s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/af16e56361dc492eb55f8065f88ceb34 : SUCCESS in 1h 02m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f0e1935f9b740de8146c2acd117ab2f : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e6d7616d2dd849d89617b898aa840afb : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d90a8b2b07c6474cacab2d8e7b5f721c : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87fb84aaf1dd4d378d358f8086b098d1 : SUCCESS in 7m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b0a679ca2f644e5692e79a0fce666dd3 : SUCCESS in 4m 02s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/988f9876ac8c4bc1b9e81253f917d066 : SUCCESS in 16m 07s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/02a3701ece5945a0ac4574bffe879ba5 : SUCCESS in 16m 53s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c898980ce7524d4596d05e182d452da1 : SUCCESS in 48m 42s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9b5d0fe4cafe4aeb8468b9aee1bbdbbe : SUCCESS in 42m 18s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/944af4fa1fcf4bb8b4366fee7d69791b : SUCCESS in 58m 03s","accounts_in_message":[],"_revision_number":5},{"id":"a24292ce38ccc2d3b926df04407295bb1b3fe467","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 15:15:55.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/008ceb3856174c5c9c22da42f1d2df14\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/3e13288899a940f2afbcbd1d6317a3d4 : SUCCESS in 50m 23s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8bc08a19bd53446985dc9b2066e6d5ce : SUCCESS in 42m 59s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b884f64e85ac43d2ad7afaeb019eb4d3 : SUCCESS in 1h 28m 45s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a06660efcae549aaa4bdbe401fecc77b : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2f45d8331ed34da5843bd819085dcf67 : SUCCESS in 4m 29s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/49f8b0fca16b4aabacde0a1a1433dc28 : FAILURE in 41m 16s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"0d80e730864e9a6c8be477260b609d93d3962640","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-08-30 16:06:19.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"a0da37b09e70272cf7ca0c4e3b754109416b9029","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 17:44:21.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5598128658f42838b8dec7c6c1587f1\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/7c1f425ca2ce438cb3ffbe78c9a4bc27 : SUCCESS in 29m 07s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/897d35989a3c43d4a073e6ab7018df5e : SUCCESS in 39m 29s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/a9406df9c4474519b6a2d3e7c2a3f81a : SUCCESS in 1h 20m 47s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/861b24ed78d14a0c811b7452124657bb : SUCCESS in 1h 19m 27s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/285e80cd86464cd6b45dfa51d9c9c93c : SUCCESS in 1h 31m 21s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/278b2ddacab54a6da96ef688fc2736f3 : SUCCESS in 21m 11s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/1840eaa4fe3e4ce882fcaab9e099e393 : FAILURE in 15m 44s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/71ce8b2905084085b468cd6efb532051 : SUCCESS in 1h 04m 35s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/149b134f5cde4952b25168b8ea346881 : SUCCESS in 1h 04m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1981acf46a2041ed83f7402cf4a6a512 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e873648b4a4b418bb50f4c985f0a8973 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3bbc8a18c88a4162ba09832a9e28b573 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbf40fa16410427f9e8665cfd137500d : SUCCESS in 5m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0935989dd7e049c5983e85df6d90bde5 : SUCCESS in 4m 19s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/81b60427dad7461ea934bd9e885d98a8 : SUCCESS in 24m 43s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bd695d84f0ed41019d086d20de09f265 : SUCCESS in 17m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/78a39a0efd7b4658b8d8ccc291b64dac : SUCCESS in 1h 05m 25s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ce94021256134fdb9816c1ca19a4d836 : SUCCESS in 45m 43s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/badffdda6b7b4bdcae26be75f9419d24 : SUCCESS in 1h 14m 02s","accounts_in_message":[],"_revision_number":6},{"id":"58833f1fe5817495908a7c9497bc3994f8291a59","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-08-30 17:53:40.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"9b72b2161462d55f90377ee2e1e840f7ff65053c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-08-30 17:53:49.000000000","message":"Patch Set 6: Backport-Candidate+1 Review-Priority+2","accounts_in_message":[],"_revision_number":6},{"id":"fc29bc60f839f7d6f5aa2ff15c64d47590d215fb","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-08-30 17:53:53.000000000","message":"Patch Set 6: Backport-Candidate+2","accounts_in_message":[],"_revision_number":6},{"id":"430a645d85db15316415ffa06fff99a8a6baf6eb","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-08-30 17:54:08.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"8158a2274dbab72e04be4c4e99a7be48c618f966","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-08-30 17:54:52.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"17942d6c6f3fd4afc828db1c7d64926dea9bd5d2","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-08-30 17:56:39.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"88afab945c34d5e463c59b4e78fe52abc179319d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 17:57:12.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"03167c79715521096b4d24fa722b34018d3d5e7f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 18:54:01.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b500e846b9544b795b319aca48b0cde\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/bc59a70037bd4abf8dbc663660d2ad63 : SUCCESS in 2h 05m 14s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ac90bd397c094b008e101938e0d8a399 : SUCCESS in 2h 11m 12s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/26aacb8aeef44c5084bd3859c0c8e5d7 : SUCCESS in 2h 44m 59s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d07af232ee1242f6baea7d58bc93a292 : SUCCESS in 3m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/916247272ccd4ef4ac0c1fd7f48825f2 : SUCCESS in 6m 55s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/ad39fb4d5d3348b7a68cd3e05e216733 : FAILURE in 26m 20s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"7b866d4d668ace1e444f80de54c69be67a96f753","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 19:35:03.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ced0d204bc974e9ea1ed22b684718a68\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/def2c7361660422494bea49365ef93aa : SUCCESS in 28m 31s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/917399ff85e44926a2e1fcf34eb535fd : SUCCESS in 1h 19m 23s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/c14d9f7faa374f00863594210cf7a598 : SUCCESS in 1h 33m 18s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b6c8a2785a58433e87a5495e3b3cdf29 : SUCCESS in 23m 42s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/82bb0c0cd9864ccea66aeeee1843230c : SUCCESS in 1h 03m 26s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/5e1898addb4a429abd923bf7a6d3c13e : SUCCESS in 1h 15m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38781a9b62cb48849a4bd78eae4a9423 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bbd24d1ac8d643a39a1c6cdcd6874c41 : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e7ab647621e4b21b7aa3d6d7f7e9346 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd3cb0b0578c4d42b04d4f634c9f5278 : SUCCESS in 9m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/49dca3fe9e3e4884883f747ec6bc5859 : SUCCESS in 4m 12s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c9f8915f2a4e4fd5a5d7824722b31139 : SUCCESS in 20m 06s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4456e74d8c3342ee9a39eaa17e294c79 : SUCCESS in 58m 28s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ec6f6579929c4266bfbb77d13db01928 : SUCCESS in 1h 08m 14s","accounts_in_message":[],"_revision_number":6},{"id":"b1dc63704ea4d7777543f146e06488e09293e080","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 19:35:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"1a67be78b692b19a098446c9bebd72be466e851d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 19:36:12.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34cd71632d1e48dca92baf9b50b849cd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2326f5253660454e87138c688076611d : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6b49b48667c94cfca18796098a5d02f8 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"b3e72210851a2a9e3e066a85bbb2341bb9fadba5","revisions":{"08791c00ddb47c7006b2b53a6eacb2a7351ecdd6":{"kind":"REWORK","_number":1,"created":"2022-08-26 08:03:38.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/46/854546/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/854546/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/854546/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/854546/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/854546/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/854546/1"}}},"commit":{"parents":[{"commit":"f0cf958d4e87a02b3698bbe9ca879c8194c2291c","subject":"Fix container image publish to Docker Hub","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f0cf958d4e87a02b3698bbe9ca879c8194c2291c"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-08-19 10:37:42.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-08-26 08:03:38.000000000","tz":0},"subject":"mariadb: Fix RPM repository URLs","message":"mariadb: Fix RPM repository URLs\n\nThe MariaDB project changed the location of the repositories, and the\nold URLs no longer work.\n\nChange-Id: Ia094cd1e29232c47d1fb26e00de569677bc5b8a1\n(cherry picked from commit 94fad9f26bc056e9f049ff63fde5ef163eae43dc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/08791c00ddb47c7006b2b53a6eacb2a7351ecdd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/08791c00ddb47c7006b2b53a6eacb2a7351ecdd6"}]},"branch":"refs/heads/stable/yoga"},"04394a6d6fe79adfba394d635b09fcef733bbdc8":{"kind":"REWORK","_number":2,"created":"2022-08-26 08:05:40.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/46/854546/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/854546/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/854546/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/854546/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/854546/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/854546/2"}}},"commit":{"parents":[{"commit":"f0cf958d4e87a02b3698bbe9ca879c8194c2291c","subject":"Fix container image publish to Docker Hub","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f0cf958d4e87a02b3698bbe9ca879c8194c2291c"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-08-19 10:37:42.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-08-26 08:05:40.000000000","tz":0},"subject":"mariadb: Fix RPM repository URLs","message":"mariadb: Fix RPM repository URLs\n\nThe MariaDB project changed the location of the repositories, and the\nold URLs no longer work.\n\nChange-Id: Ia094cd1e29232c47d1fb26e00de569677bc5b8a1\n(cherry picked from commit 94fad9f26bc056e9f049ff63fde5ef163eae43dc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/04394a6d6fe79adfba394d635b09fcef733bbdc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/04394a6d6fe79adfba394d635b09fcef733bbdc8"}]},"branch":"refs/heads/stable/yoga"},"65ad5810fcce566f3f3d806608bcac9250b80e1c":{"kind":"REWORK","_number":3,"created":"2022-08-30 10:22:42.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/854546/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/854546/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/854546/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/854546/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/854546/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/854546/3"}}},"commit":{"parents":[{"commit":"f0cf958d4e87a02b3698bbe9ca879c8194c2291c","subject":"Fix container image publish to Docker Hub","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f0cf958d4e87a02b3698bbe9ca879c8194c2291c"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-08-19 10:37:42.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-08-30 10:22:39.000000000","tz":120},"subject":"mariadb: Fix RPM repository URLs","message":"mariadb: Fix RPM repository URLs\n\nThe MariaDB project changed the location of the repositories, and the\nold URLs no longer work.\n\nPatch bundled with following one (to unbreak CI):\n\nPin docker to \u003c6\n\nIn I077b7b5acf2c1f9beb6da06d3555e2ebe30831d1 we made Kolla Zed compatible\nwith docker 6 and beyond.\nLet\u0027s pin in stable branches.\n\nChange-Id: Ia094cd1e29232c47d1fb26e00de569677bc5b8a1\n(cherry picked from commit 94fad9f26bc056e9f049ff63fde5ef163eae43dc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/65ad5810fcce566f3f3d806608bcac9250b80e1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/65ad5810fcce566f3f3d806608bcac9250b80e1c"}]},"branch":"refs/heads/stable/yoga"},"d7c7b4bcb73a25ff2bb4aae6212b1258ad045166":{"kind":"REWORK","_number":4,"created":"2022-08-30 13:28:06.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/46/854546/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/854546/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/854546/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/854546/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/854546/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/854546/4"}}},"commit":{"parents":[{"commit":"f0cf958d4e87a02b3698bbe9ca879c8194c2291c","subject":"Fix container image publish to Docker Hub","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f0cf958d4e87a02b3698bbe9ca879c8194c2291c"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-08-19 10:37:42.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-08-30 13:26:32.000000000","tz":120},"subject":"mariadb: Fix RPM repository URLs","message":"mariadb: Fix RPM repository URLs\n\nThe MariaDB project changed the location of the repositories, and the\nold URLs no longer work.\n\nFix added for stable branches: Wrap version detection for docker module\nsince it can fail for docker \u003e\u003d 6.0.\n\nChange-Id: Ia094cd1e29232c47d1fb26e00de569677bc5b8a1\n(cherry picked from commit 94fad9f26bc056e9f049ff63fde5ef163eae43dc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d7c7b4bcb73a25ff2bb4aae6212b1258ad045166"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d7c7b4bcb73a25ff2bb4aae6212b1258ad045166"}]},"branch":"refs/heads/stable/yoga"},"498af9a5e574565e2f22d3d0186f218252b20ef0":{"kind":"REWORK","_number":5,"created":"2022-08-30 13:40:27.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/46/854546/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/854546/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/854546/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/854546/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/854546/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/854546/5"}}},"commit":{"parents":[{"commit":"f0cf958d4e87a02b3698bbe9ca879c8194c2291c","subject":"Fix container image publish to Docker Hub","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f0cf958d4e87a02b3698bbe9ca879c8194c2291c"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-08-19 10:37:42.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-08-30 13:40:15.000000000","tz":120},"subject":"mariadb: Fix RPM repository URLs","message":"mariadb: Fix RPM repository URLs\n\nThe MariaDB project changed the location of the repositories, and the\nold URLs no longer work.\n\nFix added for stable branches: Wrap version detection for docker module\nsince it can fail for docker \u003e\u003d 6.0.\n\nChange-Id: Ia094cd1e29232c47d1fb26e00de569677bc5b8a1\n(cherry picked from commit 94fad9f26bc056e9f049ff63fde5ef163eae43dc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/498af9a5e574565e2f22d3d0186f218252b20ef0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/498af9a5e574565e2f22d3d0186f218252b20ef0"}]},"branch":"refs/heads/stable/yoga"},"b3e72210851a2a9e3e066a85bbb2341bb9fadba5":{"kind":"REWORK","_number":6,"created":"2022-08-30 16:06:19.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/46/854546/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/854546/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/854546/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/854546/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/854546/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/854546/6"}}},"commit":{"parents":[{"commit":"f0cf958d4e87a02b3698bbe9ca879c8194c2291c","subject":"Fix container image publish to Docker Hub","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f0cf958d4e87a02b3698bbe9ca879c8194c2291c"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-08-19 10:37:42.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-08-30 16:05:58.000000000","tz":120},"subject":"mariadb: Fix RPM repository URLs","message":"mariadb: Fix RPM repository URLs\n\nThe MariaDB project changed the location of the repositories, and the\nold URLs no longer work.\n\nFix added for stable branches: Wrap version detection for docker module\nsince it can fail for docker \u003e\u003d 6.0.\n\nChange-Id: Ia094cd1e29232c47d1fb26e00de569677bc5b8a1\n(cherry picked from commit 94fad9f26bc056e9f049ff63fde5ef163eae43dc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b3e72210851a2a9e3e066a85bbb2341bb9fadba5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b3e72210851a2a9e3e066a85bbb2341bb9fadba5"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}}]}],"submit_requirements":[]}
