)]}'
{"id":"openstack%2Fkolla~854744","triplet_id":"openstack%2Fkolla~master~I8882484b32a3a3fbde52f58c1209e8c127ea9931","project":"openstack/kolla","branch":"master","topic":"trove-api","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-01-30 17:10:52.000000000","reason":"Change was submitted"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2023-01-30 15:29:53.000000000","reason":"\u003cGERRIT_ACCOUNT_27339\u003e replied on the change","reason_account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2023-01-30 17:10:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8882484b32a3a3fbde52f58c1209e8c127ea9931","subject":"Switch trove-api to wsgi running under apache.","status":"MERGED","created":"2022-08-26 07:10:15.000000000","updated":"2023-01-30 17:12:42.000000000","submitted":"2023-01-30 17:10:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"854744-trove-api","meta_rev_id":"ccc60abd4d834cd84f8aabc75ceb258ec40dbb14","_number":854744,"virtual_id_number":854744,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-30 17:10:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-01-30 13:34:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2023-01-30 13:34:49.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2023-01-30 15:29:53.000000000","_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}],"CC":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-26 08:07:53.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2022-08-26 08:32:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-02 07:55:51.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2023-01-02 07:57:56.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-01-29 07:32:12.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2023-01-30 07:05:56.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"}],"messages":[{"id":"50dd3c3b22c76b4376495289734727055b84d240","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2022-08-26 07:10:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"59d2c3e006f3018ad6a56ebbffd084f203fdf1d1","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-08-26 08:07:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ed5f6b5e01ac02cef027466caa988862bb65c76b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 08:32:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38cd6db197634bc8b66b108c7a7028d3\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/cadd5caddb0a40a091d2c5d06178af92 : SUCCESS in 23m 02s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6c15ec5f909342fe8439685af246757c : SUCCESS in 1h 04m 15s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/52c92971edae455dbff6011a8360fee7 : SUCCESS in 1h 17m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68460239f3db4f42a89db1fe05ad2809 : SUCCESS in 3m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a2215e75ea704eb4b10535c9541bb0c0 : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ec24fdf89084fb1bc1948b27d7d324e : SUCCESS in 3m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/01466e23958e48d78a15d46d4c60ab5c : SUCCESS in 3m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9332da9ac415473aa2ac2d6f608285fe : SUCCESS in 8m 20s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/67bf6652acd443a29771548782918c9b : SUCCESS in 33m 33s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/50d1e717f6b545bfa7dd67f9e96e1056 : SUCCESS in 55m 50s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/89e9d64a309a484f80118574e4caaf01 : SUCCESS in 1h 01m 23s","accounts_in_message":[],"_revision_number":1},{"id":"066c73f3573a30f05cef7e5ccdd1858ba13ba758","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 09:26:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e45bce0fd53c4d2a9a0823616000e4ea\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d2ac109349c349769a69ed79e63bcb8d : SUCCESS in 22m 56s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/febc473ae4ab4784b6ac2fb2e06e2701 : SUCCESS in 2h 13m 14s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/55a6ac0311414465b684c50893fe2b68 : SUCCESS in 8m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/54078867f3cc45f4a1f01a59489b7f45 : SUCCESS in 9m 06s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/4f715993a32543159c20dfa45d6533a8 : SUCCESS in 1h 40m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8b5ddb33d33c71f51097f9bf6c9a4695c4396bfd","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-02 07:55:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"45f74b69222f5913fe7e5dd6faddde55e538dfe3","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-02 07:57:56.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"00375b8dcc3ef93124e18d916f06667a70aeb5d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-01-05 14:48:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7b895cbf7e1ca875229f1faa1d61885bdeed2602","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-01-05 14:54:15.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"5d36025220f5c603fc6e899340e39f4e983fd73b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 16:19:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c467a0b808bf4ff483d0e6c4d99dd9d1\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/dd271519c02b48d1a054b7e9756be690 : SUCCESS in 34m 28s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/201c02452ea449ea8bb648ca165b1d0d : SUCCESS in 1h 10m 45s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/6bd1d592785846908970c1e812117efa : SUCCESS in 31m 19s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/048d68008bc04371b4f445c4b80b425c : SUCCESS in 1h 06m 16s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/1b8a1d7643414c34b07e4f1785c879c6 : SUCCESS in 1h 19m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/136b92c726a8474cb8534eeef3f81fb1 : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d4504a043d334a7586fb9256eaccd7b4 : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1ca03768dd574725a3cd6c00c3f45f57 : SUCCESS in 3m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/40e8689b4c01474d94bda16ded091dd0 : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19ff5da6eb7e43d9a109d797b0ef480c : SUCCESS in 5m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4489c65d6914462d86b409dcd391c4a3 : SUCCESS in 4m 00s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/07db917f84524ef69fd8b4d7cb00c8af : SUCCESS in 21m 25s\n- kolla-ansible-rocky9-source https://zuul.opendev.org/t/openstack/build/3853bcb001c94e168d7ebe18b7a8f90b : SUCCESS in 1h 11m 58s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e9b9485083c04d48a735f4672b028a68 : SUCCESS in 30m 35s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4597ef9ba2e64568ac72645b04196770 : SUCCESS in 54m 55s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5a342409977349508de7cbef9800c930 : SUCCESS in 1h 14m 16s","accounts_in_message":[],"_revision_number":2},{"id":"4dcb91fc7d1033f6678bf2e0a9528f38eecf47d1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 17:16:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/346c81e25b1842f99af89182d9ab0319\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/8a5078437ce140c8a365b8c16947a78e : SUCCESS in 2h 24m 16s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d329797add364807954b29bf5f9d4aef : SUCCESS in 1h 45m 42s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e1fca755938047b6a0e3b007cf424678 : SUCCESS in 2h 21m 49s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e8a0b4a023354f8eb49495da4a191cfd : SUCCESS in 7m 04s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b3042c9720a24c0dbbc007430dce371e : SUCCESS in 6m 33s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/7dffc5555e694efa91dba864f0235dee : SUCCESS in 1h 49m 48s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/4a002a8ea27546a29c7b992694a37fcd : SUCCESS in 2h 03m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"736ea7bfa45bb21f53242395c1af10aa8757a151","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-20 11:37:51.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"81a1c49bccb7732aa600858eb25c29050390a5b7","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-01-29 07:32:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"443bf733a0f92b6a26442300e3d3089159fc4b71","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2023-01-30 07:05:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ee57c9dcd715f9d86b9ac6dcade4747c4009fb57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-30 07:06:19.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9915e6f04e1b21fec9c4d4afc5f0fac1ab7141b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-30 08:02:42.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f0414f8c630c43db9d6ed581ccc49e75\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e0826b93ef5b42feb1b99396301b000b : FAILURE in 12m 41s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ba077da589a241bb90baa3c141ffa3f8 : SUCCESS in 50m 25s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/2772b640346e42d4a3655f3d4bb6f2ce : SUCCESS in 55m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12a8d894169f4bb1b4211f01eb906147 : SUCCESS in 2m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bfe56b56f7a0439ba48e44f5a429fa49 : SUCCESS in 2m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/359c97546d2e478cbc975815d360bcbf : SUCCESS in 2m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85c5d20c6f0441099d592fa355ba722d : SUCCESS in 4m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a5f20253039941efbe04da82a5ff29bd : SUCCESS in 3m 13s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/7c09ac8d9fc64c568827a5ded10f2e3a : FAILURE in 15m 06s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/1e70e726c6ed49d0a3e5e57b8a7a670a : SUCCESS in 49m 43s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a63b6f621363404ebb184ac5676abcac : FAILURE in 18m 47s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/67ae7a9d67374e3595d73fd10981ed24 : SUCCESS in 42m 55s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/2d4cc412edfe4ffda5d04983c67d3de7 : FAILURE in 16m 24s","accounts_in_message":[],"_revision_number":2},{"id":"7846c08e797abb379f8ef6dab97332ca7e00157f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-01-30 13:34:49.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"b14783b0e9a1294e5dd6427486e25273d130fc30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-30 15:28:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32c8f37d30f640678a1b8b8c6f4e2fdb\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/ef0f35432c354bf89d03716dbbb615c0 : SUCCESS in 24m 52s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/5950faf475f14dc59d7df2d3a595b2ba : SUCCESS in 1h 22m 58s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/95c15f09c6c24ffda5e097a821e2e803 : SUCCESS in 16m 15s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/dede8b57382a48418f86b9eeabe1c1cb : SUCCESS in 1h 19m 54s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/9df64ea36c5e4a5f9d0c2dcbc36b3a1f : SUCCESS in 1h 12m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c0ed3aab85544928bc196ee7aa7c1ec : SUCCESS in 2m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c6ad3206e2844d0c817cd4c11546bdcc : SUCCESS in 2m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/102d3c2305fc42269930ff5ef39e5d6c : SUCCESS in 4m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be1fbfda0674437698f053d992e5bf9c : SUCCESS in 2m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa7e14af516740ed9b659184f5d56c3a : SUCCESS in 4m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e77686127f07475199f07e61d1940f7a : SUCCESS in 3m 14s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/85ba0d2b37524916a0ba881a80e6ab83 : SUCCESS in 23m 47s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/66d752eabe544080abeadb8c429455f2 : SUCCESS in 1h 28m 10s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ee45b54bbf0348d7ac04d7965d4a9256 : SUCCESS in 13m 30s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/fb4399e1fafb40758c415ba2627f774c : SUCCESS in 49m 40s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/14e1c5b16d5a476686815ddd6fc170d9 : SUCCESS in 1h 48m 02s","accounts_in_message":[],"_revision_number":3},{"id":"1e9884849ff140cf78771ed926f4e33910304e41","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2023-01-30 15:29:53.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"31df121d2c8a4d6c51fe731dd7932b33a820aca1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-30 15:30:19.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"05a899e5bddfcec78baa21d87fc9899ad3602d96","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-30 15:53:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c1b69bd01d44c119e56deac693c721f\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/8c864265e92b457d9aaef05f33e408e9 : SUCCESS in 2h 08m 52s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/eebf6eb1b72541198b8f5088490db05a : SUCCESS in 1h 36m 42s\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2245cde8a86d4ec6aab5a398ce62557d : SUCCESS in 2h 15m 48s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c9de545ae520462bafbde972187458a6 : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8154555eb26e47d1b3a067dbe11e5767 : SUCCESS in 4m 41s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/65c3333ad0cc4c89af5ae04cfb67fdac : FAILURE in 2h 02m 37s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/64d5fcb8efb749dc853a4abd7d64c955 : SUCCESS in 12m 18s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2df7e24166e1e0f2442e0f05bde71e1ce5c7e73b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-30 17:10:52.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1cb7183314a84434b1e481f9a897a7d0\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/55cca70d697241c493ee0b9549be33bd : SUCCESS in 16m 44s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/19f270fda525407aa79d49b2515ba5b5 : SUCCESS in 1h 06m 38s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/8eae1ddc7b4f4814a277c1bd9183811a : SUCCESS in 1h 23m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fe3d660535847908dd6189fd6687321 : SUCCESS in 2m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ad669a454234919a1f391a37eb25495 : SUCCESS in 3m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fc8520fd14a84fec89d337d67febcc24 : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db324b2ff81640718feaeb658cd8e30c : SUCCESS in 4m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb906cfafd2841cf87b1129ad4be9bc5 : SUCCESS in 3m 59s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/056912d5a64d4ad0a3e2fe75aea21328 : SUCCESS in 30m 34s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/0d18506eae9c4657a17d86bf920e99f5 : SUCCESS in 1h 04m 41s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ade475a171f7404ab76e2b131d77bd05 : SUCCESS in 33m 15s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/7151f58bf5514695a3b4c34108e036b6 : SUCCESS in 44m 50s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/0afc98309022405d890f5353dd5e392e : SUCCESS in 1h 10m 39s","accounts_in_message":[],"_revision_number":3},{"id":"d1fc5bea74b42909c3f000b3cc0da5ee800a1343","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-30 17:10:52.000000000","message":"Change has been successfully merged\n\n2 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":3},{"id":"ccc60abd4d834cd84f8aabc75ceb258ec40dbb14","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-30 17:12:42.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a36288540645470d95f41f14fb31a1e1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da127a8ee0874295a2f81b235979fc17 : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/776f46a33ecc4b63862dbb57fb43188b : SUCCESS in 38s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1bb082d35a03579d7b762d9691aef86c5b428534","revisions":{"78ee76b79cd6d7e863b66b6686e113c2d1aab417":{"kind":"REWORK","_number":1,"created":"2022-08-26 07:10:15.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/44/854744/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/854744/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/854744/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/854744/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/854744/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/854744/1"}}},"commit":{"parents":[{"commit":"5a65f36e9b6a41fb130ea89f34809d27b1ae944c","subject":"Add python-binary-memcached to openstack-base","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5a65f36e9b6a41fb130ea89f34809d27b1ae944c"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-08-26 07:06:48.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-08-26 07:06:48.000000000","tz":480},"subject":"Switch trove-api to wsgi running under apache.","message":"Switch trove-api to wsgi running under apache.\n\nDrop \u0027USER trove\u0027 from trove-api, trove-api runs under\napache needs run as root\n\nChange-Id: I8882484b32a3a3fbde52f58c1209e8c127ea9931\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/78ee76b79cd6d7e863b66b6686e113c2d1aab417"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/78ee76b79cd6d7e863b66b6686e113c2d1aab417"}]},"branch":"refs/heads/master"},"a7ccbe114c84360b351cbe4ff370b16dc924ecbb":{"kind":"REWORK","_number":2,"created":"2023-01-05 14:48:41.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/44/854744/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/854744/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/854744/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/854744/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/854744/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/854744/2"}}},"commit":{"parents":[{"commit":"5a65f36e9b6a41fb130ea89f34809d27b1ae944c","subject":"Add python-binary-memcached to openstack-base","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5a65f36e9b6a41fb130ea89f34809d27b1ae944c"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-08-26 07:06:48.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2023-01-05 14:44:53.000000000","tz":480},"subject":"Switch trove-api to wsgi running under apache.","message":"Switch trove-api to wsgi running under apache.\n\nDrop \u0027USER trove\u0027 from trove-api, trove-api runs under\napache needs run as root\n\nChange-Id: I8882484b32a3a3fbde52f58c1209e8c127ea9931\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a7ccbe114c84360b351cbe4ff370b16dc924ecbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a7ccbe114c84360b351cbe4ff370b16dc924ecbb"}]},"branch":"refs/heads/master"},"1bb082d35a03579d7b762d9691aef86c5b428534":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-01-30 13:34:49.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/44/854744/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/854744/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/854744/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/854744/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/854744/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/854744/3"}}},"commit":{"parents":[{"commit":"ab70c7618b3b7e025759fe38c8692f66ef420b15","subject":"influxdb: Update gpg key","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ab70c7618b3b7e025759fe38c8692f66ef420b15"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2022-08-26 07:06:48.000000000","tz":480},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-01-30 13:34:49.000000000","tz":0},"subject":"Switch trove-api to wsgi running under apache.","message":"Switch trove-api to wsgi running under apache.\n\nDrop \u0027USER trove\u0027 from trove-api, trove-api runs under\napache needs run as root\n\nChange-Id: I8882484b32a3a3fbde52f58c1209e8c127ea9931\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1bb082d35a03579d7b762d9691aef86c5b428534"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1bb082d35a03579d7b762d9691aef86c5b428534"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
