)]}'
{"id":"openstack%2Fkolla~855566","triplet_id":"openstack%2Fkolla~stable%2Fxena~I66476a2b396e2cbe41e68ac51f57aae1806b2ed8","project":"openstack/kolla","branch":"stable/xena","topic":"bug/1985784-stable/yoga-stable/xena","attention_set":{},"removed_from_attention_set":{"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-09-06 13:59:21.000000000","reason":"\u003cGERRIT_ACCOUNT_30491\u003e replied on the change","reason_account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}}},"hashtags":[],"change_id":"I66476a2b396e2cbe41e68ac51f57aae1806b2ed8","subject":"Fix CVE-2022-38060","status":"MERGED","created":"2022-09-06 12:04:46.000000000","updated":"2022-09-06 16:48:22.000000000","submitted":"2022-09-06 16:47:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"855566-bug/1985784-stable/yoga-stable/xena","cherry_pick_of_change":855565,"cherry_pick_of_patch_set":1,"meta_rev_id":"cafd6320c4d47b2baf1678abe0dc7a2c1fcd514d","_number":855566,"virtual_id_number":855566,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-06 16:47:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2022-09-06 12:15:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2022-09-06 12:05:21.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2022-09-06 12:15:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2022-09-06 12:15:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-06 12:05:21.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2022-09-06 12:15:25.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2022-09-06 13:47:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4a754121f1ffbd6003e7b072887e66efdfd8cf83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-09-06 12:04:46.000000000","message":"Patch Set 1: Cherry Picked from branch stable/yoga.","accounts_in_message":[],"_revision_number":1},{"id":"1dbab575559598a274948c0d4e486f31bc3e86be","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-09-06 12:05:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3a42160433103c6d5066f9546e10f9f4fe55f978","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-09-06 12:15:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1 Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"dc49b7e819ebcec0716079bb0449e4a6040e8cb2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 13:47:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/545d518f625f41259f5546a6eac5bd07\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/acc99afb1ab049b080dc8966371e4aa3 : SUCCESS in 28m 40s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/ac13913c18fe4ce8aacba5fb818b9735 : SUCCESS in 31m 02s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/2d6929eaa555491e9e61a797084058a7 : FAILURE in 59m 25s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/c7ad1affad19409189e6dba4735a0be6 : SUCCESS in 1h 14m 47s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/3292e2e0e22d45299c00d3f99fb520d4 : SUCCESS in 1h 32m 10s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f050701d718b4b80adfabcd285ac0a61 : SUCCESS in 20m 44s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/1764bb76b1a2426195e7933d641088d7 : SUCCESS in 21m 14s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/561ab1affb154772bc56eeb150003ba9 : SUCCESS in 1h 01m 41s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37c5c95cd971424894cfd48957bfec27 : SUCCESS in 7m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7006226563d4060a7ef1eb400d45306 : SUCCESS in 6m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4bd7a1ac3e934bbcb8e8e6b31e5fe995 : SUCCESS in 7m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a9463f59deb44c3d80f0f6f3b5f62a45 : SUCCESS in 6m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a90eb04464c4e959df23aa1879d9dbd : SUCCESS in 8m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21f940f6afbb493b96f7b1073b6d4b07 : SUCCESS in 6m 29s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7a7ab5e1c0164899bfea5b1762871442 : SUCCESS in 19m 12s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3c4001322547485f8a259bc03c3e0dba : SUCCESS in 20m 32s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4093c8de04304b788012e2602abe1e07 : SUCCESS in 57m 43s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d8ea22b50b59444082f0d3885ffc2af6 : SUCCESS in 1h 00m 51s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7d21f9e64ce04f8f9b8be2b9665e01f0 : SUCCESS in 1h 07m 26s","accounts_in_message":[],"_revision_number":1},{"id":"95d834c10833a5e86e7ab4385f2d51c7ee133d98","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-09-06 13:59:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f755d733fa63b68d97520eab38196977ac9f09f4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 14:55:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4e7c288d93c42fa85fd42aa75d2ee8b\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/53f0bc7c654f4e4abedbfa3ed81505b6 : SUCCESS in 1h 16m 36s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b32f3d657da54b41b250f036011e4162 : SUCCESS in 1h 06m 14s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8b7a9da26fcc4c6496163c9e6125dec4 : SUCCESS in 2h 37m 57s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/29e4b12bb5c645c782c8a85aa036c40a : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0887179917ea42249ecf7a921775710d : SUCCESS in 4m 40s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/39ae87e7e0a144ca9c0f214e683be009 : FAILURE in 1h 04m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"99872c3b995a3813f1cdac8f10d26efa8d45c9fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 15:20:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/506870a466214bf1872854da1360a221\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/e5b390da95104a9c88e6c0a44f023e44 : SUCCESS in 28m 52s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/f01a5f648cd740248226b8b410dbba58 : SUCCESS in 22m 54s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/f5315fdf7d5749d182eaf39d9b74eb38 : SUCCESS in 1h 09m 20s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/85e60b700bf84117a78f2a6477206262 : SUCCESS in 1h 02m 13s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/e189351acc1e4cc3b5cd48a700eeeeff : SUCCESS in 1h 16m 44s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5ac75074b8404047b75d3c66b0caee63 : SUCCESS in 15m 01s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f6846ab0a5ff40f8838c5a06bfd6207b : SUCCESS in 18m 08s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0654558a85cc4dc8b43ce103cb561266 : SUCCESS in 52m 58s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31b30b41f88e403eba68b356ee94c2df : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e1aa3d27b0d4b34bc8bae383411167c : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dce18931f35642f7b3fafce0dcd27843 : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9b63b7b051154344b41efefd06ee36cf : SUCCESS in 4m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f59538d6410f4696837f2cd8cc0171f4 : SUCCESS in 6m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/93a331130aa44af28f76be9f8807357f : SUCCESS in 4m 21s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/17afc01fbc0941ccadf50a376d48131e : SUCCESS in 17m 10s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6575a1bf3ccd42afb4a6fedb71b0525f : SUCCESS in 17m 31s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e0835728cf6447b4a63a29324a9c0328 : SUCCESS in 48m 19s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0b0ce62edd2a4802aef408f6a37a9f87 : SUCCESS in 46m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/47274b5ab9604d0dbc48cc3495f56dcb : SUCCESS in 58m 38s","accounts_in_message":[],"_revision_number":1},{"id":"76edc019cd6e3109ca68c67117b5525b1dd93b9e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 15:21:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"197d244947d7f0274784af1fbec45fe3812b4bfd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 16:47:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c4dd442b7f147989fdb3738e769b13f\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/8e80493c044d4919b2cac58c5d94ad66 : SUCCESS in 29m 23s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/9070495496c848778451a6bdc98c540a : SUCCESS in 1h 18m 40s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/1acb6ee380cf4974acf7446bf1d72194 : SUCCESS in 1h 25m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a2769e67ff64b8793ae0ae90c204a27 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a95a76b4f4f4864a6b76bcab40eee00 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c0225710c5ed4d4c84aad9263322a0d6 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91d9915bdfb84df3bc2c873e5145eb6a : SUCCESS in 6m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c4e421bc54948889793f248ef0262fb : SUCCESS in 4m 34s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f2c6c84e70894c57ad0f6124f421b466 : SUCCESS in 21m 28s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0682e4333c78460faefd0f6ed5c0d017 : SUCCESS in 57m 41s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ed9b96d65dbe40aabe05c7fa91fd1251 : SUCCESS in 1h 06m 10s","accounts_in_message":[],"_revision_number":1},{"id":"91b9bcfc4131d284194ea7a14e7c373ed655c2ec","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 16:47:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"cafd6320c4d47b2baf1678abe0dc7a2c1fcd514d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 16:48:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd1c553e9dd4497392bcce6adb132ab3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1140a8392cfd48cfaf63dca67f6986d0 : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c939fb9b3c2c499981ca683a9260113d : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1011fc60c333d9d2e6276ebdd20e952092ef49d8","revisions":{"1011fc60c333d9d2e6276ebdd20e952092ef49d8":{"kind":"REWORK","_number":1,"created":"2022-09-06 12:04:46.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/66/855566/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/66/855566/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/855566/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/855566/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/855566/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/66/855566/1"}}},"commit":{"parents":[{"commit":"bf4341973401a3b6ba3e6a9ac20f0c310b425883","subject":"mariadb: Fix RPM repository URLs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bf4341973401a3b6ba3e6a9ac20f0c310b425883"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-08-29 09:55:59.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-09-06 12:04:46.000000000","tz":0},"subject":"Fix CVE-2022-38060","message":"Fix CVE-2022-38060\n\nCloses-Bug: #1985784\nChange-Id: I66476a2b396e2cbe41e68ac51f57aae1806b2ed8\n(cherry picked from commit 5b1da017988c987fc68b55d1f45b5d2676474ce1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1011fc60c333d9d2e6276ebdd20e952092ef49d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1011fc60c333d9d2e6276ebdd20e952092ef49d8"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[]}
