)]}'
{"id":"openstack%2Fkolla~855567","triplet_id":"openstack%2Fkolla~stable%2Fwallaby~I66476a2b396e2cbe41e68ac51f57aae1806b2ed8","project":"openstack/kolla","branch":"stable/wallaby","topic":"bug/1985784-stable/yoga-stable/xena-stable/wallaby","attention_set":{},"removed_from_attention_set":{"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-09-06 15:16:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I66476a2b396e2cbe41e68ac51f57aae1806b2ed8","subject":"Fix CVE-2022-38060","status":"MERGED","created":"2022-09-06 12:04:54.000000000","updated":"2022-09-06 15:18:07.000000000","submitted":"2022-09-06 15:16:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"855567-bug/1985784-stable/yoga-stable/xena-stable/wallaby","cherry_pick_of_change":855566,"cherry_pick_of_patch_set":1,"meta_rev_id":"c72964da3dbdf8a7645d0c4b4153d33f8bb642fd","_number":855567,"virtual_id_number":855567,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-06 15:16:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2022-09-06 12:15:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2022-09-06 12:05:16.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2022-09-06 12:15:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2022-09-06 12:15:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-06 12:05:16.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2022-09-06 12:15:41.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2022-09-06 13:30:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"304c7e4cab3df04f7d1f2dbe911ffb857ab8f741","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-09-06 12:04:54.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"babd866af4cada5922f320f8e48d754f049d32fc","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-09-06 12:05:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"512ac585859d7fb23b94e70e73db3640bf6aa1bd","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-09-06 12:15:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1 Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"892264a4cf687b0df38cf0375485af289e895524","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 13:30:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8fd0885048d543d9bc0ee23c761bb8d2\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/a9709b8d9d6841d2b059906e6ddfc495 : SUCCESS in 27m 15s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/f7f4696d988a432fbb15b490e857d3da : SUCCESS in 32m 42s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/21d4da8dc5674bb98437747347bc468a : SUCCESS in 1h 08m 24s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/17bdad0e43824aa8ab947d4b78f0ee94 : SUCCESS in 1h 10m 29s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/2d2c554ddd25489e9d31e6d93481dc0d : SUCCESS in 1h 19m 39s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/fb438dc5d12643fd9bab9c7d8bee2191 : SUCCESS in 21m 40s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/8447c5e41a734b9cbb49b4d7a4568f8e : SUCCESS in 18m 07s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/071700021159450d8bcf45b9cefee032 : SUCCESS in 1h 00m 07s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd00f95e66724fa2972dd084dc42fa33 : SUCCESS in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/004170b86f4e4e35a0574e248c6f12c4 : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7442230af429486a832501b0a8aa12d6 : SUCCESS in 3m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9847b4cd775b4fffac9663765eafcebb : SUCCESS in 3m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7b8de67d2874b4bb6b492d98f4b5de0 : SUCCESS in 4m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3aa168be8aa94d54a9f8b782d619f193 : SUCCESS in 3m 05s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/09129d5fb255437ca52a104d9a1a952c : SUCCESS in 16m 54s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bf555d22bc2249b894b5489ad649db9b : SUCCESS in 17m 54s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/dbb25a342fe34e9cba2e6e656652091c : SUCCESS in 52m 45s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7851d7ad89f4472db70f471548c7ab52 : SUCCESS in 48m 06s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0ff907f8efdf47bdabb8a049994a741c : SUCCESS in 56m 42s","accounts_in_message":[],"_revision_number":1},{"id":"23bea98ba7355367c7e3214a0f41ed29b1eaadde","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 13:30:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"58111a36575b8c4d4cf6111af0c9f2fecbfc869b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 13:51:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9ed8ace9ae4447293f840d570cc196c\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/3b334ffd16e14582a35e7e1d4eb8a694 : SUCCESS in 1h 27m 37s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/711193144cd44739990d83972609972c : SUCCESS in 1h 07m 43s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1fe342b3eaff4314b909e14c98308cac : SUCCESS in 1h 45m 06s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/c55c483045974990b39a548703fb773c : FAILURE in 1h 05m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9414f1811edfb80f03adb2043f9b9718c1e1b464","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 15:16:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aeb6f7e74f054c0a974db06f2efc8e6c\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/b1fb180c04654315b38bd02096382f08 : SUCCESS in 32m 21s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/38b5619c312e48bca456b268aa0aaf0d : SUCCESS in 1h 09m 37s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/d87a619b49954026b38150f6dbfbb9d3 : SUCCESS in 1h 40m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/facb4ce1e7864a96af63fb08a5111d40 : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3808978696424ca1818090079822e745 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01bc598ee7bb4af7a83b3f755c2f1791 : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62696b557f7649428722ebe47c52bd59 : SUCCESS in 6m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b7234f1add654798ae8c7f0587543c6e : SUCCESS in 4m 00s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/63b1daafee384448bc53639a8d0756ab : SUCCESS in 21m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/fdec4836d0d549b5a2c57aca643cf13c : SUCCESS in 1h 00m 37s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/54138c3b5755442094980c48580398fb : SUCCESS in 1h 04m 48s","accounts_in_message":[],"_revision_number":1},{"id":"ee5be9c705cbf6556d6b87b8f8227293781f3899","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 15:16:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c72964da3dbdf8a7645d0c4b4153d33f8bb642fd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 15:18:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/916f32b401e540edb328296df03999a2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2cadc79999b4d758e73d2ac68a36ed0 : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e67b12b41fc43b6971ef6ec77510a48 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5056b65bffd9ae180693800dfbf4fecbd6b20960","revisions":{"5056b65bffd9ae180693800dfbf4fecbd6b20960":{"kind":"REWORK","_number":1,"created":"2022-09-06 12:04:54.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/67/855567/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/67/855567/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/855567/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/855567/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/67/855567/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/67/855567/1"}}},"commit":{"parents":[{"commit":"b64e71746c1a499b90ac22b290e286b201da532e","subject":"mariadb: Fix repository URLs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b64e71746c1a499b90ac22b290e286b201da532e"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-08-29 09:55:59.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-09-06 12:04:54.000000000","tz":0},"subject":"Fix CVE-2022-38060","message":"Fix CVE-2022-38060\n\nCloses-Bug: #1985784\nChange-Id: I66476a2b396e2cbe41e68ac51f57aae1806b2ed8\n(cherry picked from commit 5b1da017988c987fc68b55d1f45b5d2676474ce1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5056b65bffd9ae180693800dfbf4fecbd6b20960"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5056b65bffd9ae180693800dfbf4fecbd6b20960"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[]}
