)]}'
{"id":"openstack%2Fkolla~857709","triplet_id":"openstack%2Fkolla~master~I97d9c8261351b8fd88f088c1463cc508e07792e7","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-09-23 15:35:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I97d9c8261351b8fd88f088c1463cc508e07792e7","subject":"Python 3.8+ has better shutil.copytree()","status":"MERGED","created":"2022-09-14 13:52:36.000000000","updated":"2022-09-23 15:36:59.000000000","submitted":"2022-09-23 15:35:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"857709","meta_rev_id":"2d2b0267b909c01d66fa79d66c7953575407162c","_number":857709,"virtual_id_number":857709,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-23 15:35:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":2,"date":"2022-09-14 14:12:10.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2022-09-23 14:01:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":1,"date":"2022-09-23 14:01:58.000000000","_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-14 14:12:10.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-09-14 14:22:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-14 15:04:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-23 14:01:58.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"}],"messages":[{"id":"e9caec6671a680044f76a94f96a4edebe26930dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-09-14 13:52:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"70f1866df19f57d43acc883bf631019813c329fe","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-09-14 14:12:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"30fe465b1b639ee800974563bb2466e20b24a116","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 14:22:51.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e88079fc42b1442c984395f3c1153d6d\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/515a81a4180d47b48f351fa921be976c : FAILURE in 6m 05s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6421dd8a9ed642a38555b7b93fa60174 : FAILURE in 23m 14s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/05afba1ae3de435b9e5e7ad5cf3fc9e3 : SUCCESS in 7m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a32a7e62f7a74cfd9cd578e9aeb53d91 : SUCCESS in 3m 39s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/a095ee5043544050ae44a3c017592c84 : SUCCESS in 23m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"938218ef6fa5f95784fc06e2c9c045a22ad40905","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 15:04:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a2a5cd4c1484d79b95b73a4e809cd52\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/0dd888bb590741a5b15dc1ffef5fc117 : FAILURE in 5m 21s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9f6062c33529485abfbb557c5c571395 : FAILURE in 15m 14s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/d78ad69d05be42d48eaaf12506c39660 : FAILURE in 13m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d062968a31fe477089b1db2fb475f557 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e93dd791973840d190e4389cd2f7aa6b : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7652093e323340f89982b8cbedef0118 : SUCCESS in 4m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dde375c37f6f4c0193b95f87d69adc55 : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/447c0186a4884454be7aeb902b113a10 : SUCCESS in 6m 32s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/1aa980b8679f4f0fbb52fa87c6aefe57 : SUCCESS in 23m 31s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a5de50e5015c4760835faf400d02e4e3 : SUCCESS in 58m 12s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3d4e07fcca6e4e1c8ba32d46fd30e8ba : SUCCESS in 1h 10m 06s","accounts_in_message":[],"_revision_number":1},{"id":"c3c8ca74d53c4a2db6753d9a76d057562c14bd17","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-09-22 10:31:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"710d390bd62cd6012a59bd42bd923c15192f64df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-22 12:03:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba0e97373b254cb9874a982fe913c6e7\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f16dff2c4ba34210a538756ef6ce6119 : SUCCESS in 26m 05s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9efd88afe9344c3fa3617c6572e400ce : SUCCESS in 1h 21m 40s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/0cf2416eb9044b4c9e0aae2da8035e81 : SUCCESS in 1h 26m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d12893a8267140b19a8e86c4ff7c7b2f : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/689ab6e5acfb456f83629c3774255ae4 : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ce7c18b4ba6c453b86892a895a7d751a : SUCCESS in 3m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5969c1520bf34afba86a166e332e7cdb : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f52554dc926495c83e053c0a6397d19 : SUCCESS in 5m 44s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/95ec6bad18a5401fbe91aa696258d890 : SUCCESS in 27m 28s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4466132b80374ebab2027aa13ae57a3b : SUCCESS in 1h 13m 43s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/17c9d22df23a4b9ebaf5d7d6f9c0c355 : SUCCESS in 1h 23m 00s","accounts_in_message":[],"_revision_number":1},{"id":"c307cc3dd94b27be0dec18b376c95241d8fa4988","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-22 12:49:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/448b3467803d42c6b4d1818dfb1c0e33\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2c86354815174b6ab90348d33b1df842 : SUCCESS in 27m 56s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/df7b0a98e44042068963ee2f126381bc : SUCCESS in 2h 14m 59s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/aeebcc5b269f42d7bf088034d8c5d17f : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fe65f1704bc5465da84dc74a05785f21 : SUCCESS in 10m 00s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/4411fface05246fa8a85982bf9dd04e2 : SUCCESS in 25m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4c360666071737f8aa7e9b31ef8dc520a6e2bfbb","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-09-23 14:01:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5337f9a70545314be021c31a1ec30023b3ecb826","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-23 14:02:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b4a6d73b2f5dafc18cd6c9cbdc2dbbb4c186a0ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-23 15:35:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e9d0ceb6df246a1bf3bff39f9579aaa\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/afe2e40e988f4d06bd422e936cf82b4e : SUCCESS in 25m 25s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e5b1cf3a35cf43bf83491f4d1f34ba63 : SUCCESS in 1h 06m 57s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/9c2288855e624a578cd9f4817dbf7162 : SUCCESS in 1h 27m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8b3b8cc65db462088d169ea4da2d5d9 : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eaa58d73e5044e2e9833d31f7969c618 : SUCCESS in 2m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0fd42552d1b44b18a3a5a519c1c0c5a1 : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7779bc8136344289e4f3fe659b5f99b : SUCCESS in 5m 26s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e1c16c5f6fdb49d190005b192f4cf98b : SUCCESS in 23m 35s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bec61a95f19b48338fc4be5f4781d007 : SUCCESS in 56m 20s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e15421a291074322a8cb3ef548510a13 : SUCCESS in 1h 06m 02s","accounts_in_message":[],"_revision_number":1},{"id":"d2df424a310530e40e45c3ed5a966ac20c74d418","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-23 15:35:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2d2b0267b909c01d66fa79d66c7953575407162c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-23 15:36:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d06969636a7448aab45e72ea5182fa8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a25540e25bf4e5cbb0d29beed5b5243 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0e3a55252b04e8e81db7b0f5636fb6624824a5cb","revisions":{"0e3a55252b04e8e81db7b0f5636fb6624824a5cb":{"kind":"REWORK","_number":1,"created":"2022-09-14 13:52:36.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/09/857709/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/09/857709/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/857709/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/857709/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/857709/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/09/857709/1"}}},"commit":{"parents":[{"commit":"0d4e8e6c11947e07953121d6009809db69f4965b","subject":"Merge \"Drop support for python 3.6 and 3.7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0d4e8e6c11947e07953121d6009809db69f4965b"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-09-14 13:49:48.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-09-14 13:52:28.000000000","tz":120},"subject":"Python 3.8+ has better shutil.copytree()","message":"Python 3.8+ has better shutil.copytree()\n\ndoc says:\n\n\u003e If dirs_exist_ok is false (the default) and dst already exists, a\n\u003e FileExistsError is raised. If dirs_exist_ok is true, the copying\n\u003e operation will continue if it encounters existing directories, and files\n\u003e within the dst tree will be overwritten by corresponding files from the\n\u003e src tree.\n\nSo we do not have to copy dirs/files one by one and tell\nshutil.copytree() to handle everything.\n\nChange-Id: I97d9c8261351b8fd88f088c1463cc508e07792e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0e3a55252b04e8e81db7b0f5636fb6624824a5cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0e3a55252b04e8e81db7b0f5636fb6624824a5cb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
