)]}'
{"id":"openstack%2Fkolla~857883","triplet_id":"openstack%2Fkolla~master~Ifb7e0f1632805807851720873a70179218bdf372","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-09-28 13:25:13.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-09-26 18:14:57.000000000","reason":"\u003cGERRIT_ACCOUNT_30491\u003e replied on the change","reason_account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}}},"hashtags":[],"change_id":"Ifb7e0f1632805807851720873a70179218bdf372","subject":"nova-libvirt: provide TPM emulation","status":"MERGED","created":"2022-09-15 14:17:17.000000000","updated":"2022-09-28 13:26:21.000000000","submitted":"2022-09-28 13:25:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"857883","meta_rev_id":"b7057dfb9261199da8ec4525a3aa791a91b4616e","_number":857883,"virtual_id_number":857883,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-28 13:25:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-09-28 12:06:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-09-27 10:02:53.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-09-28 12:06:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-15 15:23:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-15 15:45:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-15 15:51:20.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-09-15 17:00:11.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2022-09-26 18:14:57.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"2c99048f618e01c7045b547b9ebcc9c030bda77f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-09-15 14:17:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"43060c07a0278804a19bcbd15ff11d25d4cdc6c8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-15 15:23:19.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/068d00fd036b412eae7750c959d9cee3\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/9acf4e0d30954139aba9dfba650a47a9 : FAILURE in 29m 39s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/bef76f5a8b33471e86500f65215260a7 : FAILURE in 1h 04m 05s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7ab6f3ef86874c83a5c3e3ea3b3a9357 : FAILURE in 32m 41s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c28992055aa843b5bf4d501a98fc1cbe : SUCCESS in 3m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bb46f375c21d466b9541918ce0733453 : SUCCESS in 4m 21s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/2605496764a04b0db004b73f401d61b2 : SUCCESS in 28m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c5da84359cd01df5c9fcab67b06e6445a9b212f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-15 15:45:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14b5a6d6c76341ec94bf55374f861940\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/e31c644f18344971a788078870d9b516 : SUCCESS in 27m 23s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/c797f198d35040f0bbe909e66d44b257 : SUCCESS in 1h 20m 44s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/45b448ed51374c38b8c1e4e7843a0f6d : SUCCESS in 26m 26s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c859a935bfec4954b2bff208b81f909e : SUCCESS in 1h 06m 45s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/ed5693a38c52419fbbb10370cb39b95b : SUCCESS in 1h 21m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9488d688fb34254b86d4a5cf2906b6e : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b6bfca86f454c13bb711acc0e06d152 : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/38ee8010309a404aab980a717de5b7f2 : SUCCESS in 3m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dc7f041db1ad4aaab1c91130303d2852 : SUCCESS in 3m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a4c93722e7a435fb8d7093e8b8a31ef : SUCCESS in 6m 07s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a7f462834bc846c4a200577c88a1e010 : SUCCESS in 25m 10s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/242bf39fad024f7b9a1498a18ba0de62 : SUCCESS in 1h 04m 55s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7cab9d8601714144b7e7ab64024e09e7 : SUCCESS in 1h 11m 45s","accounts_in_message":[],"_revision_number":1},{"id":"690425bd408fd913346e5e485e89abb2c8de3e08","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-15 15:51:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ae7c241a942cd4339ddea8b0ec0dc73dc80619fd","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-09-15 17:00:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7d5471bf3e4c1ed56b3db1201fed8e49057dc4cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-09-26 13:26:29.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"f6305ca914cf7d1ce60136bd5d9bf23713c4a34e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-09-26 13:26:44.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"b09a091733572c8ab16e8ea3faceedb0ca0f99e1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-26 13:27:05.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla for 857883,2","accounts_in_message":[],"_revision_number":2},{"id":"2f7db838c1625992d3cb4b098c5f3f078e91da4c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-09-26 13:29:04.000000000","message":"Topic zed-add-centos-stream-9-patch54 removed","accounts_in_message":[],"_revision_number":3},{"id":"8c789a866b28e9768f1a56cb0ce37e2185e2abc5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-26 14:14:33.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/559efbe77bb44d57a741ea53be913b8a\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/255398127b34476aba18098186889b9e : FAILURE in 30m 00s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/386b342a4b9e48bd936cfe3bd765c845 : FAILURE in 27m 45s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e10864dbda6b401eb03e3a0ca7b8cac5 : FAILURE in 33m 15s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1645a40d47184752b51b55f4ba70d48d : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/87e3f584f20040b1a5676697b1486785 : SUCCESS in 4m 44s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/971bd0f5e3e542cf9180e8a17485f530 : SUCCESS in 32m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"dbc750f73db1a8e46342b456be12309cbd414ccb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-26 14:49:31.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8fb97736fb8f428ca6f76f320c6d43f1\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/f44418da21484b29b0a1de6028be810a : FAILURE in 48m 43s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/e075bf07143840bf82b55db07c89c295 : SUCCESS in 1h 21m 14s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e694b9ce7411486098bab9241c57f1f5 : FAILURE in 16m 24s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a8300f37c25244028bbeefe39275635f : FAILURE in 18m 19s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/55aed1c4b3f443c89176419f96330328 : FAILURE in 25m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b681abcc33754d8fa0be027f053920ae : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c0319eeed66465eb7ef6f040ad7bf44 : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dd7b28d9b95545d1a0d55428585eeaeb : SUCCESS in 5m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b0ae35a3af684efd87b038e70761092a : SUCCESS in 3m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6906e2f0c9b34418b44ab57a8b2eb218 : SUCCESS in 6m 22s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/1f512ad25d1344fd9902edcbda633119 : SUCCESS in 22m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/370e73da1e0048be9ff9ec4da6fdebe8 : SUCCESS in 56m 26s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2f85f11c29ce4477ada717db7df816b0 : SUCCESS in 1h 08m 44s","accounts_in_message":[],"_revision_number":3},{"id":"97311e67de55c4cda63c7547b33e6378611e2eb3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-09-26 17:51:58.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"552f302251a671e8482dc88ccf322fd52c2b1c23","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-09-26 18:13:55.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"bb3f2b2cde67f13041c70b288e40ffe1c6f1430e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-09-26 18:14:57.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"026f4ec1012982b309801461046dc13020427a3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-09-26 18:43:43.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0d858100a545bccfe88f01cf2b296d6812495d39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-26 20:01:22.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ce39249e7fa74c7baf7d83b3661aa202\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/e57699c9e54f4646b430a84f55a6911d : SUCCESS in 26m 21s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/628730612dfb4a49b9cd587e5912e92c : SUCCESS in 1h 14m 45s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b3935bb79e484b248a18fed7f9b865a9 : FAILURE in 23m 43s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/beac15a444214d149d573786d4a0b231 : FAILURE in 25m 04s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/23f33c199f0e43228b81dee84a1aacf7 : FAILURE in 20m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7dd735ed82094cfb89338985f7f25714 : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a0eed363cb5149549d0de1da4e4ca95c : SUCCESS in 2m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f9c5b0c857494a41b61a421b3624791d : SUCCESS in 6m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0ea1bc01a3dd46dcbf0d2cd8ab47084d : SUCCESS in 2m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff2f483c2a39464cbae42226597f208d : SUCCESS in 8m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c64d29b5385d4f408c73be4cb4fd4845 : SUCCESS in 3m 09s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/6e394415aa5645c083952dbb9e9e0d78 : SUCCESS in 23m 13s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/44ca9857928a4373b5e813b21788bb56 : SUCCESS in 52m 28s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/24a0abc106614cc1bfef7304ab9067c5 : SUCCESS in 1h 02m 13s","accounts_in_message":[],"_revision_number":4},{"id":"9c15dc4cec2c7dbf39753fb97cb5b59509faf72c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-26 20:14:58.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/205cc6facc1b4ce6b792c9af6e986ba4\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/b433d66e291b477fbb6f2bea422a5ec4 : FAILURE in 33m 47s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/45751746c5174410b9d6be7c49344d6c : FAILURE in 1h 28m 54s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/bab11d83bd8f4a19bc12672963f0e855 : FAILURE in 30m 09s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e3741eea0d0c43f9b7a5cea273ad716e : SUCCESS in 7m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3dbaea26430c46beb3155f154ec5279d : SUCCESS in 3m 39s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/db90aa6833224f96a043617f090680cd : SUCCESS in 1h 27m 28s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"57c8b2c4fd8c7d16f7bae3820c7ac5b946ec592e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-09-27 10:02:23.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"1a4b1e12f1fdd2d88781e8be6f6a5f34a1374774","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-09-27 10:02:53.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"cab771cd5e753d5c9850331baa1be62b7b280086","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-09-27 10:07:42.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"751a685dc2f3d62166b2603bdbb409b9856f30c4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-27 11:14:15.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26763a0de9f345d89483d765d52e4cfb\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/4ce8ca303de1491bb52e0fcb95185b62 : FAILURE in 1h 09m 28s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/70dd0e535426416dafc5124f428897c1 : SUCCESS in 3m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/664e45b35a1e46dcbe31d71f400bba06 : SUCCESS in 7m 14s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/2e9f386ed1714702b7a7d46894335d7c : SUCCESS in 25m 07s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"efa5e4d40ca7953ccdb38264b9d9a7273970216d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-27 11:17:19.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3092295a7ae4102bd8e3c91ea2f74d2\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/f66df81f630144a6b85efc24ffe68186 : SUCCESS in 23m 06s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/df3eb36322104271abc9fdeeacbf6ec9 : SUCCESS in 1h 00m 48s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/226b2dc9311e4bccb2677c1376d79228 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/559caa5c44654b76bd8e520ab017464b : SUCCESS in 3m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bda44afed4ae48cfa407ede22f2e1481 : SUCCESS in 5m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ff7fd888d39f4102957022d1ac8f4616 : SUCCESS in 3m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5aa168a85324fe38f1d4bc3458d0a5c : SUCCESS in 5m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d83ddfc5c7584617bb70cb0c2434c7b3 : SUCCESS in 3m 40s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a47e700ccc6f45ff923c23efb5693164 : SUCCESS in 23m 40s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/17b117fbe74b4cb89cdd04196936e2d6 : SUCCESS in 52m 21s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2a0127e865d0466aa6fe5e912769eaad : SUCCESS in 1h 10m 09s","accounts_in_message":[],"_revision_number":5},{"id":"d7e30c0aedbd0fda2e939db242a7a48d1bfcba81","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-28 12:06:27.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"a6cd4305ca9f1de55dc72f0e6dd026f916bd91ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-28 12:06:52.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"ab8923fe02ee002044a870b72728c8fd2309728b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-28 13:25:13.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40c0412c548e4894b0ea1be2d89c9426\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/f8529ff2e9b4439e8ac3d7cd8a1a2ad3 : SUCCESS in 24m 10s (non-voting)\n- kolla-ansible-centos9s-source https://zuul.opendev.org/t/openstack/build/eb1d2baf4a584814bca1de3d45f82e37 : SUCCESS in 1h 16m 38s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d9933eeb408431ebe7870dff49b4298 : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/62e460cfa21a4a83933dafd89cea6960 : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/582ce27d2f9e419188dc311084544a77 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d26d4dbd94774b2eadd3d7bcae1f74de : SUCCESS in 6m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ce0da3b75c14105b9592a0bf4400e9b : SUCCESS in 6m 55s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/3ec8940cfd8e43da8f0dc9aca92d79bf : SUCCESS in 20m 57s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2abc07b01dbb401bb0dd9b8e6f84bce9 : SUCCESS in 56m 18s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/05b2303ed3e04800be565396441f9fc4 : SUCCESS in 55m 03s","accounts_in_message":[],"_revision_number":5},{"id":"cb35cf1f9fc8e18e6f4778cd83f9b027633e6328","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-28 13:25:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"b7057dfb9261199da8ec4525a3aa791a91b4616e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-28 13:26:21.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c5f7207d3934688b630827eea808793\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd6fbf47478e4a398aef345f9658822a : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/25e7df87a7e64a7faa13181f34ea76ff : SUCCESS in 39s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"8d209901fa86c58eee45502f3bb8e9ee7c84a885","revisions":{"c7d59a917155527a39f5ba06850e1d16046f9919":{"kind":"REWORK","_number":1,"created":"2022-09-15 14:17:17.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/83/857883/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/83/857883/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/857883/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/857883/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/857883/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/83/857883/1"}}},"commit":{"parents":[{"commit":"35a8be3df0232a9e59f488b9f923008c0ef7a91e","subject":"Move to CentOS Stream 9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/35a8be3df0232a9e59f488b9f923008c0ef7a91e"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-09-15 14:13:12.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-09-15 14:16:01.000000000","tz":120},"subject":"nova-libvirt: install \u0027swtpm\u0027 for TPM emulation","message":"nova-libvirt: install \u0027swtpm\u0027 for TPM emulation\n\nRHEL9 removed \u0027trousers\u0027 and has \u0027swtpm\u0027. Nova supports it for quite a\nwhile: https://review.opendev.org/c/openstack/nova/+/631363\n\nChange-Id: Ifb7e0f1632805807851720873a70179218bdf372\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c7d59a917155527a39f5ba06850e1d16046f9919"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c7d59a917155527a39f5ba06850e1d16046f9919"}]},"branch":"refs/heads/master"},"4d8174f5762811ed29667827c695a735c894cde4":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-09-26 13:26:29.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/83/857883/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/83/857883/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/857883/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/857883/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/857883/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/83/857883/2"}}},"commit":{"parents":[{"commit":"35a8be3df0232a9e59f488b9f923008c0ef7a91e","subject":"Move to CentOS Stream 9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/35a8be3df0232a9e59f488b9f923008c0ef7a91e"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-09-15 14:13:12.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-09-26 13:26:29.000000000","tz":0},"subject":"nova-libvirt: provide TPM emulation","message":"nova-libvirt: provide TPM emulation\n\nThe goal of the swtpm project is to provide a TPM emulator (TPM 1.2 \u0026\nTPM 2) that can be integrated into virtualized environments, such as\nvirtual machines and containers.\n\nNova supports it for quite a while:\nhttps://review.opendev.org/c/openstack/nova/+/631363\n\nChange-Id: Ifb7e0f1632805807851720873a70179218bdf372\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4d8174f5762811ed29667827c695a735c894cde4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4d8174f5762811ed29667827c695a735c894cde4"}]},"branch":"refs/heads/master","description":"Edit commit message"},"e31815c4e00474843669fed0b51c6120dbd7c939":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-09-26 13:26:44.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/83/857883/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/83/857883/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/857883/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/857883/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/857883/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/83/857883/3"}}},"commit":{"parents":[{"commit":"a840bd10e1b380034c1fb5d04035d94cb8638adb","subject":"Move to CentOS Stream 9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a840bd10e1b380034c1fb5d04035d94cb8638adb"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-09-15 14:13:12.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-09-26 13:26:44.000000000","tz":0},"subject":"nova-libvirt: provide TPM emulation","message":"nova-libvirt: provide TPM emulation\n\nThe goal of the swtpm project is to provide a TPM emulator (TPM 1.2 \u0026\nTPM 2) that can be integrated into virtualized environments, such as\nvirtual machines and containers.\n\nNova supports it for quite a while:\nhttps://review.opendev.org/c/openstack/nova/+/631363\n\nChange-Id: Ifb7e0f1632805807851720873a70179218bdf372\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e31815c4e00474843669fed0b51c6120dbd7c939"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e31815c4e00474843669fed0b51c6120dbd7c939"}]},"branch":"refs/heads/master","description":"Rebase"},"a12ccad4777c1af609f26d49480fda0d2cb570d3":{"kind":"REWORK","_number":4,"created":"2022-09-26 18:43:43.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/83/857883/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/83/857883/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/857883/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/857883/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/857883/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/83/857883/4"}}},"commit":{"parents":[{"commit":"cf985ca9db3ec091b9ec7d8a363d44bc2fae446a","subject":"Move to CentOS Stream 9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cf985ca9db3ec091b9ec7d8a363d44bc2fae446a"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-09-15 14:13:12.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-09-26 18:43:28.000000000","tz":120},"subject":"nova-libvirt: provide TPM emulation","message":"nova-libvirt: provide TPM emulation\n\nThe goal of the swtpm project is to provide a TPM emulator (TPM 1.2 \u0026\nTPM 2) that can be integrated into virtualized environments, such as\nvirtual machines and containers.\n\nNova supports it for quite a while:\nhttps://review.opendev.org/c/openstack/nova/+/631363\n\nChange-Id: Ifb7e0f1632805807851720873a70179218bdf372\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a12ccad4777c1af609f26d49480fda0d2cb570d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a12ccad4777c1af609f26d49480fda0d2cb570d3"}]},"branch":"refs/heads/master"},"8d209901fa86c58eee45502f3bb8e9ee7c84a885":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-09-27 10:02:23.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/83/857883/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/83/857883/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/857883/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/857883/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/857883/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/83/857883/5"}}},"commit":{"parents":[{"commit":"878b00013b01e3d02a704ce7c9fac1629cd1a705","subject":"Move to CentOS Stream 9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/878b00013b01e3d02a704ce7c9fac1629cd1a705"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-09-15 14:13:12.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-09-27 10:02:23.000000000","tz":0},"subject":"nova-libvirt: provide TPM emulation","message":"nova-libvirt: provide TPM emulation\n\nThe goal of the swtpm project is to provide a TPM emulator (TPM 1.2 \u0026\nTPM 2) that can be integrated into virtualized environments, such as\nvirtual machines and containers.\n\nNova supports it for quite a while:\nhttps://review.opendev.org/c/openstack/nova/+/631363\n\nChange-Id: Ifb7e0f1632805807851720873a70179218bdf372\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8d209901fa86c58eee45502f3bb8e9ee7c84a885"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8d209901fa86c58eee45502f3bb8e9ee7c84a885"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
