)]}'
{"id":"openstack%2Fkolla~858311","triplet_id":"openstack%2Fkolla~stable%2Fvictoria~I46f03329853a2025bff20b7b60e35c06aba619eb","project":"openstack/kolla","branch":"stable/victoria","topic":"zuul-queue-stable/yoga-stable/xena-stable/wallaby-stable/victoria","attention_set":{},"removed_from_attention_set":{"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-09-19 18:34:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I46f03329853a2025bff20b7b60e35c06aba619eb","subject":"[CI] Move queue setting to project level","status":"MERGED","created":"2022-09-19 11:56:03.000000000","updated":"2022-09-19 18:39:20.000000000","submitted":"2022-09-19 18:34:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"858311-zuul-queue-stable/yoga-stable/xena-stable/wallaby-stable/victoria","cherry_pick_of_change":858310,"cherry_pick_of_patch_set":1,"meta_rev_id":"cb8d5c2f045a62b97d089bc3e732d8fbce5767e9","_number":858311,"virtual_id_number":858311,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-19 18:34:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-09-19 15:01:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-09-19 11:56:09.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2022-09-19 11:56:09.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-19 13:19:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-19 15:01:25.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-09-19 15:03:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0cd45db0f1461f6456e9571a6925538fbdfa8f67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-09-19 11:56:03.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.","accounts_in_message":[],"_revision_number":1},{"id":"192f4b50bc1ed6ce7edb17bcfcce90121e49d82f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-09-19 11:56:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e7a63ed8580579fd7b2079c197aa2b6817681478","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-19 13:19:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6272dee8439d4e9ab6a55a053cde657f\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/9acc9fe364fe495d94def8036ef7ca22 : SUCCESS in 1h 14m 23s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/551405c5118846b49e81147f8b652895 : FAILURE in 1h 01m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"68879e1af62071c398c46b8337a58d5b4aab9f95","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-09-19 15:01:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9a7d5ec0ff94e82ad8bf368d606abe90adf1d365","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-19 15:03:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e60f01f0d7cc42cf832e2c671028f2db\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/037db7e84f3e44768d0c6bc5734f8ed1 : SUCCESS in 34m 31s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/1d85adc9a9bc46b28c9d7a5ba5f6cdd3 : SUCCESS in 33m 08s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/82536f7630b84c6c971b0a64ddb97a75 : SUCCESS in 1h 15m 02s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/21b9fa48fd1a43779dae2285610cb4cf : SUCCESS in 1h 14m 33s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/f09cbab4a2894936959e5ae5e62448af : SUCCESS in 1h 27m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0e207b37831423da928f131cbf19501 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a56c721ef33c4806bfabf646cce020bc : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/615bb1f7b5fd4be79fce1896fdcc0f01 : SUCCESS in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2888ca8bd993482a9dd075a0de7eab33 : SUCCESS in 5m 05s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/52cb84ca22d04a9f82741d874397a216 : SUCCESS in 33m 33s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/827af519f29e407ebe9b48f4d677977a : SUCCESS in 39m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f7ae2b2c39a74c20a840f5d8592c631b : SUCCESS in 1h 02m 01s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/48f2a70f9c464396bb91baef31765ca0 : SUCCESS in 55m 13s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a68211cad3324fd2abc91a320b6791fc : SUCCESS in 55m 56s","accounts_in_message":[],"_revision_number":1},{"id":"4dad6dcd5bc8a72025a413804dc85da2d57d50e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-19 15:04:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"eb6198834c4d4fa03bcd91d0eea3e37e6495ff2c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-19 18:34:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd816df6d35b45099721dc921410b329\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/fd807644154543c1ae5d9d4bf26d871a : SUCCESS in 30m 54s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/5f2b6f18118440798209418755e5914d : SUCCESS in 1h 08m 54s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/1670014cc2374efe901a9d2bf9d69599 : SUCCESS in 1h 13m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9d4269232e94ab697b7d0da0f525fa5 : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf1f5233a3644d9b954d38fde29f81cb : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a1b180b201e44f7ba21785edf4e7e78a : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f8519109f5c4eacb6d198d3553fa4b5 : SUCCESS in 5m 07s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1c2fec34c20c4361a07fd2271b991605 : SUCCESS in 36m 27s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/30b185286c5a4ecf846d613076c3566d : SUCCESS in 57m 10s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c6ec3e1b27364ed2bd463d2a1db88047 : SUCCESS in 59m 45s","accounts_in_message":[],"_revision_number":1},{"id":"79323c0bac47cafb4793a5963db2d6746fd5615b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-19 18:34:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"cb8d5c2f045a62b97d089bc3e732d8fbce5767e9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-19 18:39:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/981b826908254360ad91f5728ad69087\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa6e03c61cbf4c918563e74de7f98d1f : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"267d01bfecbe114edd286adf2947d018bbc41f0b","revisions":{"267d01bfecbe114edd286adf2947d018bbc41f0b":{"kind":"REWORK","_number":1,"created":"2022-09-19 11:56:03.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/11/858311/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/11/858311/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/858311/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/858311/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/858311/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/11/858311/1"}}},"commit":{"parents":[{"commit":"57c16d0da23908a3aec2b234a92f6b6979f59daa","subject":"mariadb: Fix repository URLs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/57c16d0da23908a3aec2b234a92f6b6979f59daa"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-05-18 07:49:38.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-09-19 11:56:03.000000000","tz":0},"subject":"[CI] Move queue setting to project level","message":"[CI] Move queue setting to project level\n\nPer [1].\n\n[1] http://lists.zuul-ci.org/pipermail/zuul-discuss/2022-May/001801.html\n\nChange-Id: I46f03329853a2025bff20b7b60e35c06aba619eb\n(cherry picked from commit 3967e5c24bd3de931ee3e3b19f98c72edf4d1e61)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/267d01bfecbe114edd286adf2947d018bbc41f0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/267d01bfecbe114edd286adf2947d018bbc41f0b"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
